)]}'
{"id":"openstack%2Fskyline-console~973093","triplet_id":"openstack%2Fskyline-console~master~I099d2cdff1b90611001aa75b083a6b87ee412c30","project":"openstack/skyline-console","branch":"master","topic":"fix-973093","attention_set":{},"removed_from_attention_set":{"28706":{"account":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"last_update":"2026-01-28 10:14:06.000000000","reason":"Change was submitted"},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-01-28 09:18:29.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"33689":{"account":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"last_update":"2026-01-28 10:14:06.000000000","reason":"Change was submitted"},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-01-28 09:38:21.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"34671":{"account":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"last_update":"2026-01-28 10:14:06.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I099d2cdff1b90611001aa75b083a6b87ee412c30","subject":"feat: Rebuild action added instance snapshot and bootable volume.","status":"MERGED","created":"2026-01-12 13:35:06.000000000","updated":"2026-01-28 10:15:09.000000000","submitted":"2026-01-28 10:14:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":31,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"973093-fix-973093","meta_rev_id":"0dbb7d7a2afd1f9914a0c0d222c27a77d29957ff","_number":973093,"virtual_id_number":973093,"owner":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-28 10:14:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-01-28 09:18:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-01-28 09:38:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-01-28 09:18:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-12 14:22:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-12 15:48:11.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2026-01-13 09:17:21.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-01-13 09:17:21.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2026-01-26 05:53:57.000000000","updated_by":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"},{"updated":"2026-01-26 05:53:57.000000000","updated_by":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"reviewer":{"_account_id":33689,"name":"xusongfu","email":"xu.songfu@99cloud.net","username":"xusongfu"},"state":"REVIEWER"}],"messages":[{"id":"5881e19641a208a1347e1b438fa6d2651a684af5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"date":"2026-01-12 13:35:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f6e8c47bc98b781271b3e16d183bc0b5e600f680","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 14:22:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9a83179ad5544b51a3052041a573b405\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d5023a70282e409c81cc0b7dc8f6f6b5 : SUCCESS in 2m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcc6777fc7c347568a0a97818b4f1c04 : SUCCESS in 4m 09s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/f80a660922374ae4a676f5793a2d5d06 : FAILURE in 4m 14s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/54d0844fe6614500a5ad88fc2c4cd45a : SUCCESS in 3m 49s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/2702911259fd4aaeaeba4ac2c7e64af2 : SUCCESS in 3m 43s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/753a8ecab958423fa7a2873031f458ed : SUCCESS in 46m 49s","accounts_in_message":[],"_revision_number":1},{"id":"fc4828c17b2561113dda674e350a90f6c69654be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-12 15:46:12.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"405983b012f4210ad913eca87b82a31c58733b60","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-12 15:48:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"36ea968d716d0caa4cb679ba94703823de601977","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 17:02:14.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b9aae5442b9649c5ab16c217cc46c1c7\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cc5662d1cc5c433795da8ca0f3c57ea4 : SUCCESS in 2m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8f772d9a3f5488ea8a9e1df7a843e8f : SUCCESS in 3m 57s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/c3b4ac58d68648e6822ed7f07d1ee649 : FAILURE in 5m 00s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/3a84ec3b59af4be18de936c1df3c7691 : SUCCESS in 3m 51s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/561751d90bb747be9fb9bd7aa9b12e41 : SUCCESS in 6m 23s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/61360c32520c4ad7b80346799fba8a33 : SUCCESS in 1h 14m 25s","accounts_in_message":[],"_revision_number":2},{"id":"ce0717aa04dadc8affb36dafed7341709cc70efb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"date":"2026-01-13 05:30:13.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"230b96dc10624801d20b1c0b47b73baaa305704c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 06:54:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b584d109990c4e92955d8ea858600c0e\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/563f6071de57478fbf0dc9d731229edc : SUCCESS in 2m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b71ee5a9de9a4f79a62c387fd504fe18 : SUCCESS in 3m 14s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/be231a040f9b42f19602188e7da74785 : SUCCESS in 5m 17s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/81608869bc244c8dbad6d6f6115ffd83 : SUCCESS in 3m 49s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/374bb894eb0d4899b2d1895dc709dd96 : SUCCESS in 7m 38s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/95ebefd469f74653814e932df67b2ba7 : SUCCESS in 1h 18m 42s","accounts_in_message":[],"_revision_number":3},{"id":"1f3d65839a2b806320f0d63e47ca98f3cc4ca697","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-13 09:17:21.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"b43c2320198292f783f3f0c5e87ea86a9eb47db0","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-01-14 02:09:49.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ab9e12a62d38446ba09cfe901eb899e5b13f22a2","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-14 06:46:30.000000000","message":"Patch Set 3: -Code-Review","accounts_in_message":[],"_revision_number":3},{"id":"14456a250704c2918703c0be322cc7755853e37d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"date":"2026-01-14 08:54:08.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"d65fea7cde3b6ade963e4a1d1513c1b9c6d75372","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"date":"2026-01-14 08:56:14.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"32c9f13cf92a3b2981b8a13bc8b58879f54c5281","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"date":"2026-01-14 08:56:50.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"bf37bd68752f36b284bc57cb6f9843e3cad6c9f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 10:12:48.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c3854630ae1d48c7aff7c3acafd205e6\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ffdbbbaff9d347e380d9ea51863208fe : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5360838bc5d14cd898f7c9206ae6249a : SUCCESS in 6m 32s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/746455a390654c2c8e2f98651d088cc8 : FAILURE in 13m 43s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/9258486165a84b12b17df10d97420b3c : SUCCESS in 4m 23s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/bc76a4ebc70e4f248ab3886b384d664b : SUCCESS in 7m 27s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/eefb867416af42e0bc92a3b79b19f06d : FAILURE in 1h 12m 01s","accounts_in_message":[],"_revision_number":5},{"id":"95ceba413e47d21bed75f9d8f04dc1aa64e3ed46","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"date":"2026-01-14 10:34:08.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"80e6e0b3886b36a7c00072aec66d43f421aedd4c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"date":"2026-01-14 10:46:00.000000000","message":"Uploaded patch set 7: New patch set was added with same tree, parent tree, and commit message as Patch Set 6.","accounts_in_message":[],"_revision_number":7},{"id":"5b8f17ac22a8d656bb7866cfaa87926d46dc8baa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 12:09:09.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/eef51922fd8348b8bc27e6f6fff08682\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e19456db33ca4687a2fde0c8989b6fe3 : SUCCESS in 3m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c77cf7dda5884cc89bec2a2a5a9ffa09 : SUCCESS in 7m 05s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/28328474414b4de092ac78f1c98f57a8 : SUCCESS in 4m 20s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/62b16cbd818d4cc49f72c56acf48404c : SUCCESS in 6m 58s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/ddf4e24c0bbd4271809691fc32e8d947 : SUCCESS in 7m 04s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/45763d6cbfb94a5db05464315817a940 : FAILURE in 1h 21m 01s","accounts_in_message":[],"_revision_number":7},{"id":"640242257fcbe5f89674e0264c065e8b36b6b6d9","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"date":"2026-01-14 18:35:43.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"5c9ee428205e528e267d05430a55a96a335f547b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 20:08:27.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3400fe7f55034ac3ab9a031ec6be6cd5\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8fcdcc1dbab943deb4f2db293d210c7c : SUCCESS in 1m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d974061aae94e2a8c7233a5f7400e08 : SUCCESS in 5m 07s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/453cdcc71cc34772b8a6dead9b8825df : SUCCESS in 10m 33s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/bd106df9381c42eeb3338d15469ff9b0 : SUCCESS in 7m 17s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/4b6fb4a7b0bd4244815ac2dde7cf8bea : SUCCESS in 8m 17s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/08479708bfcf4880968e29aef8ee30b2 : FAILURE in 1h 27m 17s","accounts_in_message":[],"_revision_number":7},{"id":"11892df7718e62bbc5b14e0c8ba461472255eaee","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"date":"2026-01-15 12:23:16.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"321028dcaf493fb58173a21604d7b38a40e810a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-15 13:39:40.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/eb89bfab536343368d74437ce02e3ff9\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7be0631661834733bdf281c3d98248dc : SUCCESS in 2m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3ddf97079af40b491f31a142da94a5d : SUCCESS in 5m 46s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/26e13480f6ab43e8aa6d94bb7c7ae3f2 : SUCCESS in 4m 06s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/750ba894737645a4a81c0be7545528c4 : SUCCESS in 3m 09s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/1fd185d86de149968c64fa5644aed16f : SUCCESS in 3m 12s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/e9d6ebc3596e4ede9e4ffde9ef80bbc4 : FAILURE in 1h 14m 48s","accounts_in_message":[],"_revision_number":7},{"id":"757c020a52a242d076df933dfa3f67f25da71bcd","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"date":"2026-01-20 14:42:15.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"29909a28cfeb5396d026c2a3b5dafa993a22aab1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 16:22:58.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c7108ec8ffeb4a838cbca45186775862\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/08f6b9348686447eb77380182e2ed30e : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/305aa261d816485dba283e68b957b4d1 : SUCCESS in 6m 35s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/0391ecffa00d4194962a9b29d17edf40 : SUCCESS in 6m 47s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/d2c6e23773dd43908b1d9d23ff110e0c : SUCCESS in 9m 24s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/5b66f918ebdc43f68add4b89ac578bf2 : SUCCESS in 5m 11s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/1c67d5dfb3f64b13962f9e4651d7a777 : FAILURE in 1h 34m 59s","accounts_in_message":[],"_revision_number":7},{"id":"6b1a51def7db2a3863defcde88542f9fa3758762","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"date":"2026-01-26 05:52:42.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"3b7ed627b285410672518fb793093eb5d42fb9a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 06:54:19.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ca654886e2c54b149196a8340752fb7f\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b4781fdae4a44368a27f5b377923b340 : SUCCESS in 2m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ed22cf600c54cb8bb4dafddf63f69cb : SUCCESS in 6m 25s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/1b3a8fc695a641e1ac3c14584f374883 : SUCCESS in 10m 54s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/920449f78eca4b7c87a84e3ed4637fe9 : SUCCESS in 4m 13s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/1ba1b15930ad409fbb2f10bd9817c8fb : SUCCESS in 8m 07s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/01cfe3ae519141f29f05df7b9d785205 : FAILURE in 1h 00m 53s","accounts_in_message":[],"_revision_number":7},{"id":"0cc617f7baa432e1b6a72a4ae545af92aa52a473","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"date":"2026-01-26 07:25:40.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"55168ce1c6c0a4f1ccd247d56ef95f5698ac3855","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"date":"2026-01-26 07:33:27.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"274a735f9de56c9edee67726680f2ac1e0d42e3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 07:33:33.000000000","message":"Patch Set 9: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/skyline-console for 973093,9","accounts_in_message":[],"_revision_number":9},{"id":"c545831c4f7918043f0acc91f2083ac05b56fcf7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"date":"2026-01-26 08:45:53.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"8bb07cee9f2cb9fc701fabd8a0941c4192388a5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 10:12:04.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8b1bd77c8774da48934cd4e2f2fc3fb\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f5a46b4d1f1443b6ae751ef6f13185d3 : SUCCESS in 2m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8908cf7040a044acbed3c25e9ed1063e : SUCCESS in 5m 24s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/e9a03ad2268c4451b9b58ab2e6b21422 : SUCCESS in 10m 45s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/fa857ed90b314e7bab5e89b80fb65b05 : SUCCESS in 8m 34s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/8cc69505370d499eaa5266c8bef71b13 : SUCCESS in 7m 06s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/346bfcf62b6a48a390225cf9f4e6bf4e : SUCCESS in 1h 25m 45s","accounts_in_message":[],"_revision_number":10},{"id":"83805421c798f4e4b57427ea28c971be5b32527a","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-26 10:14:24.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"1c7c39dbfc2360bb1c7a3b854723cdb45b3b59ce","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"date":"2026-01-26 10:24:56.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"5e2c0170c92d13396562f3b4e6c6923cdd65eb78","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"date":"2026-01-26 10:31:48.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"2fc0dca94a59dbe5bbd49cf31921a94c5e48f1e6","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-01-27 02:48:19.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"a18f31d486def59f0c576bc2e173002a5bce8764","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"date":"2026-01-27 05:50:33.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"c0dc180206eec3f39b5d916bcf3b2875fa7da68f","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"date":"2026-01-27 05:51:45.000000000","message":"Patch Set 11:\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"7f49cad5767a065a9bc5a319cb951a559bee03b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 06:44:41.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2043c41134074137b54045ffb2f7b776\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6cd39dbe50f4483d934303b9513b70b1 : SUCCESS in 1m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6da711bb64b043509cd18a3fd7951a4a : SUCCESS in 6m 17s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/5c71e08bc1d04905af13bb2b6b297653 : SUCCESS in 4m 15s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/ccfd2ad51f8e4838b1fcb6513bb70912 : SUCCESS in 8m 07s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/e3b13231c385446ea1b852616d9363bf : SUCCESS in 7m 34s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/798701166dbf42678727471238a3377c : SUCCESS in 47m 31s","accounts_in_message":[],"_revision_number":11},{"id":"c8ed2ed98963ae27f9627082d32ab89f4a539841","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"date":"2026-01-27 08:29:25.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"74bb3c39456bd78d39707157ab4340322bea1f84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 09:59:40.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd53e3583c4641c8b5087d199aec5d0a\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/700b8de9730b42c8aa9d1f7c8bd77bef : SUCCESS in 2m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c04843efb3d54d06aa0257db084bcba9 : SUCCESS in 6m 24s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/60cf939b096b45f895e0f3cb79947dde : SUCCESS in 10m 35s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/07275f06dd7a40268c33be16a21b2722 : SUCCESS in 5m 29s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/d89c8bba86ad44c3b84f7265d7211e95 : SUCCESS in 5m 12s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/ba1dc00930264fd89311f58ebd3508a0 : SUCCESS in 1h 24m 12s","accounts_in_message":[],"_revision_number":12},{"id":"7925c0d6b289798efdf1127eb31b2b9d40f6e6d0","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-01-28 03:04:31.000000000","message":"Patch Set 12:\n\n(8 comments)","accounts_in_message":[],"_revision_number":12},{"id":"b5902a6966b2fb6ef061b76c0465b955708fec4b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"date":"2026-01-28 06:06:50.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"f325e64072d5ac0a6c48cec14b64c57ee59e4314","author":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"date":"2026-01-28 06:07:12.000000000","message":"Patch Set 13:\n\n(8 comments)","accounts_in_message":[],"_revision_number":13},{"id":"728fe50b416250814c7e48e71a49a8fb53004c69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 07:40:20.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/863744dbb36d4c06a4810059cd819870\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a43b97b0950d4ab8bcb6058ccf950fea : SUCCESS in 2m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa498e61c82740f29356bbd1a0c638fb : SUCCESS in 4m 49s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/e3883e494fd14129ad35de2b0cd7f3aa : SUCCESS in 10m 42s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/9bc2116f46bc44b6ad02d52f4abe9716 : SUCCESS in 7m 43s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/7b24261fe5404c88b168c7fd76e20d5d : SUCCESS in 3m 48s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/be01fca0e7474c1f8e5c9d97ac768079 : SUCCESS in 1h 28m 02s","accounts_in_message":[],"_revision_number":13},{"id":"9eb4a2a9cf75ef989d9184546aad6deb0c1e4348","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-01-28 09:18:29.000000000","message":"Patch Set 13: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":13},{"id":"a1c760f9436cc44874c40d76dfc5d744eae23938","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 09:18:56.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"246b8aa51becc96873cb1842b9d3cba03e41bbd7","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-28 09:38:21.000000000","message":"Patch Set 13: Code-Review+1","accounts_in_message":[],"_revision_number":13},{"id":"06bd6b58511ae60b07c81be2f60e2bb654bd7e43","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 10:14:06.000000000","message":"Patch Set 13: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca686810c4594738a6645447a20a3637\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/76a38a281597458f974504d7c1e73230 : SUCCESS in 3m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ab16ee0859a4d0ebdbfec82262264e2 : SUCCESS in 3m 59s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/b2c88431de2047359261c36774053c3d : SUCCESS in 4m 15s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/c65aa11759754ebb9a4a14ff66f81f97 : SUCCESS in 5m 02s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/55a8d11946de4d2f89372866677d82b2 : SUCCESS in 5m 13s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/e7bb564d1f0f42bc92264af76bfe25ba : SUCCESS in 53m 21s","accounts_in_message":[],"_revision_number":13},{"id":"96bb887fc51898af6013a7ef8a78f05026ad0ae0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 10:14:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":13},{"id":"0dbb7d7a2afd1f9914a0c0d222c27a77d29957ff","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 10:15:09.000000000","message":"Patch Set 13:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/045acf6c1dad49508eae22b2bf875ee6\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/32acc29647a54c8ba858e4b176b3d604 : SUCCESS in 40s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a79de19e4794ad4bd33d7b85a211e33 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"e0bc8b888737661fbc18aa1ab34d9529202e43c9","revisions":{"6bed6a3cd14c43a253e753aee5cdeedbdd4aad2d":{"kind":"REWORK","_number":1,"created":"2026-01-12 13:35:06.000000000","uploader":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"ref":"refs/changes/93/973093/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/93/973093/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/93/973093/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/93/973093/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/93/973093/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/93/973093/1"}}},"commit":{"parents":[{"commit":"4d9c0153a412f1486460879c752d96a46b6cca54","subject":"Merge \"Correct Volume status naming on Resource Page\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/4d9c0153a412f1486460879c752d96a46b6cca54"}]}],"author":{"name":"resitdemir","email":"resit.demir@tubitak.gov.tr","date":"2026-01-12 12:48:31.000000000","tz":0},"committer":{"name":"resitdemir","email":"resit.demir@tubitak.gov.tr","date":"2026-01-12 13:32:57.000000000","tz":0},"subject":"feat: Rebuild action added instance snapshot and bootable volume.","message":"feat: Rebuild action added instance snapshot and bootable volume.\n\nChange-Id: I099d2cdff1b90611001aa75b083a6b87ee412c30\nSigned-off-by: resitdemir \u003cresit.demir@tubitak.gov.tr\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6bed6a3cd14c43a253e753aee5cdeedbdd4aad2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6bed6a3cd14c43a253e753aee5cdeedbdd4aad2d"}]},"branch":"refs/heads/master"},"696078affcae2ff609c0b39bbe0214094fc121bf":{"kind":"REWORK","_number":2,"created":"2026-01-12 15:46:12.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/93/973093/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/93/973093/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/93/973093/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/93/973093/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/93/973093/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/93/973093/2"}}},"commit":{"parents":[{"commit":"4d9c0153a412f1486460879c752d96a46b6cca54","subject":"Merge \"Correct Volume status naming on Resource Page\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/4d9c0153a412f1486460879c752d96a46b6cca54"}]}],"author":{"name":"resitdemir","email":"resit.demir@tubitak.gov.tr","date":"2026-01-12 12:48:31.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-12 15:46:11.000000000","tz":0},"subject":"feat: Rebuild action added instance snapshot and bootable volume.","message":"feat: Rebuild action added instance snapshot and bootable volume.\n\nChange-Id: I099d2cdff1b90611001aa75b083a6b87ee412c30\nSigned-off-by: resitdemir \u003cresit.demir@tubitak.gov.tr\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/696078affcae2ff609c0b39bbe0214094fc121bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/696078affcae2ff609c0b39bbe0214094fc121bf"}]},"branch":"refs/heads/master"},"45d33b7c7ffa0af174c103a94ba65c1778aae604":{"kind":"REWORK","_number":3,"created":"2026-01-13 05:30:13.000000000","uploader":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"ref":"refs/changes/93/973093/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/93/973093/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/93/973093/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/93/973093/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/93/973093/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/93/973093/3"}}},"commit":{"parents":[{"commit":"4d9c0153a412f1486460879c752d96a46b6cca54","subject":"Merge \"Correct Volume status naming on Resource Page\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/4d9c0153a412f1486460879c752d96a46b6cca54"}]}],"author":{"name":"resitdemir","email":"resit.demir@tubitak.gov.tr","date":"2026-01-12 12:48:31.000000000","tz":0},"committer":{"name":"resitdemir","email":"resit.demir@tubitak.gov.tr","date":"2026-01-13 05:29:28.000000000","tz":0},"subject":"feat: Rebuild action added instance snapshot and bootable volume.","message":"feat: Rebuild action added instance snapshot and bootable volume.\n\nChange-Id: I099d2cdff1b90611001aa75b083a6b87ee412c30\nSigned-off-by: resitdemir \u003cresit.demir@tubitak.gov.tr\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/45d33b7c7ffa0af174c103a94ba65c1778aae604"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/45d33b7c7ffa0af174c103a94ba65c1778aae604"}]},"branch":"refs/heads/master"},"926b0d78f61197e31e2a98c11bb3221fcc8ec84e":{"kind":"REWORK","_number":4,"created":"2026-01-14 08:54:08.000000000","uploader":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"ref":"refs/changes/93/973093/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/93/973093/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/93/973093/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/93/973093/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/93/973093/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/93/973093/4"}}},"commit":{"parents":[{"commit":"4d9c0153a412f1486460879c752d96a46b6cca54","subject":"Merge \"Correct Volume status naming on Resource Page\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/4d9c0153a412f1486460879c752d96a46b6cca54"}]}],"author":{"name":"resitdemir","email":"resit.demir@tubitak.gov.tr","date":"2026-01-12 12:48:31.000000000","tz":0},"committer":{"name":"resitdemir","email":"resit.demir@tubitak.gov.tr","date":"2026-01-14 08:53:20.000000000","tz":0},"subject":"feat: Rebuild action added instance snapshot and bootable volume.","message":"feat: Rebuild action added instance snapshot and bootable volume.\n\nChange-Id: I099d2cdff1b90611001aa75b083a6b87ee412c30\nSigned-off-by: resitdemir \u003cresit.demir@tubitak.gov.tr\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/926b0d78f61197e31e2a98c11bb3221fcc8ec84e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/926b0d78f61197e31e2a98c11bb3221fcc8ec84e"}]},"branch":"refs/heads/master"},"3118fddfd4eff42400b12dc9d4f6baf572b1410f":{"kind":"REWORK","_number":5,"created":"2026-01-14 08:56:14.000000000","uploader":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"ref":"refs/changes/93/973093/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/93/973093/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/93/973093/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/93/973093/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/93/973093/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/93/973093/5"}}},"commit":{"parents":[{"commit":"4d9c0153a412f1486460879c752d96a46b6cca54","subject":"Merge \"Correct Volume status naming on Resource Page\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/4d9c0153a412f1486460879c752d96a46b6cca54"}]}],"author":{"name":"resitdemir","email":"resit.demir@tubitak.gov.tr","date":"2026-01-12 12:48:31.000000000","tz":0},"committer":{"name":"resitdemir","email":"resit.demir@tubitak.gov.tr","date":"2026-01-14 08:55:30.000000000","tz":0},"subject":"feat: Rebuild action added instance snapshot and bootable volume.","message":"feat: Rebuild action added instance snapshot and bootable volume.\n\nChange-Id: I099d2cdff1b90611001aa75b083a6b87ee412c30\nSigned-off-by: resitdemir \u003cresit.demir@tubitak.gov.tr\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/3118fddfd4eff42400b12dc9d4f6baf572b1410f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/3118fddfd4eff42400b12dc9d4f6baf572b1410f"}]},"branch":"refs/heads/master"},"862ac6c96fe174f9709091f58eab271f1845209f":{"kind":"REWORK","_number":6,"created":"2026-01-14 10:34:08.000000000","uploader":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"ref":"refs/changes/93/973093/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/93/973093/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/93/973093/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/93/973093/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/93/973093/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/93/973093/6"}}},"commit":{"parents":[{"commit":"4d9c0153a412f1486460879c752d96a46b6cca54","subject":"Merge \"Correct Volume status naming on Resource Page\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/4d9c0153a412f1486460879c752d96a46b6cca54"}]}],"author":{"name":"resitdemir","email":"resit.demir@tubitak.gov.tr","date":"2026-01-12 12:48:31.000000000","tz":0},"committer":{"name":"resitdemir","email":"resit.demir@tubitak.gov.tr","date":"2026-01-14 10:28:50.000000000","tz":0},"subject":"feat: Rebuild action added instance snapshot and bootable volume.","message":"feat: Rebuild action added instance snapshot and bootable volume.\n\nChange-Id: I099d2cdff1b90611001aa75b083a6b87ee412c30\nSigned-off-by: resitdemir \u003cresit.demir@tubitak.gov.tr\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/862ac6c96fe174f9709091f58eab271f1845209f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/862ac6c96fe174f9709091f58eab271f1845209f"}]},"branch":"refs/heads/master"},"294676e2a37050d3ed1813ff5945b590b16d5442":{"kind":"NO_CHANGE","_number":7,"created":"2026-01-14 10:46:00.000000000","uploader":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"ref":"refs/changes/93/973093/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/93/973093/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/93/973093/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/93/973093/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/93/973093/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/93/973093/7"}}},"commit":{"parents":[{"commit":"4d9c0153a412f1486460879c752d96a46b6cca54","subject":"Merge \"Correct Volume status naming on Resource Page\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/4d9c0153a412f1486460879c752d96a46b6cca54"}]}],"author":{"name":"resitdemir","email":"resit.demir@tubitak.gov.tr","date":"2026-01-12 12:48:31.000000000","tz":0},"committer":{"name":"resitdemir","email":"resit.demir@tubitak.gov.tr","date":"2026-01-14 10:44:57.000000000","tz":0},"subject":"feat: Rebuild action added instance snapshot and bootable volume.","message":"feat: Rebuild action added instance snapshot and bootable volume.\n\nChange-Id: I099d2cdff1b90611001aa75b083a6b87ee412c30\nSigned-off-by: resitdemir \u003cresit.demir@tubitak.gov.tr\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/294676e2a37050d3ed1813ff5945b590b16d5442"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/294676e2a37050d3ed1813ff5945b590b16d5442"}]},"branch":"refs/heads/master"},"051df6656d0e9a0a91f7625d0d54b3482ec8f38c":{"kind":"REWORK","_number":8,"created":"2026-01-26 07:25:40.000000000","uploader":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"ref":"refs/changes/93/973093/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/93/973093/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/93/973093/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/93/973093/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/93/973093/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/93/973093/8"}}},"commit":{"parents":[{"commit":"4d9c0153a412f1486460879c752d96a46b6cca54","subject":"Merge \"Correct Volume status naming on Resource Page\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/4d9c0153a412f1486460879c752d96a46b6cca54"}]}],"author":{"name":"resitdemir","email":"resit.demir@tubitak.gov.tr","date":"2026-01-12 12:48:31.000000000","tz":0},"committer":{"name":"resitdemir","email":"resit.demir@tubitak.gov.tr","date":"2026-01-26 07:25:04.000000000","tz":0},"subject":"feat: Rebuild action added instance snapshot and bootable volume.","message":"feat: Rebuild action added instance snapshot and bootable volume.\n\nChange-Id: I099d2cdff1b90611001aa75b083a6b87ee412c30\nSigned-off-by: resitdemir \u003cresit.demir@tubitak.gov.tr\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/051df6656d0e9a0a91f7625d0d54b3482ec8f38c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/051df6656d0e9a0a91f7625d0d54b3482ec8f38c"}]},"branch":"refs/heads/master"},"1a8d0328f5981f64d95bfb409409627059c59fe9":{"kind":"REWORK","_number":9,"created":"2026-01-26 07:33:27.000000000","uploader":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"ref":"refs/changes/93/973093/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/93/973093/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/93/973093/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/93/973093/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/93/973093/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/93/973093/9"}}},"commit":{"parents":[{"commit":"4d9c0153a412f1486460879c752d96a46b6cca54","subject":"Merge \"Correct Volume status naming on Resource Page\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/4d9c0153a412f1486460879c752d96a46b6cca54"}]}],"author":{"name":"resitdemir","email":"resit.demir@tubitak.gov.tr","date":"2026-01-12 12:48:31.000000000","tz":0},"committer":{"name":"resitdemir","email":"resit.demir@tubitak.gov.tr","date":"2026-01-26 07:31:29.000000000","tz":0},"subject":"feat: Rebuild action added instance snapshot and bootable volume.","message":"feat: Rebuild action added instance snapshot and bootable volume.\n\nChange-Id: I099d2cdff1b90611001aa75b083a6b87ee412c30\nSigned-off-by: resitdemir \u003cresit.demir@tubitak.gov.tr\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/1a8d0328f5981f64d95bfb409409627059c59fe9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/1a8d0328f5981f64d95bfb409409627059c59fe9"}]},"branch":"refs/heads/master"},"a6032d8145811aa3422e743c010c71c9c73d5795":{"kind":"REWORK","_number":10,"created":"2026-01-26 08:45:53.000000000","uploader":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"ref":"refs/changes/93/973093/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/93/973093/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/93/973093/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/93/973093/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/93/973093/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/93/973093/10"}}},"commit":{"parents":[{"commit":"e3d4e427eecddc57bbb5b7c0a740cc83475e0a2e","subject":"Merge \"docs: add missing pymysql dependency for database connectivity\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e3d4e427eecddc57bbb5b7c0a740cc83475e0a2e"}]}],"author":{"name":"resitdemir","email":"resit.demir@tubitak.gov.tr","date":"2026-01-12 12:48:31.000000000","tz":0},"committer":{"name":"resitdemir","email":"resit.demir@tubitak.gov.tr","date":"2026-01-26 08:45:24.000000000","tz":0},"subject":"feat: Rebuild action added instance snapshot and bootable volume.","message":"feat: Rebuild action added instance snapshot and bootable volume.\n\nChange-Id: I099d2cdff1b90611001aa75b083a6b87ee412c30\nSigned-off-by: resitdemir \u003cresit.demir@tubitak.gov.tr\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a6032d8145811aa3422e743c010c71c9c73d5795"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a6032d8145811aa3422e743c010c71c9c73d5795"}]},"branch":"refs/heads/master"},"d33bbec4ccee8a3f94c540b08a16aa44139c2191":{"kind":"REWORK","_number":11,"created":"2026-01-27 05:50:33.000000000","uploader":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"ref":"refs/changes/93/973093/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/93/973093/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/93/973093/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/93/973093/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/93/973093/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/93/973093/11"}}},"commit":{"parents":[{"commit":"e3d4e427eecddc57bbb5b7c0a740cc83475e0a2e","subject":"Merge \"docs: add missing pymysql dependency for database connectivity\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e3d4e427eecddc57bbb5b7c0a740cc83475e0a2e"}]}],"author":{"name":"resitdemir","email":"resit.demir@tubitak.gov.tr","date":"2026-01-12 12:48:31.000000000","tz":0},"committer":{"name":"resitdemir","email":"resit.demir@tubitak.gov.tr","date":"2026-01-27 05:50:02.000000000","tz":0},"subject":"feat: Rebuild action added instance snapshot and bootable volume.","message":"feat: Rebuild action added instance snapshot and bootable volume.\n\nChange-Id: I099d2cdff1b90611001aa75b083a6b87ee412c30\nSigned-off-by: resitdemir \u003cresit.demir@tubitak.gov.tr\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d33bbec4ccee8a3f94c540b08a16aa44139c2191"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d33bbec4ccee8a3f94c540b08a16aa44139c2191"}]},"branch":"refs/heads/master"},"8b8aafad9ec0bf4e211c677749548649015523fa":{"kind":"REWORK","_number":12,"created":"2026-01-27 08:29:25.000000000","uploader":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"ref":"refs/changes/93/973093/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/93/973093/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/93/973093/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/93/973093/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/93/973093/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/93/973093/12"}}},"commit":{"parents":[{"commit":"e3d4e427eecddc57bbb5b7c0a740cc83475e0a2e","subject":"Merge \"docs: add missing pymysql dependency for database connectivity\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e3d4e427eecddc57bbb5b7c0a740cc83475e0a2e"}]}],"author":{"name":"resitdemir","email":"resit.demir@tubitak.gov.tr","date":"2026-01-12 12:48:31.000000000","tz":0},"committer":{"name":"resitdemir","email":"resit.demir@tubitak.gov.tr","date":"2026-01-27 08:26:54.000000000","tz":0},"subject":"feat: Rebuild action added instance snapshot and bootable volume.","message":"feat: Rebuild action added instance snapshot and bootable volume.\n\nChange-Id: I099d2cdff1b90611001aa75b083a6b87ee412c30\nSigned-off-by: resitdemir \u003cresit.demir@tubitak.gov.tr\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/8b8aafad9ec0bf4e211c677749548649015523fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/8b8aafad9ec0bf4e211c677749548649015523fa"}]},"branch":"refs/heads/master"},"e0bc8b888737661fbc18aa1ab34d9529202e43c9":{"kind":"REWORK","_number":13,"created":"2026-01-28 06:06:50.000000000","uploader":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"ref":"refs/changes/93/973093/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/93/973093/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/93/973093/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/93/973093/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/93/973093/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/93/973093/13"}}},"commit":{"parents":[{"commit":"e3d4e427eecddc57bbb5b7c0a740cc83475e0a2e","subject":"Merge \"docs: add missing pymysql dependency for database connectivity\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e3d4e427eecddc57bbb5b7c0a740cc83475e0a2e"}]}],"author":{"name":"resitdemir","email":"resit.demir@tubitak.gov.tr","date":"2026-01-12 12:48:31.000000000","tz":0},"committer":{"name":"resitdemir","email":"resit.demir@tubitak.gov.tr","date":"2026-01-28 06:06:30.000000000","tz":0},"subject":"feat: Rebuild action added instance snapshot and bootable volume.","message":"feat: Rebuild action added instance snapshot and bootable volume.\n\nChange-Id: I099d2cdff1b90611001aa75b083a6b87ee412c30\nSigned-off-by: resitdemir \u003cresit.demir@tubitak.gov.tr\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e0bc8b888737661fbc18aa1ab34d9529202e43c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e0bc8b888737661fbc18aa1ab34d9529202e43c9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
