)]}'
{"id":"openstack%2Fskyline-console~973209","triplet_id":"openstack%2Fskyline-console~master~I4a87f64fea699b82d06d1e3982fc32d5a931a77d","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-01-14 09:42:17.000000000","reason":"Change was submitted"},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-01-14 02:01:27.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-01-14 06:45:03.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"I4a87f64fea699b82d06d1e3982fc32d5a931a77d","subject":"Correct various grammar mistakes","status":"MERGED","created":"2026-01-13 09:23:29.000000000","updated":"2026-01-14 09:43:09.000000000","submitted":"2026-01-14 09:42:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"973209","meta_rev_id":"892cf259a9e4a92cbb1e000db5b805ec9f0bb565","_number":973209,"virtual_id_number":973209,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-14 09:42:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-01-14 02:01:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-01-14 02:01:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-13 09:25:21.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"},{"updated":"2026-01-13 09:25:21.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2026-01-13 10:19:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a1421aa2863614cf3f72e45a87337385d018aef9","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-13 09:23:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a37e5b9cf49723a1e4e222d55afe519bfc5aeeb6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-13 09:24:20.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"77afe646a9a4bb9692860ccce599374b7a0518e4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-13 09:24:49.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"7d9ee4c30ede45d9ddfd9eeedd306edbf99d850f","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-13 09:25:04.000000000","message":"Patch Set 3:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"38550ba63ff615b3a6cfdf950d03120e79364cb5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 10:19:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5e59e907036491d82588c1af7a8439e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21e4548a8fa14f41b2172b8303ef6390 : SUCCESS in 6m 27s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/d2ac5224ce4a491398519ea1e28ed43b : SUCCESS in 5m 19s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/c482d771a02b4456a1bdebd17ea2c4f0 : SUCCESS in 6m 25s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/18592b9dd8024ef2a1f8485999fcd21d : SUCCESS in 4m 47s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/fe324718eb5042d6a248a7d6a5ad9252 : SUCCESS in 49m 13s","accounts_in_message":[],"_revision_number":3},{"id":"26b10b9bb25415a6a60999644bf85b6edda1d8bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-13 10:26:41.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"557c44408072e2342155a462328c2fae15ab8324","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-13 11:15:12.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":4},{"id":"abda04e58f981df6c1ed2382739b30d35e399c8b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-13 11:15:20.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"e801a9e04a28edb592ce07adc13ca9990e383943","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-13 11:21:43.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"d4c069564a4c9dbc1f8c5cc808c4bbc600d19dc3","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-13 11:21:47.000000000","message":"Patch Set 6:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":6},{"id":"f55834d209ce7502f56e8aaf1c41aadf1d3cace3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-13 11:26:12.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"89c5dff0ed8af832f0fabfaa524d408c581d0ed9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-13 11:28:37.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"1a1e6ccbf90ceff4cde9deb770fb058e89faecd9","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-13 11:29:17.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":8},{"id":"73a49fc18a38953a9dea637cb33963745362693d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-13 11:36:40.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"4afe2e35487720826f69b0e6e2f63c19d300730a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-13 11:41:33.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"9f7d4c3409632f1d27a5ffccdb9c97d8b7153fb8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-13 11:48:19.000000000","message":"Patch Set 11: Published edit on patch set 10.","accounts_in_message":[],"_revision_number":11},{"id":"51b645c5ec1d90ddcf0bc20b7b317a9987c6ab4b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-13 11:49:25.000000000","message":"Patch Set 12: Published edit on patch set 11.","accounts_in_message":[],"_revision_number":12},{"id":"2b0b0490561ebc8894c460483a8abf5bcf9e3864","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-13 11:50:25.000000000","message":"Patch Set 13: Published edit on patch set 12.","accounts_in_message":[],"_revision_number":13},{"id":"58669f1bac0bd74cdd0459646890c27bd5595473","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-13 11:52:22.000000000","message":"Patch Set 14: Published edit on patch set 13.","accounts_in_message":[],"_revision_number":14},{"id":"5a29830239edb0fdb2adf5b08d4d962c7744280e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-13 11:54:07.000000000","message":"Patch Set 15: Published edit on patch set 14.","accounts_in_message":[],"_revision_number":15},{"id":"cadc0831be294fd95b80eaab6846f9562f128283","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-13 11:58:46.000000000","message":"Patch Set 16: Published edit on patch set 15.","accounts_in_message":[],"_revision_number":16},{"id":"28fff5891912f9760920212e37af6b97f64e9845","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-13 12:03:02.000000000","message":"Patch Set 17: Published edit on patch set 16.","accounts_in_message":[],"_revision_number":17},{"id":"e2e19eaba24de5f1bc9e65d247db8586010bc3a6","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-13 12:04:19.000000000","message":"Patch Set 17:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":17},{"id":"da5a73e942943389739c4b5e5f6f3c391efd1dea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-13 12:32:23.000000000","message":"Patch Set 18: Published edit on patch set 17.","accounts_in_message":[],"_revision_number":18},{"id":"e50bd8731374f9b588aef542ff3ffad143fa47f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-13 12:35:29.000000000","message":"Patch Set 19: Commit message was updated.","accounts_in_message":[],"_revision_number":19},{"id":"1e359fb17b62d7cdfe679d5bc0f84f8279585052","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-13 13:33:56.000000000","message":"Patch Set 20: Published edit on patch set 19.","accounts_in_message":[],"_revision_number":20},{"id":"c3cdbbe236b12660da2a178048ef197cc7293129","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-13 13:37:47.000000000","message":"Patch Set 21: Published edit on patch set 20.","accounts_in_message":[],"_revision_number":21},{"id":"5e82318a948e4f3cbd1e773de322e3c3ceab9ccb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-13 13:39:11.000000000","message":"Patch Set 22: Commit message was updated.","accounts_in_message":[],"_revision_number":22},{"id":"cc75e2922d99bd9df02de2613601c2e538dfaaee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-13 13:56:06.000000000","message":"Patch Set 23: Published edit on patch set 22.","accounts_in_message":[],"_revision_number":23},{"id":"1013fcb7d45f1c14c91281188a1d15efed339925","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-13 13:57:58.000000000","message":"Patch Set 24: Published edit on patch set 23.","accounts_in_message":[],"_revision_number":24},{"id":"b38cbbbc20c847381253ade1f0f1bd7dd6bd9423","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-13 13:58:45.000000000","message":"Patch Set 25: Published edit on patch set 24.","accounts_in_message":[],"_revision_number":25},{"id":"20c46a17f5857d4a6678f7aa0dbb7df79e73fc5c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-13 14:02:29.000000000","message":"Patch Set 26: Published edit on patch set 25.","accounts_in_message":[],"_revision_number":26},{"id":"63a74f2083ec319b720a625e51cecfc01b3e5352","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-13 14:03:14.000000000","message":"Patch Set 27: Published edit on patch set 26.","accounts_in_message":[],"_revision_number":27},{"id":"b53e6484f73bee5beaecc33f991071dfe5126861","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-13 14:04:23.000000000","message":"Patch Set 28: Published edit on patch set 27.","accounts_in_message":[],"_revision_number":28},{"id":"3c1c3b9a5ea9f43e9da6499a8dbed09df60f5871","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 15:30:20.000000000","message":"Patch Set 28: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c167fb65f4154223a5bc0a51f88dd8cb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2198d3dbd71f44438870eaaa2e92a401 : SUCCESS in 6m 36s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/594e590253314e16ae3ca7f72d4bb3b7 : SUCCESS in 12m 11s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/b41c0f7a9af2420cae1792ae7252acb2 : SUCCESS in 7m 14s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/cabea12aa9cf4a909203b6207663a566 : SUCCESS in 4m 22s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/fcf916907c134812b810989190af1928 : SUCCESS in 1h 25m 17s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/0f15162318504a9c960e3c414b8aac44 : SUCCESS in 1h 17m 30s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/e8f80a7a4cdb4f17bf16f7fa3c753098 : SUCCESS in 49m 05s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/bcd692663b3e474aa6bff083dd5dff30 : SUCCESS in 52m 21s","accounts_in_message":[],"_revision_number":28},{"id":"3f38f3fc578206a815a41688f438cdc10f63a63e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-13 23:14:20.000000000","message":"Patch Set 29: Published edit on patch set 28.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":29},{"id":"1202c1fb8c2b2e4a4c3fd19d17eae65f1a785932","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-13 23:14:52.000000000","message":"Patch Set 30: Published edit on patch set 29.","accounts_in_message":[],"_revision_number":30},{"id":"5c6711ca27292737d604f5674bb98d250b45bcba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 00:21:30.000000000","message":"Patch Set 30: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ef87e988e60415cad0f56fbd3684cd1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/866ec4a181a24ac48e87dc7c3e3c55ec : SUCCESS in 6m 12s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/cc0d6fa1736c42e3854cac11728e20b7 : SUCCESS in 11m 37s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/43a88efdd82d4eb782998ae10b5c96f2 : SUCCESS in 4m 31s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/cf6201f4b23046cdb5f557d97c26054a : SUCCESS in 8m 45s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/4e5b2fc73c6b486f8bd54c49a65b651b : SUCCESS in 1h 05m 34s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/11305825072c4e8aa4dcc51caa982a41 : SUCCESS in 1h 00m 46s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/46e4e09a4d964014a5f531b4b25b7d00 : SUCCESS in 52m 55s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/ac5283f351684b2a996b02d5f887a063 : SUCCESS in 50m 03s","accounts_in_message":[],"_revision_number":30},{"id":"d33dca1752319cc477616340986b196b2d727a4d","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-01-14 02:01:27.000000000","message":"Patch Set 30: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":30},{"id":"3b0a85cae56caaacdde4aba3f2aca6f6d21fa063","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 02:04:58.000000000","message":"Patch Set 30: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":30},{"id":"aaacb9153a346fc1788aa553237dab77da5a9e84","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 04:06:32.000000000","message":"Patch Set 30: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dce5dfee3a0a4acea3d6147977ce145a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a428a4f5ff1640129a95c15f20886d2c : SUCCESS in 6m 34s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/b2b80bb7967a4cc588a62a677326f0d0 : SUCCESS in 10m 23s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/af6adb26717c4dc1b289979837f9dcc6 : SUCCESS in 3m 40s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/3b2ebc91a5074e878ff443af4067f4c2 : SUCCESS in 8m 43s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/0ea0589d5f5748e6b6ba78fe3944e78b : SUCCESS in 1h 27m 03s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/07dca9c28dac4bde8d304ec3cc1e6200 : SUCCESS in 1h 08m 25s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/239aff24a00d4f519d127408359f20fa : FAILURE in 45m 17s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/08a75e4f55ec4854a3b2f6516199b427 : SUCCESS in 46m 53s","accounts_in_message":[],"_revision_number":30},{"id":"71756e62a63bb841913ec0d30a0a0bb3df92fcb6","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-14 06:45:03.000000000","message":"Patch Set 30:\n\n(1 comment)","accounts_in_message":[],"_revision_number":30},{"id":"f08e22f69733a2f3c366d715df157f5fcac5916e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 08:14:58.000000000","message":"Patch Set 30: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf17e430154848b28e6fbe683e38b30f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fbc5c61a6bc46eda8352abcd176964e : SUCCESS in 3m 48s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/cee24d96475b4ae5829b39278e8e24b5 : SUCCESS in 11m 37s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/f5328d3834ce4871a2a60781544f555d : SUCCESS in 7m 42s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/77914cc759984b6b9fd9c3785a3b7f41 : SUCCESS in 8m 37s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/8087718616344f8eb1a762649ec4adff : SUCCESS in 1h 25m 00s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/30539d9364604bbbbb9a422b363d0c1c : SUCCESS in 31m 41s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/bbc7795656344f638e92a144606e0bf7 : SUCCESS in 55m 59s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/063e747e9333479481b3ef4da95976b9 : SUCCESS in 50m 08s","accounts_in_message":[],"_revision_number":30},{"id":"3f0cc5dbca1c8ba56af420b31baf889302240a07","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 08:15:28.000000000","message":"Patch Set 30: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":30},{"id":"e3e70ba0ec8b4d2a5016d4d1e0feb9417805f060","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 09:42:16.000000000","message":"Patch Set 30: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37e607a2e3f54a71ac6ca00239085a74\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/849afce28dc34ad6b53a5f9d60c21ae4 : SUCCESS in 4m 10s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/38c3f81ab6a6435aab4ec5ccd5bcc194 : SUCCESS in 4m 49s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/fc67fff6916a46f5b2f971d794b2f884 : SUCCESS in 7m 51s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/34fa15a0f599484994beb1977e1a1cfe : SUCCESS in 3m 41s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/69f4d0e380a047edb500d19b1fe7ee18 : SUCCESS in 1h 22m 13s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/94bc25efb16f43e7ae9f2b27f2ef5ba0 : SUCCESS in 1h 05m 21s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/d5a036ebfad64eefbed5e54b207215a0 : SUCCESS in 32m 07s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/583af60ada894fdba012fa2ec4daf0c8 : SUCCESS in 30m 49s","accounts_in_message":[],"_revision_number":30},{"id":"7a1082d167268fbd14e1a68a7beee473ffef9e5f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 09:42:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":30},{"id":"892cf259a9e4a92cbb1e000db5b805ec9f0bb565","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 09:43:09.000000000","message":"Patch Set 30:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b80a9d78fae2496489c2a27bbd0a90ff\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65738b0644c240a9857f0f4ec14d84b7 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":30}],"current_revision_number":30,"current_revision":"a144251d1a8d03773d77073e2b947a7d1d518ff1","revisions":{"a86a645962da364a5632a98d6bbdc54f8f3440e0":{"kind":"REWORK","_number":1,"created":"2026-01-13 09:23:29.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/09/973209/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/09/973209/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/1"}}},"commit":{"parents":[{"commit":"cc77a164297c4d3f835f0a7b9569d54ac74ceaa4","subject":"Merge \"Rename \"Show Instance\" checkbox\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/cc77a164297c4d3f835f0a7b9569d54ac74ceaa4"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-13 09:23:29.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-13 09:23:29.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: I4a87f64fea699b82d06d1e3982fc32d5a931a77d\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a86a645962da364a5632a98d6bbdc54f8f3440e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a86a645962da364a5632a98d6bbdc54f8f3440e0"}]},"branch":"refs/heads/master"},"6bc065095204b9e2ca72efb605a132ab3a5f0e21":{"kind":"REWORK","_number":2,"created":"2026-01-13 09:24:20.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/09/973209/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/09/973209/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/2"}}},"commit":{"parents":[{"commit":"cc77a164297c4d3f835f0a7b9569d54ac74ceaa4","subject":"Merge \"Rename \"Show Instance\" checkbox\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/cc77a164297c4d3f835f0a7b9569d54ac74ceaa4"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-13 09:23:29.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-13 09:24:20.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: I4a87f64fea699b82d06d1e3982fc32d5a931a77d\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6bc065095204b9e2ca72efb605a132ab3a5f0e21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6bc065095204b9e2ca72efb605a132ab3a5f0e21"}]},"branch":"refs/heads/master"},"25bf702c1316c3049c07fa0ffc03b7a40f14f6d4":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-01-13 09:24:49.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/09/973209/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/09/973209/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/3"}}},"commit":{"parents":[{"commit":"cc77a164297c4d3f835f0a7b9569d54ac74ceaa4","subject":"Merge \"Rename \"Show Instance\" checkbox\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/cc77a164297c4d3f835f0a7b9569d54ac74ceaa4"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-13 09:23:29.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-13 09:24:49.000000000","tz":0},"subject":"Small fix for writing Front-End / Back-End","message":"Small fix for writing Front-End / Back-End\n\ntrivial fix\n\nChange-Id: I4a87f64fea699b82d06d1e3982fc32d5a931a77d\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/25bf702c1316c3049c07fa0ffc03b7a40f14f6d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/25bf702c1316c3049c07fa0ffc03b7a40f14f6d4"}]},"branch":"refs/heads/master","description":"Edit commit message"},"96a319a4e126899798bfc15ea679ad877f4b6a56":{"kind":"REWORK","_number":4,"created":"2026-01-13 10:26:41.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/09/973209/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/09/973209/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/4"}}},"commit":{"parents":[{"commit":"cc77a164297c4d3f835f0a7b9569d54ac74ceaa4","subject":"Merge \"Rename \"Show Instance\" checkbox\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/cc77a164297c4d3f835f0a7b9569d54ac74ceaa4"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-13 09:23:29.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-13 10:26:41.000000000","tz":0},"subject":"Small fix for writing Front-End / Back-End","message":"Small fix for writing Front-End / Back-End\n\ntrivial fix\n\nChange-Id: I4a87f64fea699b82d06d1e3982fc32d5a931a77d\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/96a319a4e126899798bfc15ea679ad877f4b6a56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/96a319a4e126899798bfc15ea679ad877f4b6a56"}]},"branch":"refs/heads/master"},"a32f49be5af6c15886fb3a4537c8394050ca4660":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2026-01-13 11:15:20.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/09/973209/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/09/973209/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/5"}}},"commit":{"parents":[{"commit":"cc77a164297c4d3f835f0a7b9569d54ac74ceaa4","subject":"Merge \"Rename \"Show Instance\" checkbox\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/cc77a164297c4d3f835f0a7b9569d54ac74ceaa4"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-13 09:23:29.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-13 11:15:20.000000000","tz":0},"subject":"Fix for writing Front-End / Back-End","message":"Fix for writing Front-End / Back-End\n\nAnd few small fixes.\n\nChange-Id: I4a87f64fea699b82d06d1e3982fc32d5a931a77d\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a32f49be5af6c15886fb3a4537c8394050ca4660"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a32f49be5af6c15886fb3a4537c8394050ca4660"}]},"branch":"refs/heads/master","description":"Edit commit message"},"9652a1b35421be5c3f9777450d0427568084a538":{"kind":"REWORK","_number":6,"created":"2026-01-13 11:21:43.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/09/973209/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/09/973209/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/6"}}},"commit":{"parents":[{"commit":"cc77a164297c4d3f835f0a7b9569d54ac74ceaa4","subject":"Merge \"Rename \"Show Instance\" checkbox\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/cc77a164297c4d3f835f0a7b9569d54ac74ceaa4"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-13 09:23:29.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-13 11:21:39.000000000","tz":0},"subject":"Fix for writing Front-End / Back-End","message":"Fix for writing Front-End / Back-End\n\nAnd few small fixes.\n\nChange-Id: I4a87f64fea699b82d06d1e3982fc32d5a931a77d\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/9652a1b35421be5c3f9777450d0427568084a538"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/9652a1b35421be5c3f9777450d0427568084a538"}]},"branch":"refs/heads/master"},"bcc9f92cc7246aa51d44f4e56d83091ea4c91ac0":{"kind":"REWORK","_number":7,"created":"2026-01-13 11:26:12.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/09/973209/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/09/973209/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/7"}}},"commit":{"parents":[{"commit":"cc77a164297c4d3f835f0a7b9569d54ac74ceaa4","subject":"Merge \"Rename \"Show Instance\" checkbox\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/cc77a164297c4d3f835f0a7b9569d54ac74ceaa4"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-13 09:23:29.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-13 11:26:12.000000000","tz":0},"subject":"Fix for writing Front-End / Back-End","message":"Fix for writing Front-End / Back-End\n\nAnd few small fixes.\n\nChange-Id: I4a87f64fea699b82d06d1e3982fc32d5a931a77d\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/bcc9f92cc7246aa51d44f4e56d83091ea4c91ac0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/bcc9f92cc7246aa51d44f4e56d83091ea4c91ac0"}]},"branch":"refs/heads/master"},"98e8109e8d5ea5f778816b823e5f114ff4076d0e":{"kind":"REWORK","_number":8,"created":"2026-01-13 11:28:37.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/09/973209/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/09/973209/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/8"}}},"commit":{"parents":[{"commit":"cc77a164297c4d3f835f0a7b9569d54ac74ceaa4","subject":"Merge \"Rename \"Show Instance\" checkbox\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/cc77a164297c4d3f835f0a7b9569d54ac74ceaa4"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-13 09:23:29.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-13 11:28:37.000000000","tz":0},"subject":"Fix for writing Front-End / Back-End","message":"Fix for writing Front-End / Back-End\n\nAnd few small fixes.\n\nChange-Id: I4a87f64fea699b82d06d1e3982fc32d5a931a77d\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/98e8109e8d5ea5f778816b823e5f114ff4076d0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/98e8109e8d5ea5f778816b823e5f114ff4076d0e"}]},"branch":"refs/heads/master"},"48aa2ab1f0b16f77618333d60d02169cac2495c4":{"kind":"REWORK","_number":9,"created":"2026-01-13 11:36:40.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/09/973209/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/09/973209/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/9"}}},"commit":{"parents":[{"commit":"cc77a164297c4d3f835f0a7b9569d54ac74ceaa4","subject":"Merge \"Rename \"Show Instance\" checkbox\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/cc77a164297c4d3f835f0a7b9569d54ac74ceaa4"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-13 09:23:29.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-13 11:36:40.000000000","tz":0},"subject":"Fix for writing Front-End / Back-End","message":"Fix for writing Front-End / Back-End\n\nAnd few small fixes.\n\nChange-Id: I4a87f64fea699b82d06d1e3982fc32d5a931a77d\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/48aa2ab1f0b16f77618333d60d02169cac2495c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/48aa2ab1f0b16f77618333d60d02169cac2495c4"}]},"branch":"refs/heads/master"},"f3acf9b8303bb1950aec8ca628c03f09e5b1dc23":{"kind":"REWORK","_number":10,"created":"2026-01-13 11:41:33.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/09/973209/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/09/973209/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/10"}}},"commit":{"parents":[{"commit":"cc77a164297c4d3f835f0a7b9569d54ac74ceaa4","subject":"Merge \"Rename \"Show Instance\" checkbox\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/cc77a164297c4d3f835f0a7b9569d54ac74ceaa4"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-13 09:23:29.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-13 11:41:33.000000000","tz":0},"subject":"Fix for writing Front-End / Back-End","message":"Fix for writing Front-End / Back-End\n\nAnd few small fixes.\n\nChange-Id: I4a87f64fea699b82d06d1e3982fc32d5a931a77d\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f3acf9b8303bb1950aec8ca628c03f09e5b1dc23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f3acf9b8303bb1950aec8ca628c03f09e5b1dc23"}]},"branch":"refs/heads/master"},"dafa6fb9cb6bf317c0c3aefb150b41f423e31e37":{"kind":"REWORK","_number":11,"created":"2026-01-13 11:48:19.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/09/973209/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/09/973209/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/11"}}},"commit":{"parents":[{"commit":"cc77a164297c4d3f835f0a7b9569d54ac74ceaa4","subject":"Merge \"Rename \"Show Instance\" checkbox\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/cc77a164297c4d3f835f0a7b9569d54ac74ceaa4"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-13 09:23:29.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-13 11:48:18.000000000","tz":0},"subject":"Fix for writing Front-End / Back-End","message":"Fix for writing Front-End / Back-End\n\nAnd few small fixes.\n\nChange-Id: I4a87f64fea699b82d06d1e3982fc32d5a931a77d\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/dafa6fb9cb6bf317c0c3aefb150b41f423e31e37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/dafa6fb9cb6bf317c0c3aefb150b41f423e31e37"}]},"branch":"refs/heads/master"},"e92ca7861b4a0c6c769dfb60f489c5c7b042cac6":{"kind":"REWORK","_number":12,"created":"2026-01-13 11:49:25.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/09/973209/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/09/973209/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/12"}}},"commit":{"parents":[{"commit":"cc77a164297c4d3f835f0a7b9569d54ac74ceaa4","subject":"Merge \"Rename \"Show Instance\" checkbox\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/cc77a164297c4d3f835f0a7b9569d54ac74ceaa4"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-13 09:23:29.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-13 11:49:25.000000000","tz":0},"subject":"Fix for writing Front-End / Back-End","message":"Fix for writing Front-End / Back-End\n\nAnd few small fixes.\n\nChange-Id: I4a87f64fea699b82d06d1e3982fc32d5a931a77d\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e92ca7861b4a0c6c769dfb60f489c5c7b042cac6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e92ca7861b4a0c6c769dfb60f489c5c7b042cac6"}]},"branch":"refs/heads/master"},"6bef618f12e00ef8c18d280f9b19cb11b6d1c16c":{"kind":"REWORK","_number":13,"created":"2026-01-13 11:50:25.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/09/973209/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/09/973209/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/13"}}},"commit":{"parents":[{"commit":"cc77a164297c4d3f835f0a7b9569d54ac74ceaa4","subject":"Merge \"Rename \"Show Instance\" checkbox\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/cc77a164297c4d3f835f0a7b9569d54ac74ceaa4"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-13 09:23:29.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-13 11:50:25.000000000","tz":0},"subject":"Fix for writing Front-End / Back-End","message":"Fix for writing Front-End / Back-End\n\nAnd few small fixes.\n\nChange-Id: I4a87f64fea699b82d06d1e3982fc32d5a931a77d\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6bef618f12e00ef8c18d280f9b19cb11b6d1c16c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6bef618f12e00ef8c18d280f9b19cb11b6d1c16c"}]},"branch":"refs/heads/master"},"f42face2d2158d28786109b54a586b02e21f298f":{"kind":"REWORK","_number":14,"created":"2026-01-13 11:52:22.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/09/973209/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/09/973209/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/14"}}},"commit":{"parents":[{"commit":"cc77a164297c4d3f835f0a7b9569d54ac74ceaa4","subject":"Merge \"Rename \"Show Instance\" checkbox\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/cc77a164297c4d3f835f0a7b9569d54ac74ceaa4"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-13 09:23:29.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-13 11:52:22.000000000","tz":0},"subject":"Fix for writing Front-End / Back-End","message":"Fix for writing Front-End / Back-End\n\nAnd few small fixes.\n\nChange-Id: I4a87f64fea699b82d06d1e3982fc32d5a931a77d\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f42face2d2158d28786109b54a586b02e21f298f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f42face2d2158d28786109b54a586b02e21f298f"}]},"branch":"refs/heads/master"},"d360bb0f591a80cb651b1df52e5738facf201fd6":{"kind":"REWORK","_number":15,"created":"2026-01-13 11:54:07.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/09/973209/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/09/973209/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/15"}}},"commit":{"parents":[{"commit":"cc77a164297c4d3f835f0a7b9569d54ac74ceaa4","subject":"Merge \"Rename \"Show Instance\" checkbox\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/cc77a164297c4d3f835f0a7b9569d54ac74ceaa4"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-13 09:23:29.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-13 11:54:07.000000000","tz":0},"subject":"Fix for writing Front-End / Back-End","message":"Fix for writing Front-End / Back-End\n\nAnd few small fixes.\n\nChange-Id: I4a87f64fea699b82d06d1e3982fc32d5a931a77d\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d360bb0f591a80cb651b1df52e5738facf201fd6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d360bb0f591a80cb651b1df52e5738facf201fd6"}]},"branch":"refs/heads/master"},"aa02e092b340b8aba72efd1e558f93f3f425db1f":{"kind":"REWORK","_number":16,"created":"2026-01-13 11:58:46.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/09/973209/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/09/973209/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/16"}}},"commit":{"parents":[{"commit":"cc77a164297c4d3f835f0a7b9569d54ac74ceaa4","subject":"Merge \"Rename \"Show Instance\" checkbox\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/cc77a164297c4d3f835f0a7b9569d54ac74ceaa4"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-13 09:23:29.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-13 11:58:46.000000000","tz":0},"subject":"Fix for writing Front-End / Back-End","message":"Fix for writing Front-End / Back-End\n\nAnd few small fixes.\n\nChange-Id: I4a87f64fea699b82d06d1e3982fc32d5a931a77d\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/aa02e092b340b8aba72efd1e558f93f3f425db1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/aa02e092b340b8aba72efd1e558f93f3f425db1f"}]},"branch":"refs/heads/master"},"f0bc7800ef419605a279431ffc41e18343bf03c2":{"kind":"REWORK","_number":17,"created":"2026-01-13 12:03:02.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/09/973209/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/09/973209/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/17"}}},"commit":{"parents":[{"commit":"cc77a164297c4d3f835f0a7b9569d54ac74ceaa4","subject":"Merge \"Rename \"Show Instance\" checkbox\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/cc77a164297c4d3f835f0a7b9569d54ac74ceaa4"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-13 09:23:29.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-13 12:01:36.000000000","tz":0},"subject":"Fix for writing Front-End / Back-End","message":"Fix for writing Front-End / Back-End\n\nAnd few small fixes.\n\nChange-Id: I4a87f64fea699b82d06d1e3982fc32d5a931a77d\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f0bc7800ef419605a279431ffc41e18343bf03c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f0bc7800ef419605a279431ffc41e18343bf03c2"}]},"branch":"refs/heads/master"},"8d81bcd15b22a91ef99ae8768f697452b92d0e08":{"kind":"REWORK","_number":18,"created":"2026-01-13 12:32:23.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/09/973209/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/09/973209/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/18"}}},"commit":{"parents":[{"commit":"cc77a164297c4d3f835f0a7b9569d54ac74ceaa4","subject":"Merge \"Rename \"Show Instance\" checkbox\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/cc77a164297c4d3f835f0a7b9569d54ac74ceaa4"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-13 09:23:29.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-13 12:32:23.000000000","tz":0},"subject":"Fix for writing Front-End / Back-End","message":"Fix for writing Front-End / Back-End\n\nAnd few small fixes.\n\nChange-Id: I4a87f64fea699b82d06d1e3982fc32d5a931a77d\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/8d81bcd15b22a91ef99ae8768f697452b92d0e08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/8d81bcd15b22a91ef99ae8768f697452b92d0e08"}]},"branch":"refs/heads/master"},"f8a2f277207dba3cceb842aa2d77082fdbf7ec5c":{"kind":"NO_CODE_CHANGE","_number":19,"created":"2026-01-13 12:35:29.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/09/973209/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/09/973209/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/19"}}},"commit":{"parents":[{"commit":"cc77a164297c4d3f835f0a7b9569d54ac74ceaa4","subject":"Merge \"Rename \"Show Instance\" checkbox\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/cc77a164297c4d3f835f0a7b9569d54ac74ceaa4"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-13 09:23:29.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-13 12:35:29.000000000","tz":0},"subject":"Correct various grammar mistakes","message":"Correct various grammar mistakes\n\nChange-Id: I4a87f64fea699b82d06d1e3982fc32d5a931a77d\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f8a2f277207dba3cceb842aa2d77082fdbf7ec5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f8a2f277207dba3cceb842aa2d77082fdbf7ec5c"}]},"branch":"refs/heads/master","description":"Edit commit message"},"75c21b65628e03ddf1a84d09b68f4e9fd2135dce":{"kind":"REWORK","_number":20,"created":"2026-01-13 13:33:56.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/09/973209/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/09/973209/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/20"}}},"commit":{"parents":[{"commit":"cc77a164297c4d3f835f0a7b9569d54ac74ceaa4","subject":"Merge \"Rename \"Show Instance\" checkbox\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/cc77a164297c4d3f835f0a7b9569d54ac74ceaa4"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-13 09:23:29.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-13 13:33:56.000000000","tz":0},"subject":"Correct various grammar mistakes","message":"Correct various grammar mistakes\n\nChange-Id: I4a87f64fea699b82d06d1e3982fc32d5a931a77d\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/75c21b65628e03ddf1a84d09b68f4e9fd2135dce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/75c21b65628e03ddf1a84d09b68f4e9fd2135dce"}]},"branch":"refs/heads/master"},"7c228ed0b7fb108d8b3aa00ffa70e96f0dac3f2e":{"kind":"REWORK","_number":21,"created":"2026-01-13 13:37:47.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/09/973209/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/09/973209/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/21"}}},"commit":{"parents":[{"commit":"cc77a164297c4d3f835f0a7b9569d54ac74ceaa4","subject":"Merge \"Rename \"Show Instance\" checkbox\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/cc77a164297c4d3f835f0a7b9569d54ac74ceaa4"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-13 09:23:29.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-13 13:37:47.000000000","tz":0},"subject":"Correct various grammar mistakes","message":"Correct various grammar mistakes\n\nChange-Id: I4a87f64fea699b82d06d1e3982fc32d5a931a77d\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7c228ed0b7fb108d8b3aa00ffa70e96f0dac3f2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7c228ed0b7fb108d8b3aa00ffa70e96f0dac3f2e"}]},"branch":"refs/heads/master"},"11d55afba5f51370837817762a3533cef5ba2d16":{"kind":"NO_CODE_CHANGE","_number":22,"created":"2026-01-13 13:39:11.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/09/973209/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/09/973209/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/22"}}},"commit":{"parents":[{"commit":"cc77a164297c4d3f835f0a7b9569d54ac74ceaa4","subject":"Merge \"Rename \"Show Instance\" checkbox\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/cc77a164297c4d3f835f0a7b9569d54ac74ceaa4"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-13 09:23:29.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-13 13:39:11.000000000","tz":0},"subject":"Correct various grammar mistakes","message":"Correct various grammar mistakes\n\nChange-Id: I4a87f64fea699b82d06d1e3982fc32d5a931a77d\nCo-authored-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/11d55afba5f51370837817762a3533cef5ba2d16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/11d55afba5f51370837817762a3533cef5ba2d16"}]},"branch":"refs/heads/master","description":"Edit commit message"},"54721b094e5e1af1dbc5c1959a94dac114f97443":{"kind":"REWORK","_number":23,"created":"2026-01-13 13:56:06.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/09/973209/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/09/973209/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/23"}}},"commit":{"parents":[{"commit":"cc77a164297c4d3f835f0a7b9569d54ac74ceaa4","subject":"Merge \"Rename \"Show Instance\" checkbox\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/cc77a164297c4d3f835f0a7b9569d54ac74ceaa4"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-13 09:23:29.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-13 13:56:05.000000000","tz":0},"subject":"Correct various grammar mistakes","message":"Correct various grammar mistakes\n\nChange-Id: I4a87f64fea699b82d06d1e3982fc32d5a931a77d\nCo-authored-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/54721b094e5e1af1dbc5c1959a94dac114f97443"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/54721b094e5e1af1dbc5c1959a94dac114f97443"}]},"branch":"refs/heads/master"},"7802633ee1f8424511bca96d3c3c00992671ab46":{"kind":"REWORK","_number":24,"created":"2026-01-13 13:57:58.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/09/973209/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/09/973209/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/24"}}},"commit":{"parents":[{"commit":"cc77a164297c4d3f835f0a7b9569d54ac74ceaa4","subject":"Merge \"Rename \"Show Instance\" checkbox\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/cc77a164297c4d3f835f0a7b9569d54ac74ceaa4"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-13 09:23:29.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-13 13:57:58.000000000","tz":0},"subject":"Correct various grammar mistakes","message":"Correct various grammar mistakes\n\nChange-Id: I4a87f64fea699b82d06d1e3982fc32d5a931a77d\nCo-authored-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7802633ee1f8424511bca96d3c3c00992671ab46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7802633ee1f8424511bca96d3c3c00992671ab46"}]},"branch":"refs/heads/master"},"ac380e2bb478a364139cdc12f851b806ea06b054":{"kind":"REWORK","_number":25,"created":"2026-01-13 13:58:45.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/09/973209/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/09/973209/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/25"}}},"commit":{"parents":[{"commit":"cc77a164297c4d3f835f0a7b9569d54ac74ceaa4","subject":"Merge \"Rename \"Show Instance\" checkbox\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/cc77a164297c4d3f835f0a7b9569d54ac74ceaa4"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-13 09:23:29.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-13 13:58:45.000000000","tz":0},"subject":"Correct various grammar mistakes","message":"Correct various grammar mistakes\n\nChange-Id: I4a87f64fea699b82d06d1e3982fc32d5a931a77d\nCo-authored-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ac380e2bb478a364139cdc12f851b806ea06b054"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ac380e2bb478a364139cdc12f851b806ea06b054"}]},"branch":"refs/heads/master"},"25bd0c80fd8afd0f830961a280c7828e7888a583":{"kind":"REWORK","_number":26,"created":"2026-01-13 14:02:29.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/09/973209/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/09/973209/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/26"}}},"commit":{"parents":[{"commit":"cc77a164297c4d3f835f0a7b9569d54ac74ceaa4","subject":"Merge \"Rename \"Show Instance\" checkbox\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/cc77a164297c4d3f835f0a7b9569d54ac74ceaa4"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-13 09:23:29.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-13 14:02:29.000000000","tz":0},"subject":"Correct various grammar mistakes","message":"Correct various grammar mistakes\n\nChange-Id: I4a87f64fea699b82d06d1e3982fc32d5a931a77d\nCo-authored-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/25bd0c80fd8afd0f830961a280c7828e7888a583"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/25bd0c80fd8afd0f830961a280c7828e7888a583"}]},"branch":"refs/heads/master"},"6f252e7ad7380be94727f339fb89caa914b7c3df":{"kind":"REWORK","_number":27,"created":"2026-01-13 14:03:14.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/09/973209/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/09/973209/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/27"}}},"commit":{"parents":[{"commit":"cc77a164297c4d3f835f0a7b9569d54ac74ceaa4","subject":"Merge \"Rename \"Show Instance\" checkbox\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/cc77a164297c4d3f835f0a7b9569d54ac74ceaa4"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-13 09:23:29.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-13 14:03:13.000000000","tz":0},"subject":"Correct various grammar mistakes","message":"Correct various grammar mistakes\n\nChange-Id: I4a87f64fea699b82d06d1e3982fc32d5a931a77d\nCo-authored-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6f252e7ad7380be94727f339fb89caa914b7c3df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6f252e7ad7380be94727f339fb89caa914b7c3df"}]},"branch":"refs/heads/master"},"5f5135e5299e82a9131117a499e9067816a240ca":{"kind":"REWORK","_number":28,"created":"2026-01-13 14:04:23.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/09/973209/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/09/973209/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/28"}}},"commit":{"parents":[{"commit":"cc77a164297c4d3f835f0a7b9569d54ac74ceaa4","subject":"Merge \"Rename \"Show Instance\" checkbox\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/cc77a164297c4d3f835f0a7b9569d54ac74ceaa4"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-13 09:23:29.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-13 14:04:23.000000000","tz":0},"subject":"Correct various grammar mistakes","message":"Correct various grammar mistakes\n\nChange-Id: I4a87f64fea699b82d06d1e3982fc32d5a931a77d\nCo-authored-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5f5135e5299e82a9131117a499e9067816a240ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/5f5135e5299e82a9131117a499e9067816a240ca"}]},"branch":"refs/heads/master"},"9121dfeb5f8507ab995346b7043dd9190fdcf8cc":{"kind":"REWORK","_number":29,"created":"2026-01-13 23:14:20.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/09/973209/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/09/973209/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/29"}}},"commit":{"parents":[{"commit":"cc77a164297c4d3f835f0a7b9569d54ac74ceaa4","subject":"Merge \"Rename \"Show Instance\" checkbox\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/cc77a164297c4d3f835f0a7b9569d54ac74ceaa4"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-13 09:23:29.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-13 23:13:47.000000000","tz":0},"subject":"Correct various grammar mistakes","message":"Correct various grammar mistakes\n\nChange-Id: I4a87f64fea699b82d06d1e3982fc32d5a931a77d\nCo-authored-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/9121dfeb5f8507ab995346b7043dd9190fdcf8cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/9121dfeb5f8507ab995346b7043dd9190fdcf8cc"}]},"branch":"refs/heads/master"},"a144251d1a8d03773d77073e2b947a7d1d518ff1":{"kind":"REWORK","_number":30,"created":"2026-01-13 23:14:52.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/09/973209/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/09/973209/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/09/973209/30"}}},"commit":{"parents":[{"commit":"cc77a164297c4d3f835f0a7b9569d54ac74ceaa4","subject":"Merge \"Rename \"Show Instance\" checkbox\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/cc77a164297c4d3f835f0a7b9569d54ac74ceaa4"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-13 09:23:29.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-13 23:14:52.000000000","tz":0},"subject":"Correct various grammar mistakes","message":"Correct various grammar mistakes\n\nChange-Id: I4a87f64fea699b82d06d1e3982fc32d5a931a77d\nCo-authored-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a144251d1a8d03773d77073e2b947a7d1d518ff1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a144251d1a8d03773d77073e2b947a7d1d518ff1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
