)]}'
{"id":"openstack%2Fskyline-console~973256","triplet_id":"openstack%2Fskyline-console~master~I9acd36ccbdaedc2c70c3c5d8b999ae63b3d8744b","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-01-14 04:35:36.000000000","reason":"Change was submitted"},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-01-14 02:15:21.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-01-14 04:35:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9acd36ccbdaedc2c70c3c5d8b999ae63b3d8744b","subject":"Update User Password - names of fields on the form are now more precise","status":"MERGED","created":"2026-01-13 14:20:24.000000000","updated":"2026-01-14 04:37:11.000000000","submitted":"2026-01-14 04:35:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"973256","meta_rev_id":"da1a4973283f9c9459bd2923328392c67cb3df3a","_number":973256,"virtual_id_number":973256,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-14 04:35:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":2,"date":"2026-01-14 02:15:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":1,"date":"2026-01-14 02:15:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-13 14:23:14.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2026-01-13 14:23:14.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"},{"updated":"2026-01-13 15:56:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c2c3e2f2b2265bf0f2de85a6f504f7649f1ce0dc","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-13 14:20:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"57d5f5aaea6743346699efa278fb10bf841a5a43","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-13 14:22:25.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"8493cf31de4d9f6c5007eb57b8a99b71c9dbf7ca","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-13 14:23:05.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"247db4c04bd1c3c14140cd0962ab3d9d4ad742c7","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-13 14:23:14.000000000","message":"Patch Set 3:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"a3c1c0ed00524aa313b9436e76bc26da00fccd65","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-13 14:28:23.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"65f569e2deb22f1490e05688861183410f2fa326","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 15:56:02.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b317bf43caa40dc9fbf5f2216420ad9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea6c00806d9e4bbfbd35cec05de98999 : SUCCESS in 6m 41s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/9a8c4e0be66e4855bd1c54f066b73aed : SUCCESS in 8m 11s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/b5304224c3e4487baac63809ee9df10b : SUCCESS in 6m 09s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/91e2dafb77a2470b89468e3189c2dc31 : SUCCESS in 5m 07s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/5697ba59cf924c8db148423adff21368 : SUCCESS in 1h 17m 42s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/61f223f779f8426c870a6e73f20178ea : SUCCESS in 1h 19m 47s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/1399a056efbb4b4eba982c2840e99eeb : SUCCESS in 29m 16s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/703631f1f50548b596a007d9fce827e6 : SUCCESS in 58m 00s","accounts_in_message":[],"_revision_number":3},{"id":"f36fab96a05d31cb4959d058dd29463e18f37f46","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-13 23:29:32.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"83b1b3e2dcf51cfc8f8359a26edf6b6ea4d6870d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 00:56:16.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/babcabeafac647a4ad943bfe80eea499\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a86f6e2aa5a2488997aa82b760818002 : SUCCESS in 5m 40s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/50c1e8f30b454bc296c66fdc7cd8de0c : SUCCESS in 8m 52s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/711d2c5e19ee4cfcbb8343ff1a0d29e9 : SUCCESS in 7m 17s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/0947f6fa787b465e871e55a5a4b6afc7 : SUCCESS in 7m 26s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/1da56f1fc6ea4946b1d29a5f2faea367 : SUCCESS in 1h 21m 36s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/14227f9f16be43b88449166411530f08 : SUCCESS in 1h 00m 04s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/1117dc229b494215aa348087a525e16e : SUCCESS in 50m 19s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/5a5a521de9184c048d8e5c2445938fe2 : SUCCESS in 42m 50s","accounts_in_message":[],"_revision_number":4},{"id":"52a2fca113f9a7cb6844719fb40ed74ffff1848c","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-01-14 02:15:21.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"e6b49b8de0601b638bf80475dfc8d6a74c9653d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 02:16:40.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"24f56ed45f9bc9188a4cb46478e3d4367d9c071e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 04:35:35.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/24086334a93d451bafb9a71478e225e9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41096ec4f7d348dfac191f45657b2b5c : SUCCESS in 5m 57s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/8e157887bf8a48d3b53fbd7627f06a93 : SUCCESS in 9m 48s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/b70e43f72dcf424db46c197f58df3a47 : SUCCESS in 7m 41s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/b910e12e43164e5283eb4c38c135d534 : SUCCESS in 7m 09s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/184f521bb9a64c059d9df1a6f6ee4b7d : SUCCESS in 1h 23m 26s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/3bb44f9034d94c0aa22a68bb045e0fea : SUCCESS in 52m 30s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/0bbe9b9c0d8d43b0887b88dff9c6cc39 : SUCCESS in 40m 30s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/0d11a3aa7eb944bbba48723f818131fb : SUCCESS in 1h 00m 20s","accounts_in_message":[],"_revision_number":4},{"id":"373dd2c9a06d978d6e76aab51ff84a8c1fbe3171","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 04:35:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"da1a4973283f9c9459bd2923328392c67cb3df3a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 04:37:11.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32c622dec2cf4f96a785fbb102b73db4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2e8c2008f134fb8a00cb00198bccd56 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"68f170ac277576ac7174a60246a4739f2153147c","revisions":{"8b2dfee285116852535dbd32b30148ab4e8d8ab5":{"kind":"REWORK","_number":1,"created":"2026-01-13 14:20:24.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/56/973256/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/56/973256/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/56/973256/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/56/973256/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/56/973256/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/56/973256/1"}}},"commit":{"parents":[{"commit":"301b238b6c362d948f4b432595041b2115398c8c","subject":"feat: Add i18n in the pre-comiit hooks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/301b238b6c362d948f4b432595041b2115398c8c"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-13 14:20:24.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-13 14:20:24.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: I9acd36ccbdaedc2c70c3c5d8b999ae63b3d8744b\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/8b2dfee285116852535dbd32b30148ab4e8d8ab5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/8b2dfee285116852535dbd32b30148ab4e8d8ab5"}]},"branch":"refs/heads/master"},"a4ab2e32c581943552b8cb37d9b84cb299f957c5":{"kind":"REWORK","_number":2,"created":"2026-01-13 14:22:25.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/56/973256/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/56/973256/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/56/973256/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/56/973256/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/56/973256/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/56/973256/2"}}},"commit":{"parents":[{"commit":"301b238b6c362d948f4b432595041b2115398c8c","subject":"feat: Add i18n in the pre-comiit hooks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/301b238b6c362d948f4b432595041b2115398c8c"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-13 14:20:24.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-13 14:22:25.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: I9acd36ccbdaedc2c70c3c5d8b999ae63b3d8744b\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a4ab2e32c581943552b8cb37d9b84cb299f957c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a4ab2e32c581943552b8cb37d9b84cb299f957c5"}]},"branch":"refs/heads/master"},"e4ed1fcac28dd77c0e9cf88c7034a499b6357a72":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-01-13 14:23:05.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/56/973256/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/56/973256/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/56/973256/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/56/973256/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/56/973256/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/56/973256/3"}}},"commit":{"parents":[{"commit":"301b238b6c362d948f4b432595041b2115398c8c","subject":"feat: Add i18n in the pre-comiit hooks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/301b238b6c362d948f4b432595041b2115398c8c"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-13 14:20:24.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-13 14:23:05.000000000","tz":0},"subject":"Update User Password - names of fields on the form are now more precise","message":"Update User Password - names of fields on the form are now more precise\n\nChange-Id: I9acd36ccbdaedc2c70c3c5d8b999ae63b3d8744b\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e4ed1fcac28dd77c0e9cf88c7034a499b6357a72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e4ed1fcac28dd77c0e9cf88c7034a499b6357a72"}]},"branch":"refs/heads/master","description":"Edit commit message"},"68f170ac277576ac7174a60246a4739f2153147c":{"kind":"REWORK","_number":4,"created":"2026-01-13 23:29:32.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/56/973256/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/56/973256/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/56/973256/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/56/973256/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/56/973256/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/56/973256/4"}}},"commit":{"parents":[{"commit":"301b238b6c362d948f4b432595041b2115398c8c","subject":"feat: Add i18n in the pre-comiit hooks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/301b238b6c362d948f4b432595041b2115398c8c"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-13 14:20:24.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-13 23:29:32.000000000","tz":0},"subject":"Update User Password - names of fields on the form are now more precise","message":"Update User Password - names of fields on the form are now more precise\n\nChange-Id: I9acd36ccbdaedc2c70c3c5d8b999ae63b3d8744b\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/68f170ac277576ac7174a60246a4739f2153147c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/68f170ac277576ac7174a60246a4739f2153147c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
