)]}'
{"id":"openstack%2Fskyline-console~973258","triplet_id":"openstack%2Fskyline-console~master~I4832bbe4c08f1ff5aec21b2d16ca5f9f1b54439c","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-01-14 02:27:29.000000000","reason":"Change was submitted"},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-01-14 01:55:10.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}},"hashtags":[],"change_id":"I4832bbe4c08f1ff5aec21b2d16ca5f9f1b54439c","subject":"Rename \"Port Security Enabled\" to \"Port Security\"","status":"MERGED","created":"2026-01-13 14:39:47.000000000","updated":"2026-01-14 02:28:49.000000000","submitted":"2026-01-14 02:27:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"973258","meta_rev_id":"b967c47c715eced311ee8166bba97eee72ff74f5","_number":973258,"virtual_id_number":973258,"owner":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-14 02:27:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-01-14 01:55:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-01-13 23:13:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-01-14 01:55:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-13 15:41:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-13 23:13:10.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-01-13 23:13:10.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"}],"messages":[{"id":"d9c3cfa88cf4bca1de4e2fab88d5f38460cef8c1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-13 14:39:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"da102e3e954663ca879be1f1f824713e7a1cae16","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-13 14:42:45.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"80cf506dc867ad933c245433e540cfea9d55cfe5","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-13 14:42:47.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"e922dd7b4b26746ab6d94a5d5c9306cb0eadbde4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-13 14:44:25.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"257e5e2d1f4001844fb8558c9e219b269cef9239","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 15:41:11.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c824fced64843aa90fd02570cd6685b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d44dce055e143b1aa71269a6913b983 : SUCCESS in 3m 32s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/4804e5dbe83e4a4ba3be0f3cad2befbd : SUCCESS in 11m 01s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/94dbe3711c3a4b4cb5f7051fd42f785e : SUCCESS in 3m 13s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/b7aba5805ed94d3496532fe979d8e208 : SUCCESS in 8m 12s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/6bf8e91aa997461fa99b8dcd93050ba7 : SUCCESS in 55m 08s","accounts_in_message":[],"_revision_number":3},{"id":"99976d5f835bf6f94e40b68838ce662deb0d08ba","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-13 23:13:10.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"4e5da168bc6064353dcd3cee6d1b8fbbad95d562","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-01-14 01:55:10.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"d5d18955148bf8d60ee5ec34e67b4543035fa74e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 01:55:34.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0ad33a87b6767322cf6f54fa2e0feb3cf7ef94d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 02:27:29.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4bdf208f35e40efaab71adbf9e125c6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d461d80d718247f18a4565352641347f : SUCCESS in 5m 11s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/defba07014a54e269da1151d0a58723a : SUCCESS in 4m 52s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/ef15f00cdc894ee6847ca929cd1f166f : SUCCESS in 4m 24s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/755e952c30924d368b355b8a3b613c34 : SUCCESS in 3m 11s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/c035fff6414645fd8e3ee7593643352c : SUCCESS in 31m 19s","accounts_in_message":[],"_revision_number":3},{"id":"817161a4ed5093e35572e6a0807a85f53ec93796","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 02:27:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"b967c47c715eced311ee8166bba97eee72ff74f5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 02:28:49.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e1f24cbe55144c286c67425333c7c52\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a0a92768cba41629d033dee6dd8b0c9 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"06fb626c5c27dd43f08b197b73b1188d9124d390","revisions":{"bf464cd65ef4294a76722498b8128469537c700b":{"kind":"REWORK","_number":1,"created":"2026-01-13 14:39:47.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/58/973258/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/58/973258/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/58/973258/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/58/973258/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/58/973258/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/58/973258/1"}}},"commit":{"parents":[{"commit":"301b238b6c362d948f4b432595041b2115398c8c","subject":"feat: Add i18n in the pre-comiit hooks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/301b238b6c362d948f4b432595041b2115398c8c"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-13 14:39:47.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-13 14:39:47.000000000","tz":0},"subject":"Rename \"Port Security Enabled\" to \"Port Security\"","message":"Rename \"Port Security Enabled\" to \"Port Security\"\n\nRemoves redundant word \"Enabled\" since the field is represented as a toggle and already implies on/off behavior.\n\nChange-Id: I4832bbe4c08f1ff5aec21b2d16ca5f9f1b54439c\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/bf464cd65ef4294a76722498b8128469537c700b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/bf464cd65ef4294a76722498b8128469537c700b"}]},"branch":"refs/heads/master"},"71df01418bee5d87123b440ec08347c0850a1ef2":{"kind":"REWORK","_number":2,"created":"2026-01-13 14:42:45.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/58/973258/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/58/973258/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/58/973258/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/58/973258/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/58/973258/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/58/973258/2"}}},"commit":{"parents":[{"commit":"301b238b6c362d948f4b432595041b2115398c8c","subject":"feat: Add i18n in the pre-comiit hooks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/301b238b6c362d948f4b432595041b2115398c8c"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-13 14:39:47.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-13 14:42:26.000000000","tz":0},"subject":"Rename \"Port Security Enabled\" to \"Port Security\"","message":"Rename \"Port Security Enabled\" to \"Port Security\"\n\nRemoves redundant word \"Enabled\" since the field is represented as a toggle and already implies on/off behavior.\n\nChange-Id: I4832bbe4c08f1ff5aec21b2d16ca5f9f1b54439c\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/71df01418bee5d87123b440ec08347c0850a1ef2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/71df01418bee5d87123b440ec08347c0850a1ef2"}]},"branch":"refs/heads/master"},"06fb626c5c27dd43f08b197b73b1188d9124d390":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-01-13 14:44:25.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/58/973258/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/58/973258/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/58/973258/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/58/973258/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/58/973258/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/58/973258/3"}}},"commit":{"parents":[{"commit":"301b238b6c362d948f4b432595041b2115398c8c","subject":"feat: Add i18n in the pre-comiit hooks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/301b238b6c362d948f4b432595041b2115398c8c"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-13 14:39:47.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-13 14:44:25.000000000","tz":0},"subject":"Rename \"Port Security Enabled\" to \"Port Security\"","message":"Rename \"Port Security Enabled\" to \"Port Security\"\n\nRemove redundant word \"Enabled\" since the field is represented as a toggle and already implies on/off behavior.\n\nChange-Id: I4832bbe4c08f1ff5aec21b2d16ca5f9f1b54439c\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/06fb626c5c27dd43f08b197b73b1188d9124d390"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/06fb626c5c27dd43f08b197b73b1188d9124d390"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
