)]}'
{"id":"openstack%2Fskyline-console~973269","triplet_id":"openstack%2Fskyline-console~master~Id3a3bcb2d71249f2ed427f40e11197b0c2e611ce","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-01-14 03:53:59.000000000","reason":"Change was submitted"},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-01-14 02:03:22.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}},"hashtags":[],"change_id":"Id3a3bcb2d71249f2ed427f40e11197b0c2e611ce","subject":"Update welcome page","status":"MERGED","created":"2026-01-13 16:05:48.000000000","updated":"2026-01-14 03:55:10.000000000","submitted":"2026-01-14 03:53:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"973269","meta_rev_id":"fca9e4bc572a6e1dc4e2a2a72075a9d592eef319","_number":973269,"virtual_id_number":973269,"owner":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-14 03:53:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-01-14 02:03:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-01-13 23:17:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-01-14 02:03:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-13 17:24:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-13 23:17:21.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-01-13 23:17:21.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"}],"messages":[{"id":"271eda7e38aa38c651ea1c41eebde75982d8ad48","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-13 16:05:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"720b63a52178295a499cd42b392bea3ad0afe922","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-13 16:11:54.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"bc432fe45df6742a5a12f3ace956daa6520e7a6c","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-13 16:11:58.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"58ddd1f4b2f089adc7ecfddab31fb6e50c5279a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 17:24:03.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/aad00e15d2ae4a2abb98d5f7496f8ecb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8581f86585d44bacbca5f8e940905222 : SUCCESS in 6m 28s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/31f6db1b00d341a5b7b8a81cddfe1dfc : SUCCESS in 11m 03s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/9ebd9ed8c81549fd8e9d9dcb6c7dede6 : SUCCESS in 7m 29s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/c0b47529746d459d949c65181a5dd850 : SUCCESS in 4m 03s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/3461b9cc959e480099d2ca62c16267f9 : FAILURE in 1h 07m 00s","accounts_in_message":[],"_revision_number":2},{"id":"1dc0db2c8ad90a16249b44314291eb57773f91db","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-13 20:00:03.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"50564456efb3b032e2d4d34c3d3bbbeac57436dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 20:17:10.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ba24521b3bbe428fb97256f62b4a2d66\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71758eb9a7de4596a1b4dbd386b79a38 : RETRY_LIMIT in 1m 35s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/cc18b2330a95460285e22738f48b8c4e : RETRY_LIMIT in 1m 10s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/307b911fe36743278320aba37fc92d30 : RETRY_LIMIT in 1m 36s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/66d4512b62d144de98a80f5b754ca1b7 : RETRY_LIMIT in 2m 20s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/29f29b81310442bc96b6cbc56dd30530 : RETRY_LIMIT in 2m 21s","accounts_in_message":[],"_revision_number":2},{"id":"ae5d6666dcd84c8b92f245b1de75e127677dc5f6","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-13 20:27:20.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1616ba469d61f38432a59b2106900f0aefac4fa1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 20:38:28.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/60f01f7a36764a99b9795668d2d964fa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85782e069aa04097a48e9d149f769e20 : RETRY_LIMIT in 1m 09s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/9b3ed09c4f3640a8ae876cb42aa5ef5f : RETRY_LIMIT in 51s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/04ee77e257304f06a825de3f05e62b43 : RETRY_LIMIT in 1m 08s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/8b8f027cb58f4c5ca3d4374dba047bd2 : RETRY_LIMIT in 1m 11s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/69deb348fc7443fd929e065dea14a199 : RETRY_LIMIT in 1m 37s","accounts_in_message":[],"_revision_number":2},{"id":"ccd1194cfe1d2e20c99bb1cc6ebfa3429001cfd7","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-13 22:23:55.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a345c27c99c3b582fc680c3561c97e184d53f494","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 23:03:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5cf2b946ea7b40a49939ccf42aaf41dc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbd552b2082044fea9b244dfb3f43aca : SUCCESS in 6m 30s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/e364bdd3965b40d5842ab718123d87c4 : SUCCESS in 9m 40s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/3e234b63d7cc46169e5ea905cef1e4dd : SUCCESS in 3m 44s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/d66328a1dd78437d8921aa76be4bd693 : SUCCESS in 7m 55s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/f371e423f13e4a85b1664b04224a0e05 : SUCCESS in 38m 34s","accounts_in_message":[],"_revision_number":2},{"id":"3744a30939de27008c9ae3f06c716a21b6fe0b40","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-13 23:17:21.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"7dfc084a0317b8bece55a2fcfe092a10cad6bfc8","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-01-14 02:03:22.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ed0d66c854cb3bf8396d6c3c5b74dbe0a43bc0ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 02:16:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5cd2f8a1c27d3a997ee9f30e359a27d5f5be3275","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 03:53:59.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f599408cae0441597e59bac1d0c4112\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52ad23d3007f4951a0e1f49cc7e507f6 : SUCCESS in 5m 41s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/af4c1861b110439891bf20008ef58a23 : SUCCESS in 4m 59s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/91a3cefb920643d8a4444fbb8f0e19bd : SUCCESS in 7m 00s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/1540a5b91aaf43d9ae0c57915327b7af : SUCCESS in 7m 31s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/50701feac48e4e178960bced0ac58bee : SUCCESS in 51m 13s","accounts_in_message":[],"_revision_number":2},{"id":"628fecbdadabf5fbe224f3e50565b43b9a3ffcd3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 03:53:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"fca9e4bc572a6e1dc4e2a2a72075a9d592eef319","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 03:55:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f944310688f04d0e8f17ab38765e479e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2340016bb174943a4fc01defa33c49d : SUCCESS in 59s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8822e65294ff78d572fc68e8ddb5ec7d0dfbd35e","revisions":{"c821e232517bc9995f570461501e23e6229c05ea":{"kind":"REWORK","_number":1,"created":"2026-01-13 16:05:48.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/69/973269/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/69/973269/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/69/973269/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/69/973269/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/69/973269/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/69/973269/1"}}},"commit":{"parents":[{"commit":"301b238b6c362d948f4b432595041b2115398c8c","subject":"feat: Add i18n in the pre-comiit hooks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/301b238b6c362d948f4b432595041b2115398c8c"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-13 16:05:48.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-13 16:05:48.000000000","tz":0},"subject":"Update welcome page","message":"Update welcome page\n\nModify welcome page to be grammatically more logic.\n\nChange-Id: Id3a3bcb2d71249f2ed427f40e11197b0c2e611ce\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/c821e232517bc9995f570461501e23e6229c05ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/c821e232517bc9995f570461501e23e6229c05ea"}]},"branch":"refs/heads/master"},"8822e65294ff78d572fc68e8ddb5ec7d0dfbd35e":{"kind":"REWORK","_number":2,"created":"2026-01-13 16:11:54.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/69/973269/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/69/973269/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/69/973269/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/69/973269/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/69/973269/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/69/973269/2"}}},"commit":{"parents":[{"commit":"301b238b6c362d948f4b432595041b2115398c8c","subject":"feat: Add i18n in the pre-comiit hooks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/301b238b6c362d948f4b432595041b2115398c8c"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-13 16:05:48.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-13 16:07:23.000000000","tz":0},"subject":"Update welcome page","message":"Update welcome page\n\nModify welcome page to be grammatically more logic.\n\nChange-Id: Id3a3bcb2d71249f2ed427f40e11197b0c2e611ce\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/8822e65294ff78d572fc68e8ddb5ec7d0dfbd35e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/8822e65294ff78d572fc68e8ddb5ec7d0dfbd35e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
