)]}'
{"id":"openstack%2Fskyline-console~973314","triplet_id":"openstack%2Fskyline-console~master~Icb3a9a310cb77568574f58bf8799c4099ba98868","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-01-14 04:35:34.000000000","reason":"Change was submitted"},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-01-14 02:14:13.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}},"hashtags":[],"change_id":"Icb3a9a310cb77568574f58bf8799c4099ba98868","subject":"Rename \"Owned Project\" Field","status":"MERGED","created":"2026-01-13 22:38:28.000000000","updated":"2026-01-14 04:36:27.000000000","submitted":"2026-01-14 04:35:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"973314","meta_rev_id":"12a5f935eb5b290b032647229ae68a13ba7bff75","_number":973314,"virtual_id_number":973314,"owner":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-14 04:35:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-01-14 02:14:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-01-13 23:15:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-01-14 02:14:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-13 23:15:47.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-01-13 23:15:47.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2026-01-13 23:49:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bc85772ba87a7e631443a620ac51491b1fd56363","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-13 22:38:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4bc148fee5a9dadcca04b11ecb1c7e3ece397e00","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-13 22:40:19.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"5660d3bb84d87d7064898a4c7083b0c721cfabb5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-13 22:40:25.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"bba7a66ec74b63c5f717d69061d39b33c3d6fdf0","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-13 22:40:33.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":3},{"id":"2a2060439f0965c902903e4f7f001ff3b1011217","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-13 23:15:47.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e9bb6e24ff50838967e6f298c213462c256b3c9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-13 23:49:47.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/256e11f3b0b943e58c4786b7324b719b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e42ab0965264d8ab6975c940b7ae867 : SUCCESS in 5m 38s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/270159e70b064b1d94829e09e851f4df : SUCCESS in 6m 51s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/bfb048d905614097b3088596baaf0c72 : SUCCESS in 4m 28s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/4d0891bc00fb4a36bcc38bc7241871f9 : SUCCESS in 4m 36s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/4359a74909284c82ad7ec780c86faf92 : SUCCESS in 1h 07m 58s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/fd4e99861b4745858f33327d180e1e5b : SUCCESS in 30m 34s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/aed6b821b2814a1f930636521668aa1d : SUCCESS in 31m 09s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/d6a945034fad47eb97d3467b0f356a2b : SUCCESS in 38m 24s","accounts_in_message":[],"_revision_number":3},{"id":"b9e0034ec1d40b3005aaa0183e9595fac0bc18df","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-01-14 02:14:13.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"d894bc3d88710a65bbc5d9e5b7f78c7b1360c072","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 02:16:39.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"918824e51ddd00c02c54a472ba34f6c4a44b53de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 04:35:33.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e8849431a874f35acd6f32f1576df32\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd259700f5fd4c0e9c722e2d7fe07838 : SUCCESS in 6m 17s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/d94422b97a1f4f268188918e8360dd03 : SUCCESS in 5m 07s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/7a39a7239f114aa9a85ac2e9a4524436 : SUCCESS in 8m 52s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/79f968e3d43f4d79ade7405903b7e458 : SUCCESS in 6m 49s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/34bb77466e14407a9b3cea1aebb7e08c : SUCCESS in 1h 28m 48s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/3da2667c807f4e658515fa9a233047b6 : SUCCESS in 59m 26s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/6e863b4fa9da469699c915b15b8dfbca : SUCCESS in 52m 07s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/0e67baf6be904b88a8cba54d1a94c90c : SUCCESS in 48m 00s","accounts_in_message":[],"_revision_number":3},{"id":"29bfcd54427aa6aa579ad3b9597179ba2f4871d3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 04:35:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"12a5f935eb5b290b032647229ae68a13ba7bff75","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 04:36:27.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9859964b292f408db323961df668c4b0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f650395777b4fa59f1365b4a73fa972 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e021ab86a8172f712b4da75e9ef61cee001dd11b","revisions":{"cf5127066afe982b54adbc4ca762f956a4fccf5b":{"kind":"REWORK","_number":1,"created":"2026-01-13 22:38:28.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/14/973314/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/14/973314/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/14/973314/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/14/973314/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/14/973314/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/14/973314/1"}}},"commit":{"parents":[{"commit":"301b238b6c362d948f4b432595041b2115398c8c","subject":"feat: Add i18n in the pre-comiit hooks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/301b238b6c362d948f4b432595041b2115398c8c"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-13 22:38:28.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-13 22:38:28.000000000","tz":0},"subject":"Rename \"Owned Project\u0027\" Field","message":"Rename \"Owned Project\u0027\" Field\n\nSimplify \"Owned Project\u0027 to “Project” to make it\nclearer and easier to understand.\n\nChange-Id: Icb3a9a310cb77568574f58bf8799c4099ba98868\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/cf5127066afe982b54adbc4ca762f956a4fccf5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/cf5127066afe982b54adbc4ca762f956a4fccf5b"}]},"branch":"refs/heads/master"},"b74f211cee00d783915b9602cb30d1279a943d98":{"kind":"REWORK","_number":2,"created":"2026-01-13 22:40:19.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/14/973314/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/14/973314/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/14/973314/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/14/973314/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/14/973314/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/14/973314/2"}}},"commit":{"parents":[{"commit":"301b238b6c362d948f4b432595041b2115398c8c","subject":"feat: Add i18n in the pre-comiit hooks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/301b238b6c362d948f4b432595041b2115398c8c"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-13 22:38:28.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-13 22:39:52.000000000","tz":0},"subject":"Rename \"Owned Project\u0027\" Field","message":"Rename \"Owned Project\u0027\" Field\n\nSimplify \"Owned Project\u0027 to “Project” to make it\nclearer and easier to understand.\n\nChange-Id: Icb3a9a310cb77568574f58bf8799c4099ba98868\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b74f211cee00d783915b9602cb30d1279a943d98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b74f211cee00d783915b9602cb30d1279a943d98"}]},"branch":"refs/heads/master"},"e021ab86a8172f712b4da75e9ef61cee001dd11b":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-01-13 22:40:25.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/14/973314/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/14/973314/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/14/973314/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/14/973314/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/14/973314/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/14/973314/3"}}},"commit":{"parents":[{"commit":"301b238b6c362d948f4b432595041b2115398c8c","subject":"feat: Add i18n in the pre-comiit hooks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/301b238b6c362d948f4b432595041b2115398c8c"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-13 22:38:28.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-13 22:40:25.000000000","tz":0},"subject":"Rename \"Owned Project\" Field","message":"Rename \"Owned Project\" Field\n\nSimplify \"Owned Project\u0027 to “Project” to make it\nclearer and easier to understand.\n\nChange-Id: Icb3a9a310cb77568574f58bf8799c4099ba98868\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e021ab86a8172f712b4da75e9ef61cee001dd11b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e021ab86a8172f712b4da75e9ef61cee001dd11b"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
