)]}'
{"id":"openstack%2Fskyline-console~973336","triplet_id":"openstack%2Fskyline-console~master~I1bd9a6d8e42fb79d1ba8fdb02e69e06ed3f11dae","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-01-16 11:42:59.000000000","reason":"\u003cGERRIT_ACCOUNT_37632\u003e replied on the change","reason_account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-01-16 02:22:21.000000000","reason":"removed on reply"},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-01-16 13:32:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1bd9a6d8e42fb79d1ba8fdb02e69e06ed3f11dae","subject":"Remove duplicate policy","status":"MERGED","created":"2026-01-14 01:29:05.000000000","updated":"2026-01-16 13:33:43.000000000","submitted":"2026-01-16 13:32:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"973336","meta_rev_id":"e57237ffa70be753207151b198625780c81360b8","_number":973336,"virtual_id_number":973336,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-16 13:32:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-01-16 02:22:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-01-16 02:22:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-14 01:31:32.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"},{"updated":"2026-01-14 01:31:32.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2026-01-14 02:29:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"09ce7b2e33699fd8a9092573bfef8fe6b1eb3921","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-14 01:29:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f2bc6bda69b520faaed39482e52603660979c8df","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-14 01:30:38.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"71dd25be1908abdd677736f7b59c36af1648799e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-14 01:31:11.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"79857e8cd134861ddc582207b74870eeb4d6ad1d","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-14 01:31:32.000000000","message":"Patch Set 3:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"af04ee8191e800133eb7e80904df8af1aea1360f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 02:29:36.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/847bbb8994ae4d539ea09e2aea223b8a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7e4fcdda7c44df48ece401dc85e4eda : SUCCESS in 6m 25s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/544434fd83e7462da92a605ba6d7d78e : SUCCESS in 6m 40s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/dfe56067bcb64e40aa2abd15b6640281 : SUCCESS in 7m 45s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/6d0375ccb08543b39bfca84956f69f55 : SUCCESS in 3m 37s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/bf959eac7b074f8a84a3d466c2452e17 : SUCCESS in 55m 06s","accounts_in_message":[],"_revision_number":3},{"id":"bc3e97e79341a23122bd083242d3754a163f448d","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-01-16 02:22:21.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"d3ada264917f9fdc3fbd23da1be23fdfaa642467","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-01-16 02:22:27.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"1e101b24bf78682f4ba60d7b2d4f9f3bed04d2bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 02:22:48.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"aac4ba59f8923d3c718b0ed3c90bd0e4661d14fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 03:14:02.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/540f3daed8a2424daad7cf8b76064fb3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edd519fc693f47d79c3c1495104abbf7 : SUCCESS in 4m 43s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/c5826b457c4e4dbe9e0c23c6358bc9c4 : SUCCESS in 8m 43s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/c1cb4ced933b4aabb5ae2669e35cac40 : SUCCESS in 5m 20s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/c4453ec50cd44dffa17c96870c706bd9 : SUCCESS in 3m 36s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/98a0aa16be7f4a80a6ec3a6071261a0a : FAILURE in 42m 54s","accounts_in_message":[],"_revision_number":3},{"id":"483b44ce382575884727643c995837b63b92fbf0","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-16 11:42:59.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2a2d78d21ee1b4a9f6821c97eedeb1b6f9d6e457","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 12:41:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/71a1ff98654648aa888b6a0e4b75f0d6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec5f76ecb8ed4d2a81c6fedb0e3c1be2 : SUCCESS in 6m 12s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/309475ee020447a18827707877a982c7 : SUCCESS in 11m 11s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/7df314aeef204f84af39b0e67cfeaa8d : SUCCESS in 8m 00s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/a35fd2df49dc4b9d83176df60fd747d6 : SUCCESS in 4m 52s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/b66888e799c54d94bdba6dfb465f0a92 : SUCCESS in 50m 56s","accounts_in_message":[],"_revision_number":3},{"id":"834d8373121b046d15ab6d6f72e60924d8a0fbec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 12:42:12.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e1e4bbcd2ca2d71d106286058f92b463762d53e0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 13:32:46.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d11bcdce616454daad99f52b1b61121\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9c5fc9031604bc9a9cc9738eae41fe4 : SUCCESS in 6m 48s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/b194ab5a803246db9395879433924796 : SUCCESS in 8m 26s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/bb2a1a5b10f84e0aa916f9be8a4bd55f : SUCCESS in 7m 28s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/a16ccc9d717f4aff9f457a87ed06eb7c : SUCCESS in 4m 19s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/7de733de956c4cd39d17835dc8bb6d17 : SUCCESS in 49m 10s","accounts_in_message":[],"_revision_number":3},{"id":"6b85ff7987b9101440515ca467503a2871c4d8c8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 13:32:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"e57237ffa70be753207151b198625780c81360b8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 13:33:43.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/622f9f07922d4533a7742a39802dd4d4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/625d0cc600f54da5ac5ad4aff79c116e : SUCCESS in 45s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"512a2ed2cd5293bd976881aae1211d85eea7ace6","revisions":{"9c64401662562c5f84a3412de341b1e3d13a7cc3":{"kind":"REWORK","_number":1,"created":"2026-01-14 01:29:05.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/36/973336/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/36/973336/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/36/973336/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/36/973336/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/36/973336/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/36/973336/1"}}},"commit":{"parents":[{"commit":"301b238b6c362d948f4b432595041b2115398c8c","subject":"feat: Add i18n in the pre-comiit hooks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/301b238b6c362d948f4b432595041b2115398c8c"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-14 01:29:05.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-14 01:29:05.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: I1bd9a6d8e42fb79d1ba8fdb02e69e06ed3f11dae\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/9c64401662562c5f84a3412de341b1e3d13a7cc3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/9c64401662562c5f84a3412de341b1e3d13a7cc3"}]},"branch":"refs/heads/master"},"35a8c5960f336f61b07c92eb6d950eae17e36247":{"kind":"REWORK","_number":2,"created":"2026-01-14 01:30:38.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/36/973336/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/36/973336/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/36/973336/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/36/973336/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/36/973336/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/36/973336/2"}}},"commit":{"parents":[{"commit":"301b238b6c362d948f4b432595041b2115398c8c","subject":"feat: Add i18n in the pre-comiit hooks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/301b238b6c362d948f4b432595041b2115398c8c"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-14 01:29:05.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-14 01:30:38.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: I1bd9a6d8e42fb79d1ba8fdb02e69e06ed3f11dae\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/35a8c5960f336f61b07c92eb6d950eae17e36247"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/35a8c5960f336f61b07c92eb6d950eae17e36247"}]},"branch":"refs/heads/master"},"512a2ed2cd5293bd976881aae1211d85eea7ace6":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-01-14 01:31:11.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/36/973336/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/36/973336/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/36/973336/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/36/973336/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/36/973336/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/36/973336/3"}}},"commit":{"parents":[{"commit":"301b238b6c362d948f4b432595041b2115398c8c","subject":"feat: Add i18n in the pre-comiit hooks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/301b238b6c362d948f4b432595041b2115398c8c"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-14 01:29:05.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-14 01:31:11.000000000","tz":0},"subject":"Remove duplicate policy","message":"Remove duplicate policy\n\nChange-Id: I1bd9a6d8e42fb79d1ba8fdb02e69e06ed3f11dae\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/512a2ed2cd5293bd976881aae1211d85eea7ace6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/512a2ed2cd5293bd976881aae1211d85eea7ace6"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
