)]}'
{"id":"openstack%2Fskyline-console~973351","triplet_id":"openstack%2Fskyline-console~master~I5704697f2c8bd0321014844cadea202911ec291c","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"20865":{"account":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"last_update":"2026-01-22 03:57:51.000000000","reason":"Change was submitted"},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-01-22 02:24:19.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"7556":{"account":{"_account_id":7556,"name":"Zuul (dev)","email":"zuul-dev@openstack.org","username":"zuul-dev"},"last_update":"2026-01-22 03:57:51.000000000","reason":"Change was submitted"},"36902":{"account":{"_account_id":36902,"name":"Nevra Gürses Yakar","display_name":"nevra","email":"nevra.gurses@tubitak.gov.tr","username":"nevragurses"},"last_update":"2026-01-22 03:57:51.000000000","reason":"Change was submitted"},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-01-22 03:57:51.000000000","reason":"Change was submitted"},"34671":{"account":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"last_update":"2026-01-22 03:57:51.000000000","reason":"Change was submitted"}},"hashtags":["CVE-2026-40212"],"change_id":"I5704697f2c8bd0321014844cadea202911ec291c","subject":"Update: Instance detail view full log opening on url and re-fetches log data on page reload.","status":"MERGED","created":"2026-01-14 07:26:55.000000000","updated":"2026-04-10 17:08:10.000000000","submitted":"2026-01-22 03:57:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"973351","meta_rev_id":"803af02ede28292879efb615953892b9c171cf49","_number":973351,"virtual_id_number":973351,"owner":{"_account_id":36902,"name":"Nevra Gürses Yakar","display_name":"nevra","email":"nevra.gurses@tubitak.gov.tr","username":"nevragurses"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-22 03:57:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-01-22 02:24:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-01-22 10:33:45.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-01-22 02:24:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},{"_account_id":7556,"name":"Zuul (dev)","email":"zuul-dev@openstack.org","username":"zuul-dev"},{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"}],"reviewers":{"REVIEWER":[{"_account_id":7556,"name":"Zuul (dev)","email":"zuul-dev@openstack.org","username":"zuul-dev"},{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-14 09:18:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-14 11:40:33.000000000","updated_by":{"_account_id":36902,"name":"Nevra Gürses Yakar","display_name":"nevra","email":"nevra.gurses@tubitak.gov.tr","username":"nevragurses"},"reviewer":{"_account_id":7556,"name":"Zuul (dev)","email":"zuul-dev@openstack.org","username":"zuul-dev"},"state":"REVIEWER"},{"updated":"2026-01-14 11:40:33.000000000","updated_by":{"_account_id":36902,"name":"Nevra Gürses Yakar","display_name":"nevra","email":"nevra.gurses@tubitak.gov.tr","username":"nevragurses"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2026-01-14 11:40:33.000000000","updated_by":{"_account_id":36902,"name":"Nevra Gürses Yakar","display_name":"nevra","email":"nevra.gurses@tubitak.gov.tr","username":"nevragurses"},"reviewer":{"_account_id":34671,"name":"Reşit Demir","email":"resit.demir@tubitak.gov.tr","username":"resitdemir"},"state":"REVIEWER"},{"updated":"2026-01-14 11:40:33.000000000","updated_by":{"_account_id":36902,"name":"Nevra Gürses Yakar","display_name":"nevra","email":"nevra.gurses@tubitak.gov.tr","username":"nevragurses"},"reviewer":{"_account_id":20865,"name":"Gökhan","email":"gokhan.isik@tubitak.gov.tr","username":"gokhan.isik"},"state":"REVIEWER"},{"updated":"2026-01-14 15:41:00.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2026-01-22 10:33:45.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"cb5c4bf9dd77660ce72242a79be52e5d0986517f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36902,"name":"Nevra Gürses Yakar","display_name":"nevra","email":"nevra.gurses@tubitak.gov.tr","username":"nevragurses"},"date":"2026-01-14 07:26:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9d459485a1935ad58d935a52edcdef9d1af04334","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 09:18:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/492240b7e54648e48ee1fc49f4c1e5ad\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/747df2d9951841e28d67c0b2588ea54d : SUCCESS in 2m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ed3bb19207f4e9685930d34c18c6e40 : SUCCESS in 8m 13s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/5437ac2065e342c6bb0c75c059467a01 : SUCCESS in 6m 14s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/a3cef8f3f2634c19be42e2a22d0abdb5 : SUCCESS in 7m 39s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/13c896e846c5402cbe952d5ef0fe592a : SUCCESS in 7m 20s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/bfd2e9b23ee5421e8bbb058eb1b94881 : SUCCESS in 1h 35m 24s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/911223522c194df3b2e6a6340cf6accc : SUCCESS in 1h 09m 02s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/365cfe3d6c2948ba91ac9be19ab5b03c : SUCCESS in 31m 16s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/0ed2ebe3384e4f91bffab4b816a43a3f : SUCCESS in 51m 49s","accounts_in_message":[],"_revision_number":1},{"id":"66af6c3240e9159fb19717b0f643196a142b0f69","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-14 15:41:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"20599b67755caa8ece007aa33162dfff9c405db0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36902,"name":"Nevra Gürses Yakar","display_name":"nevra","email":"nevra.gurses@tubitak.gov.tr","username":"nevragurses"},"date":"2026-01-15 10:36:55.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"54c28937aa06e422d4c765da532c6af746b6bced","author":{"_account_id":36902,"name":"Nevra Gürses Yakar","display_name":"nevra","email":"nevra.gurses@tubitak.gov.tr","username":"nevragurses"},"date":"2026-01-15 10:38:38.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f325db2c9bff34282164997d02711d1055ce37d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-15 11:57:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86d82972a6b948239b0cdb874c8e9ad9\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fab9763ec36b4cc89a7688588f7039a0 : SUCCESS in 2m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/491ea0570ed4474e9fd8bd7f6819a4e2 : SUCCESS in 6m 15s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/12df3100a0e743838a16055549a63dc2 : SUCCESS in 10m 49s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/04f8ac07268b4e2690f720cfc3db96dc : SUCCESS in 7m 27s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/9d3a26c1959d44daadbd8486677df94e : SUCCESS in 8m 33s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/5cc584bb4d864e208224982514937ee8 : SUCCESS in 1h 18m 53s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/f1777b9aa9494b02bb9d294b0dd91dc7 : SUCCESS in 43m 47s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/6636c2d8b67941f3b92bf8a1ece9b3c4 : SUCCESS in 55m 32s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/e4a5ba8c14774708b627b6c6bca24fc5 : SUCCESS in 50m 21s","accounts_in_message":[],"_revision_number":2},{"id":"82874d9d560a469f44121ba2e5f395409b90cfa0","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-18 20:18:33.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"5676097a9ecfeffbf0267b69c23da938d17fbbee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36902,"name":"Nevra Gürses Yakar","display_name":"nevra","email":"nevra.gurses@tubitak.gov.tr","username":"nevragurses"},"date":"2026-01-20 12:22:02.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"efff94949b575219ce66509186bb421094866717","author":{"_account_id":36902,"name":"Nevra Gürses Yakar","display_name":"nevra","email":"nevra.gurses@tubitak.gov.tr","username":"nevragurses"},"date":"2026-01-20 12:25:03.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"dfc027c0c2794fafddf5e3cc7ae58edf7f9f65e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 13:18:40.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5457839eb14642abb1a5d1583dbc8ed4\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1ce20fd5d760406086195a89e1a08014 : SUCCESS in 2m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc4009a4756a478589a1a424fa951797 : SUCCESS in 3m 40s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/fa67da70be204278addf9d8a8d329e49 : SUCCESS in 8m 40s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/16975da9f1f34058af53c53ab717ae52 : SUCCESS in 4m 41s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/ae5f0797c9f74a93bccc17059b45e7e6 : SUCCESS in 4m 32s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/d2e47be8de0e45e69c6e5ae7043595d3 : FAILURE in 55m 11s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/0d5362ffb9084da39671df278ad1ba7c : SUCCESS in 32m 20s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/9db9f793a0d74a12a68e2bb272e9cfb2 : SUCCESS in 51m 13s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/4921f5fbfb884474a4332328ef78d6e7 : SUCCESS in 51m 16s","accounts_in_message":[],"_revision_number":3},{"id":"d6a99fe8219bf1f09e67aa940d3310d907eac0e8","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-20 20:04:24.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c5683a804eab50e2b927f4bb06d9a35d62b29cf1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36902,"name":"Nevra Gürses Yakar","display_name":"nevra","email":"nevra.gurses@tubitak.gov.tr","username":"nevragurses"},"date":"2026-01-21 05:42:56.000000000","message":"Uploaded patch set 4: New patch set was added with same tree, parent tree, and commit message as Patch Set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"740421525e6b5db56c2ceb5682bac8b7bca0c3e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 06:48:47.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6e2119e994474395aad3cdf5b2058847\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/026e50ce7f1c4e37a4a72682048c3a4f : SUCCESS in 1m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86061aca167f4883990f8cdee07ba6c9 : SUCCESS in 6m 40s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/b1991eb8756e46448a1e80e87ff76bb3 : SUCCESS in 11m 45s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/4198d6d776e74a8283174ad87b84a98c : SUCCESS in 3m 12s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/d60ffb69090b4108bb96c49a86f74999 : SUCCESS in 6m 10s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/146b30b1d322496eae80cae7290d901a : FAILURE in 52m 47s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/1effefe7e47240a8b952c03d200387ed : SUCCESS in 59m 11s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/db653f668a374f558ee19f4fb14981ac : SUCCESS in 52m 11s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/aaf00c99752a49dd84cfe5c5275b9ba5 : SUCCESS in 52m 04s","accounts_in_message":[],"_revision_number":4},{"id":"5ffdfcf842b88896ed1f125166279e6ce22a8ff5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36902,"name":"Nevra Gürses Yakar","display_name":"nevra","email":"nevra.gurses@tubitak.gov.tr","username":"nevragurses"},"date":"2026-01-21 06:49:13.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"f2890e6ab2e8ad79d14c21028b6f70a5f55e10f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36902,"name":"Nevra Gürses Yakar","display_name":"nevra","email":"nevra.gurses@tubitak.gov.tr","username":"nevragurses"},"date":"2026-01-21 07:00:30.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"7304b9a9257e9af8f752ec9ab9776b11e412a6c4","author":{"_account_id":36902,"name":"Nevra Gürses Yakar","display_name":"nevra","email":"nevra.gurses@tubitak.gov.tr","username":"nevragurses"},"date":"2026-01-21 07:02:32.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"580b9a71f477f050ccbfd4dab65a601d8418af09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 08:04:23.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d20d34faa95a47b4b464b6451696552c\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/76615b6557974cc893252b8c83652bcd : SUCCESS in 2m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8aa115b312e4656bbd863f2dfea743d : SUCCESS in 5m 35s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/a789c7188fe84e0780a057ef91b2fe7a : SUCCESS in 6m 31s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/1197a36c3e624f7fa95b9464edd399d6 : SUCCESS in 7m 22s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/61d43ade221a449888a0712e0eec558f : SUCCESS in 4m 23s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/06f38c04ffcd4710802eae38f17d7d3b : SUCCESS in 1h 03m 01s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/c73e543cd3d24391868053a2a46f1f39 : SUCCESS in 46m 16s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/2ef5a6895daa4bb185b22bf51a31fdaa : SUCCESS in 53m 27s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/8f57c213168f41c7bf0ffbc75fd03eaf : SUCCESS in 52m 00s","accounts_in_message":[],"_revision_number":6},{"id":"4b87ebbf32592c61412c8076504cbd3501ed9a81","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-01-22 02:24:19.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"f0af90b3cd6c359511e5c14889672a4542e9cc4a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 02:25:04.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"edc2abc7b8b69a2c38f9c7fd40f34e3d54a3cf45","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 03:57:51.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fdbc7d3a2c414647a5af8074b1e4e85f\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/26e28a45b2494aa9b029a519233d87da : SUCCESS in 3m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8e90355abb2409ab983726913631f41 : SUCCESS in 6m 54s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/e7e1814b53834acc990fb2abb7069a6d : SUCCESS in 12m 20s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/53b3cf47c79b4f519077245dbd938322 : SUCCESS in 7m 09s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/3c377a01905b45129491f2945ef4dc77 : SUCCESS in 8m 34s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/4ed3d7b950cb4fd9baa239d049c2acca : SUCCESS in 1h 25m 52s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/f791529ab65249f1af239766c974e25c : SUCCESS in 32m 44s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/86fed0cc58d94e5099be4675ceb76515 : SUCCESS in 45m 50s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/b08799f85bc246cca227fc2dc1e1321b : SUCCESS in 53m 05s","accounts_in_message":[],"_revision_number":6},{"id":"e4324d744778e832104bbbfc4d18908f3bbf6821","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 03:57:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"17e4223032a8cbf824abbe3349462c15d59a1170","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 03:58:41.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4fbf5345f1124f2187df13ae1786c642\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5f89f7a3515f44c78b736c0dad2e7af5 : SUCCESS in 39s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/298ddac972a8432bae2ac40775445219 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":6},{"id":"31cbe12ea822ffb444b526717af3991b33a66786","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-22 10:33:45.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"803af02ede28292879efb615953892b9c171cf49","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-04-10 17:08:10.000000000","message":"Hashtag added: CVE-2026-40212","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"ec3692124357d73f75aed6300ac72ca6ebaa49e0","revisions":{"e8e3b9b6e314500b4004b7429078ef8640b92f31":{"kind":"REWORK","_number":1,"created":"2026-01-14 07:26:55.000000000","uploader":{"_account_id":36902,"name":"Nevra Gürses Yakar","display_name":"nevra","email":"nevra.gurses@tubitak.gov.tr","username":"nevragurses"},"ref":"refs/changes/51/973351/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/51/973351/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/51/973351/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/51/973351/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/51/973351/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/51/973351/1"}}},"commit":{"parents":[{"commit":"84827b77a1e9cbf90182d40018234fe424d37aeb","subject":"Merge \"Update User Password - names of fields on the form are now more precise\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/84827b77a1e9cbf90182d40018234fe424d37aeb"}]}],"author":{"name":"nevragurses","email":"nevra.gurses@tubitak.gov.tr","date":"2026-01-14 06:52:30.000000000","tz":0},"committer":{"name":"nevragurses","email":"nevra.gurses@tubitak.gov.tr","date":"2026-01-14 07:29:32.000000000","tz":0},"subject":"Update: Instance detail view full log opening on url and re-fetches log data on page reload.","message":"Update: Instance detail view full log opening on url\nand re-fetches log data on page reload.\n\nChange-Id: I5704697f2c8bd0321014844cadea202911ec291c\nSigned-off-by: nevragurses \u003cnevra.gurses@tubitak.gov.tr\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e8e3b9b6e314500b4004b7429078ef8640b92f31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e8e3b9b6e314500b4004b7429078ef8640b92f31"}]},"branch":"refs/heads/master"},"94658f3369d260f337fd6163cbc282f0af23941e":{"kind":"REWORK","_number":2,"created":"2026-01-15 10:36:55.000000000","uploader":{"_account_id":36902,"name":"Nevra Gürses Yakar","display_name":"nevra","email":"nevra.gurses@tubitak.gov.tr","username":"nevragurses"},"ref":"refs/changes/51/973351/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/51/973351/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/51/973351/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/51/973351/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/51/973351/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/51/973351/2"}}},"commit":{"parents":[{"commit":"84827b77a1e9cbf90182d40018234fe424d37aeb","subject":"Merge \"Update User Password - names of fields on the form are now more precise\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/84827b77a1e9cbf90182d40018234fe424d37aeb"}]}],"author":{"name":"nevragurses","email":"nevra.gurses@tubitak.gov.tr","date":"2026-01-14 06:52:30.000000000","tz":0},"committer":{"name":"nevragurses","email":"nevra.gurses@tubitak.gov.tr","date":"2026-01-15 10:41:12.000000000","tz":0},"subject":"Update: Instance detail view full log opening on url and re-fetches log data on page reload.","message":"Update: Instance detail view full log opening on url\nand re-fetches log data on page reload.\n\nChange-Id: I5704697f2c8bd0321014844cadea202911ec291c\nSigned-off-by: nevragurses \u003cnevra.gurses@tubitak.gov.tr\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/94658f3369d260f337fd6163cbc282f0af23941e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/94658f3369d260f337fd6163cbc282f0af23941e"}]},"branch":"refs/heads/master"},"f67ba99038e1328a7026aaa284b731856d0e9185":{"kind":"REWORK","_number":3,"created":"2026-01-20 12:22:02.000000000","uploader":{"_account_id":36902,"name":"Nevra Gürses Yakar","display_name":"nevra","email":"nevra.gurses@tubitak.gov.tr","username":"nevragurses"},"ref":"refs/changes/51/973351/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/51/973351/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/51/973351/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/51/973351/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/51/973351/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/51/973351/3"}}},"commit":{"parents":[{"commit":"826a5f7666cab045b0cf84fc99a49a7590cf468b","subject":"Merge \"Same subnet with LB use Yes in UI\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/826a5f7666cab045b0cf84fc99a49a7590cf468b"}]}],"author":{"name":"nevragurses","email":"nevra.gurses@tubitak.gov.tr","date":"2026-01-14 06:52:30.000000000","tz":0},"committer":{"name":"nevragurses","email":"nevra.gurses@tubitak.gov.tr","date":"2026-01-20 12:15:06.000000000","tz":0},"subject":"Update: Instance detail view full log opening on url and re-fetches log data on page reload.","message":"Update: Instance detail view full log opening on url\nand re-fetches log data on page reload.\n\nChange-Id: I5704697f2c8bd0321014844cadea202911ec291c\nSigned-off-by: nevragurses \u003cnevra.gurses@tubitak.gov.tr\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f67ba99038e1328a7026aaa284b731856d0e9185"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f67ba99038e1328a7026aaa284b731856d0e9185"}]},"branch":"refs/heads/master"},"8dcb176176fd06b740cdc8beb29f1b962bb401ff":{"kind":"NO_CHANGE","_number":4,"created":"2026-01-21 05:42:56.000000000","uploader":{"_account_id":36902,"name":"Nevra Gürses Yakar","display_name":"nevra","email":"nevra.gurses@tubitak.gov.tr","username":"nevragurses"},"ref":"refs/changes/51/973351/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/51/973351/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/51/973351/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/51/973351/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/51/973351/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/51/973351/4"}}},"commit":{"parents":[{"commit":"826a5f7666cab045b0cf84fc99a49a7590cf468b","subject":"Merge \"Same subnet with LB use Yes in UI\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/826a5f7666cab045b0cf84fc99a49a7590cf468b"}]}],"author":{"name":"nevragurses","email":"nevra.gurses@tubitak.gov.tr","date":"2026-01-14 06:52:30.000000000","tz":0},"committer":{"name":"nevragurses","email":"nevra.gurses@tubitak.gov.tr","date":"2026-01-21 05:47:12.000000000","tz":0},"subject":"Update: Instance detail view full log opening on url and re-fetches log data on page reload.","message":"Update: Instance detail view full log opening on url\nand re-fetches log data on page reload.\n\nChange-Id: I5704697f2c8bd0321014844cadea202911ec291c\nSigned-off-by: nevragurses \u003cnevra.gurses@tubitak.gov.tr\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/8dcb176176fd06b740cdc8beb29f1b962bb401ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/8dcb176176fd06b740cdc8beb29f1b962bb401ff"}]},"branch":"refs/heads/master"},"bc93b1a18428f94dbbfaaa8bfe9140016fb6bcf7":{"kind":"REWORK","_number":5,"created":"2026-01-21 06:49:13.000000000","uploader":{"_account_id":36902,"name":"Nevra Gürses Yakar","display_name":"nevra","email":"nevra.gurses@tubitak.gov.tr","username":"nevragurses"},"ref":"refs/changes/51/973351/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/51/973351/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/51/973351/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/51/973351/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/51/973351/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/51/973351/5"}}},"commit":{"parents":[{"commit":"6f0ae5d21d01dd5380fe214c5e88335b0d80c81c","subject":"Merge \"Rename “Active Status” to “Active” in Instance Overview\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6f0ae5d21d01dd5380fe214c5e88335b0d80c81c"}]}],"author":{"name":"nevragurses","email":"nevra.gurses@tubitak.gov.tr","date":"2026-01-21 06:53:25.000000000","tz":0},"committer":{"name":"nevragurses","email":"nevra.gurses@tubitak.gov.tr","date":"2026-01-21 06:53:25.000000000","tz":0},"subject":"Update: Instance detail view full log opening on url and re-fetches log data on page reload.","message":"Update: Instance detail view full log opening on url\nand re-fetches log data on page reload.\n\nChange-Id: I5704697f2c8bd0321014844cadea202911ec291c\nSigned-off-by: nevragurses \u003cnevra.gurses@tubitak.gov.tr\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/bc93b1a18428f94dbbfaaa8bfe9140016fb6bcf7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/bc93b1a18428f94dbbfaaa8bfe9140016fb6bcf7"}]},"branch":"refs/heads/master"},"ec3692124357d73f75aed6300ac72ca6ebaa49e0":{"kind":"REWORK","_number":6,"created":"2026-01-21 07:00:30.000000000","uploader":{"_account_id":36902,"name":"Nevra Gürses Yakar","display_name":"nevra","email":"nevra.gurses@tubitak.gov.tr","username":"nevragurses"},"ref":"refs/changes/51/973351/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/51/973351/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/51/973351/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/51/973351/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/51/973351/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/51/973351/6"}}},"commit":{"parents":[{"commit":"6f0ae5d21d01dd5380fe214c5e88335b0d80c81c","subject":"Merge \"Rename “Active Status” to “Active” in Instance Overview\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6f0ae5d21d01dd5380fe214c5e88335b0d80c81c"}]}],"author":{"name":"nevragurses","email":"nevra.gurses@tubitak.gov.tr","date":"2026-01-21 07:00:02.000000000","tz":0},"committer":{"name":"nevragurses","email":"nevra.gurses@tubitak.gov.tr","date":"2026-01-21 07:05:10.000000000","tz":0},"subject":"Update: Instance detail view full log opening on url and re-fetches log data on page reload.","message":"Update: Instance detail view full log opening on url\nand re-fetches log data on page reload.\n\nChange-Id: I5704697f2c8bd0321014844cadea202911ec291c\nSigned-off-by: nevragurses \u003cnevra.gurses@tubitak.gov.tr\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ec3692124357d73f75aed6300ac72ca6ebaa49e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ec3692124357d73f75aed6300ac72ca6ebaa49e0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
