)]}'
{"id":"openstack%2Fskyline-console~973390","triplet_id":"openstack%2Fskyline-console~master~Ic81ae814b79579afdc941476215ab88d5d5c8755","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-01-14 14:58:59.000000000","reason":"removed on reply"},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-01-16 02:24:37.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-01-16 14:48:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic81ae814b79579afdc941476215ab88d5d5c8755","subject":"Instance details, tab Instance Snapshots rename to Snapshots","status":"MERGED","created":"2026-01-14 14:22:06.000000000","updated":"2026-01-16 14:49:42.000000000","submitted":"2026-01-16 14:48:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"973390","meta_rev_id":"7afe0cc6635f5901fb36c1a278737fba0634a106","_number":973390,"virtual_id_number":973390,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-16 14:48:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"recommended":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"all":[{"value":1,"date":"2026-01-14 14:58:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":2,"date":"2026-01-16 02:24:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":1,"date":"2026-01-16 02:24:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-14 14:25:33.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2026-01-14 14:25:33.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"},{"updated":"2026-01-14 15:55:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"902670a938ad44237a1b273d71817662ab2aa736","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-14 14:22:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"83940cbf8fecf583a0066f56010b5185e25774e6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-14 14:24:20.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"b61318f22619f718e079956724c09bee53f40244","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-14 14:25:13.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"9fcf2eb81baabd763a0f875d0babf8d997f45976","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-14 14:25:33.000000000","message":"Patch Set 3:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"543e80c4424914788d091f392015b6d448de727a","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-14 14:58:59.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"0066d6f900d379c5260f6cd6701bb86e5695dd06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 15:55:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf50556ab65c41298c8d8049c1b314a1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c455fb9c829343c0af6c9c93fd8bbc50 : SUCCESS in 4m 23s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/b89a62dc02644b8c9ba3e873387a91d7 : SUCCESS in 10m 58s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/9f951ab1c10c4b6d891cbe8d27dfde21 : SUCCESS in 3m 53s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/3f1c333cbcb64e37a80b9a43bc3ce2be : SUCCESS in 7m 43s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/3bd68a3f5a624e21ba06ac9662de4f90 : SUCCESS in 1h 23m 31s","accounts_in_message":[],"_revision_number":3},{"id":"3c89ce59e0c32ae3588f06d621562cef9aa22ada","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-01-16 02:24:37.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"94a970e2051a10656697aa967946cfbe4a8bc335","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 02:25:05.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1f6dfff52e8f73e08e5b99215192ebfe1ca1a9b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 04:41:41.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a93efc4b2ffa45178e1717780826e327\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e6de5dec68b4c6ebfdc9b1fe0eeace9 : SUCCESS in 8m 17s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/44b8f28f162f4b1a9db22deb6ffe8904 : SUCCESS in 4m 20s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/1977f7ea4ada47be90e4e50c4a307ad2 : SUCCESS in 5m 22s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/9c395cc9872b4ae49b1ebe276307a4a9 : SUCCESS in 7m 10s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/a92b19d861024d1fb1e62b6d60c1347b : FAILURE in 33m 58s","accounts_in_message":[],"_revision_number":3},{"id":"d4402874cf26b02e97a2ddb303ed157e9628fb8e","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-16 11:42:46.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"22b898a8962ec601d45ffd593372366253969bd7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 13:13:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/11e475ff0691449685d79e998fde576f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a21a1e118dd479e9349455dd90229c6 : SUCCESS in 6m 25s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/448502f141a9480aaea1bf397aca63e1 : SUCCESS in 12m 30s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/36a84aaaf7154aa4adc955b677638139 : SUCCESS in 7m 35s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/59a77a2a400942d296fcd542bd317e97 : SUCCESS in 7m 26s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/4fdb876203954bffad749e4be8eddff4 : SUCCESS in 1h 25m 40s","accounts_in_message":[],"_revision_number":3},{"id":"ad71b0e2ba579cf6096efa47d120fdba00425b56","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 13:14:24.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e854db702c4ce04971b2dbfb913a36699ed112c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 14:48:52.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/945d3cab648a4449b640595cc3698c2d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d16412c1bfb42d0ad497a60ae9ce0e8 : SUCCESS in 5m 09s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/69201481077d42fc8596c9064dd24598 : SUCCESS in 4m 55s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/91bb14bce63540c39481451f29a276de : SUCCESS in 6m 13s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/c1ba657a06df4c138a3f5f516761d0e6 : SUCCESS in 7m 59s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/ef08cd99b2434f07893fdd7ce15d1c05 : SUCCESS in 1h 22m 04s","accounts_in_message":[],"_revision_number":3},{"id":"7628b082d451cba6a517b2828069800757e3aa1a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 14:48:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"7afe0cc6635f5901fb36c1a278737fba0634a106","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 14:49:42.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c295c04f6a9c4f2fbbba862971562492\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6e001bb9ba04e02a29dcfabf1eec09d : SUCCESS in 43s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a1079bf85094b2276b03c7192435e9236ce6d292","revisions":{"e9f362d5c2bc1343d7ac5c51cbf328418545793f":{"kind":"REWORK","_number":1,"created":"2026-01-14 14:22:06.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/90/973390/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/90/973390/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/90/973390/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/90/973390/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/90/973390/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/90/973390/1"}}},"commit":{"parents":[{"commit":"271032baab085cde5f85c5c1e9b1cffb6afcd4ce","subject":"Merge \"Correct various grammar mistakes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/271032baab085cde5f85c5c1e9b1cffb6afcd4ce"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-14 14:22:06.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-14 14:22:06.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: Ic81ae814b79579afdc941476215ab88d5d5c8755\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e9f362d5c2bc1343d7ac5c51cbf328418545793f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e9f362d5c2bc1343d7ac5c51cbf328418545793f"}]},"branch":"refs/heads/master"},"7c3d78dfa6e356efea0398c45eb2be20ef990a8a":{"kind":"REWORK","_number":2,"created":"2026-01-14 14:24:20.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/90/973390/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/90/973390/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/90/973390/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/90/973390/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/90/973390/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/90/973390/2"}}},"commit":{"parents":[{"commit":"271032baab085cde5f85c5c1e9b1cffb6afcd4ce","subject":"Merge \"Correct various grammar mistakes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/271032baab085cde5f85c5c1e9b1cffb6afcd4ce"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-14 14:22:06.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-14 14:24:19.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: Ic81ae814b79579afdc941476215ab88d5d5c8755\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7c3d78dfa6e356efea0398c45eb2be20ef990a8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7c3d78dfa6e356efea0398c45eb2be20ef990a8a"}]},"branch":"refs/heads/master"},"a1079bf85094b2276b03c7192435e9236ce6d292":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-01-14 14:25:13.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/90/973390/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/90/973390/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/90/973390/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/90/973390/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/90/973390/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/90/973390/3"}}},"commit":{"parents":[{"commit":"271032baab085cde5f85c5c1e9b1cffb6afcd4ce","subject":"Merge \"Correct various grammar mistakes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/271032baab085cde5f85c5c1e9b1cffb6afcd4ce"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-14 14:22:06.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-14 14:25:13.000000000","tz":0},"subject":"Instance details, tab Instance Snapshots rename to Snapshots","message":"Instance details, tab Instance Snapshots rename to Snapshots\n\nAfter open instance details.. We already in instance, phrase not need.\n\nChange-Id: Ic81ae814b79579afdc941476215ab88d5d5c8755\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a1079bf85094b2276b03c7192435e9236ce6d292"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a1079bf85094b2276b03c7192435e9236ce6d292"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
