)]}'
{"id":"openstack%2Fskyline-console~973403","triplet_id":"openstack%2Fskyline-console~master~Ic2c76585380d910343a3e1ddecad26fd5813b64a","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-01-15 00:00:27.000000000","reason":"removed on reply"},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-01-16 02:25:28.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-01-16 04:52:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic2c76585380d910343a3e1ddecad26fd5813b64a","subject":"Volume details, tab Volume Snapshots rename to Snapshots","status":"MERGED","created":"2026-01-14 15:21:45.000000000","updated":"2026-01-16 04:53:05.000000000","submitted":"2026-01-16 04:52:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"973403","meta_rev_id":"d8f7629ea1936174e81386de2d9e5b4f8a23ce86","_number":973403,"virtual_id_number":973403,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-16 04:52:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"recommended":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"all":[{"value":1,"date":"2026-01-15 00:00:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":2,"date":"2026-01-16 02:25:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":1,"date":"2026-01-16 02:25:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-14 15:23:52.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2026-01-14 15:23:52.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"},{"updated":"2026-01-14 16:02:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e34fe6d3b690bf5685e09f95f0d6d2a174b8337b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-14 15:21:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"56cc6b6c7128637322aaba255e1a3325cf71a3b1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-14 15:23:00.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"1c756dbfba3187548961948636d9041accf84ebe","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-14 15:23:41.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"5c7b10f6e5505b94fbb76816bcd756a97a9dd946","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-14 15:23:52.000000000","message":"Patch Set 3:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"196be114234c2de747f8667e19e75eda1c104814","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 16:02:38.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8fa76add079a4496badd625f452eff73\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b33a54e145d48e78cc29fbdae9a02e0 : SUCCESS in 6m 12s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/ccbb7f758c2b4f009c0845631777a222 : SUCCESS in 6m 42s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/1628b17328754917b277f5fa7d5dd349 : SUCCESS in 3m 28s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/2b1a530dcca546c8a14eaa233e5188f1 : SUCCESS in 6m 32s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/96be41f8a10147a8bd6f5aeda571bc6c : FAILURE in 38m 02s","accounts_in_message":[],"_revision_number":3},{"id":"7bfd407f3f16f2e939b5def4d1eda6839322cde0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-14 16:15:46.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"ce506341ff5ca4faa5ebb9562be60bc6cb78525e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 17:21:44.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cd9e09f5cce141bbae65964ef824f654\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e409fed827454b0fa0347e5b8a61ef9b : SUCCESS in 6m 24s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/aa82733028d041b496ba17c995615ee4 : SUCCESS in 4m 52s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/f163fa0ef77345a7bc01b8c7f66df7ca : SUCCESS in 4m 46s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/03d832cd42cb4af5bf61886ae3da49cd : SUCCESS in 7m 42s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/6f8c5fb9a38e43a5ad83a277e63b891b : FAILURE in 1h 00m 07s","accounts_in_message":[],"_revision_number":4},{"id":"72e092508533746c7e256c923db0854e9e69282c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-14 17:28:14.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"bae1be635d8c66893a0e844a78c90a46c23e3b5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 18:28:44.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f248624a302b437fb63c1f4c6d419d5b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f007528708da4c6bb5c9b5c7e71e8b5d : SUCCESS in 6m 06s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/04a66d9eae8841778c634146c105ded5 : SUCCESS in 11m 28s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/117758316a664ddca8d251242ca62ed4 : SUCCESS in 8m 32s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/58a5ebf0187f446b914cf206dad00d3c : SUCCESS in 3m 55s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/bd5444758e804988b94c2932c1f4c7bf : FAILURE in 54m 37s","accounts_in_message":[],"_revision_number":5},{"id":"f219e6e4a942c6fce4009449cd9980eeae7b142e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-14 19:02:04.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"43148de199cc68ac91fdcbd97ccf26536db05b0f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-14 19:22:01.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nThe following files contain Git conflicts:\n* test/e2e/integration/pages/storage/volume.spec.js\n","accounts_in_message":[],"_revision_number":7},{"id":"79135a888823f1486e1013004023830bcb6aae81","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-14 19:23:29.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"d0345d0ca22fc6eda5208cc737ecc1ce1a930b9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 20:29:38.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5d639b1f809f40a58da988a28be486a8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b13644c4453453bb2126a89a3a60008 : SUCCESS in 6m 59s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/c019147c2ffb472fbfac8f101c467993 : SUCCESS in 11m 26s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/0371ccb38860414285362bb5b8da624e : SUCCESS in 6m 40s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/907886cd6f4b49bf86904817249a1452 : SUCCESS in 7m 35s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/ae9f27816b4a4b62be82b87fd9b888ab : FAILURE in 1h 00m 59s","accounts_in_message":[],"_revision_number":8},{"id":"d89ac4fa5d0bdf1823a67d11f9ceb440ad4b0181","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-14 20:32:39.000000000","message":"Patch Set 9: Published edit on patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"d3bd58a401699462938cd951f118cad2cbd65abe","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-14 20:32:42.000000000","message":"Patch Set 9:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":9},{"id":"196f50fd9cf38ad52530e44bbbfbcc444a7466b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-14 21:23:06.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b4b11576d064c7590f40d4f0d31f01c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89b7599610a449109ef5ff5a23d7eb3e : SUCCESS in 7m 36s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/aac68b7a85b2429d84084cf406f796df : SUCCESS in 7m 37s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/986f0533aded432c903328fa5bd05e29 : SUCCESS in 9m 29s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/36ccd39e2cc547a89609f3f121d199a7 : SUCCESS in 6m 36s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/5379cfc95dcd4bb3a23803982c0d8595 : SUCCESS in 49m 04s","accounts_in_message":[],"_revision_number":9},{"id":"53a9850d737def836343c31716c735cf78b399d7","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-15 00:00:27.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"9ed77936662e3d163d2892ee7b5df900499ca21b","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-01-16 02:25:28.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"58148cb1ac24dd06a7378a451e7a5ef23055cb81","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 02:26:15.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"67b2e546ac83cbce82105d56b05a3c3ece9bcf08","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 04:52:15.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e32552e898444deb89a7e5158dcacd9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e2a6271683c475f9b7efca32f1293aa : SUCCESS in 6m 16s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/2d433dcac0de45628bfa3251242ce461 : SUCCESS in 4m 20s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/10a03ec804024d5e8b0da4155a1617aa : SUCCESS in 3m 21s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/824ce21f372e4bdfb91d1515e6716307 : SUCCESS in 7m 13s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/de110c9d3a744f7189bb61742f1ba8a7 : SUCCESS in 41m 09s","accounts_in_message":[],"_revision_number":9},{"id":"a53ff8cffc9c1a4850f76e8fcf68be690743b0de","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 04:52:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9},{"id":"d8f7629ea1936174e81386de2d9e5b4f8a23ce86","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 04:53:05.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/576ab44a131e4b2eafc45f89ffd2f236\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/739408b300b14e91a368184b8b6762ac : SUCCESS in 42s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"dcc9dd34c92fccc9cc8cd1c1cc5a591eae710b61","revisions":{"e63f7587f61f70ae097b8aff15d6fb706a6a0d95":{"kind":"REWORK","_number":1,"created":"2026-01-14 15:21:45.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/03/973403/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/03/973403/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/03/973403/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/03/973403/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/03/973403/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/03/973403/1"}}},"commit":{"parents":[{"commit":"271032baab085cde5f85c5c1e9b1cffb6afcd4ce","subject":"Merge \"Correct various grammar mistakes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/271032baab085cde5f85c5c1e9b1cffb6afcd4ce"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-14 15:21:45.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-14 15:21:45.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: Ic2c76585380d910343a3e1ddecad26fd5813b64a\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e63f7587f61f70ae097b8aff15d6fb706a6a0d95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e63f7587f61f70ae097b8aff15d6fb706a6a0d95"}]},"branch":"refs/heads/master"},"afcba2e4a6e38af057aec95004bb0d3010938195":{"kind":"REWORK","_number":2,"created":"2026-01-14 15:23:00.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/03/973403/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/03/973403/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/03/973403/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/03/973403/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/03/973403/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/03/973403/2"}}},"commit":{"parents":[{"commit":"271032baab085cde5f85c5c1e9b1cffb6afcd4ce","subject":"Merge \"Correct various grammar mistakes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/271032baab085cde5f85c5c1e9b1cffb6afcd4ce"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-14 15:21:45.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-14 15:23:00.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: Ic2c76585380d910343a3e1ddecad26fd5813b64a\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/afcba2e4a6e38af057aec95004bb0d3010938195"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/afcba2e4a6e38af057aec95004bb0d3010938195"}]},"branch":"refs/heads/master"},"7a8690ff20443db5f02417bb107b72bf07a320d0":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-01-14 15:23:41.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/03/973403/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/03/973403/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/03/973403/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/03/973403/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/03/973403/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/03/973403/3"}}},"commit":{"parents":[{"commit":"271032baab085cde5f85c5c1e9b1cffb6afcd4ce","subject":"Merge \"Correct various grammar mistakes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/271032baab085cde5f85c5c1e9b1cffb6afcd4ce"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-14 15:21:45.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-14 15:23:41.000000000","tz":0},"subject":"Volume details, tab Volume Snapshots rename to Snapshots","message":"Volume details, tab Volume Snapshots rename to Snapshots\n\nAfter open volume details.. We already in volume, phrase not need.\n\nChange-Id: Ic2c76585380d910343a3e1ddecad26fd5813b64a\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7a8690ff20443db5f02417bb107b72bf07a320d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7a8690ff20443db5f02417bb107b72bf07a320d0"}]},"branch":"refs/heads/master","description":"Edit commit message"},"09f48c88a9890424c41618e6f4f37e4fc9e2e261":{"kind":"REWORK","_number":4,"created":"2026-01-14 16:15:46.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/03/973403/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/03/973403/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/03/973403/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/03/973403/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/03/973403/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/03/973403/4"}}},"commit":{"parents":[{"commit":"271032baab085cde5f85c5c1e9b1cffb6afcd4ce","subject":"Merge \"Correct various grammar mistakes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/271032baab085cde5f85c5c1e9b1cffb6afcd4ce"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-14 15:21:45.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-14 16:15:38.000000000","tz":0},"subject":"Volume details, tab Volume Snapshots rename to Snapshots","message":"Volume details, tab Volume Snapshots rename to Snapshots\n\nAfter open volume details.. We already in volume, phrase not need.\n\nChange-Id: Ic2c76585380d910343a3e1ddecad26fd5813b64a\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/09f48c88a9890424c41618e6f4f37e4fc9e2e261"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/09f48c88a9890424c41618e6f4f37e4fc9e2e261"}]},"branch":"refs/heads/master"},"77ad899131bf587ff71245facaad41d4ead76fd8":{"kind":"REWORK","_number":5,"created":"2026-01-14 17:28:14.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/03/973403/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/03/973403/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/03/973403/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/03/973403/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/03/973403/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/03/973403/5"}}},"commit":{"parents":[{"commit":"271032baab085cde5f85c5c1e9b1cffb6afcd4ce","subject":"Merge \"Correct various grammar mistakes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/271032baab085cde5f85c5c1e9b1cffb6afcd4ce"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-14 15:21:45.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-14 17:27:59.000000000","tz":0},"subject":"Volume details, tab Volume Snapshots rename to Snapshots","message":"Volume details, tab Volume Snapshots rename to Snapshots\n\nAfter open volume details.. We already in volume, phrase not need.\n\nChange-Id: Ic2c76585380d910343a3e1ddecad26fd5813b64a\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/77ad899131bf587ff71245facaad41d4ead76fd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/77ad899131bf587ff71245facaad41d4ead76fd8"}]},"branch":"refs/heads/master"},"3e1a420dcf22cb3abe75d702720eddc9619530bd":{"kind":"REWORK","_number":6,"created":"2026-01-14 19:02:04.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/03/973403/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/03/973403/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/03/973403/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/03/973403/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/03/973403/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/03/973403/6"}}},"commit":{"parents":[{"commit":"271032baab085cde5f85c5c1e9b1cffb6afcd4ce","subject":"Merge \"Correct various grammar mistakes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/271032baab085cde5f85c5c1e9b1cffb6afcd4ce"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-14 15:21:45.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-14 19:01:54.000000000","tz":0},"subject":"Volume details, tab Volume Snapshots rename to Snapshots","message":"Volume details, tab Volume Snapshots rename to Snapshots\n\nAfter open volume details.. We already in volume, phrase not need.\n\nChange-Id: Ic2c76585380d910343a3e1ddecad26fd5813b64a\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/3e1a420dcf22cb3abe75d702720eddc9619530bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/3e1a420dcf22cb3abe75d702720eddc9619530bd"}]},"branch":"refs/heads/master"},"25c33644d0cef5b782571f9664556c7aeb451453":{"kind":"REWORK","_number":7,"created":"2026-01-14 19:22:01.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/03/973403/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/03/973403/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/03/973403/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/03/973403/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/03/973403/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/03/973403/7"}}},"commit":{"parents":[{"commit":"4d7feee1f494c9d2de46cfe1a29459993a88952a","subject":"Volume details, tab Volume Backups rename to Backups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/4d7feee1f494c9d2de46cfe1a29459993a88952a"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-14 15:21:45.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-14 19:22:01.000000000","tz":0},"subject":"Volume details, tab Volume Snapshots rename to Snapshots","message":"Volume details, tab Volume Snapshots rename to Snapshots\n\nAfter open volume details.. We already in volume, phrase not need.\n\nChange-Id: Ic2c76585380d910343a3e1ddecad26fd5813b64a\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/25c33644d0cef5b782571f9664556c7aeb451453"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/25c33644d0cef5b782571f9664556c7aeb451453"}]},"branch":"refs/heads/master","description":"Rebase"},"0ce98c476d0bed284ca88569f2af80e0822b35a5":{"kind":"REWORK","_number":8,"created":"2026-01-14 19:23:29.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/03/973403/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/03/973403/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/03/973403/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/03/973403/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/03/973403/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/03/973403/8"}}},"commit":{"parents":[{"commit":"4d7feee1f494c9d2de46cfe1a29459993a88952a","subject":"Volume details, tab Volume Backups rename to Backups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/4d7feee1f494c9d2de46cfe1a29459993a88952a"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-14 15:21:45.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-14 19:23:21.000000000","tz":0},"subject":"Volume details, tab Volume Snapshots rename to Snapshots","message":"Volume details, tab Volume Snapshots rename to Snapshots\n\nAfter open volume details.. We already in volume, phrase not need.\n\nChange-Id: Ic2c76585380d910343a3e1ddecad26fd5813b64a\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/0ce98c476d0bed284ca88569f2af80e0822b35a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/0ce98c476d0bed284ca88569f2af80e0822b35a5"}]},"branch":"refs/heads/master"},"dcc9dd34c92fccc9cc8cd1c1cc5a591eae710b61":{"kind":"REWORK","_number":9,"created":"2026-01-14 20:32:39.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/03/973403/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/03/973403/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/03/973403/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/03/973403/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/03/973403/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/03/973403/9"}}},"commit":{"parents":[{"commit":"4d7feee1f494c9d2de46cfe1a29459993a88952a","subject":"Volume details, tab Volume Backups rename to Backups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/4d7feee1f494c9d2de46cfe1a29459993a88952a"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-14 15:21:45.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-14 20:32:39.000000000","tz":0},"subject":"Volume details, tab Volume Snapshots rename to Snapshots","message":"Volume details, tab Volume Snapshots rename to Snapshots\n\nAfter open volume details.. We already in volume, phrase not need.\n\nChange-Id: Ic2c76585380d910343a3e1ddecad26fd5813b64a\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/dcc9dd34c92fccc9cc8cd1c1cc5a591eae710b61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/dcc9dd34c92fccc9cc8cd1c1cc5a591eae710b61"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
