)]}'
{"id":"openstack%2Fskyline-console~973459","triplet_id":"openstack%2Fskyline-console~master~Ic6aa2362b60101b279f2c23a10596763daca9012","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-01-16 05:07:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic6aa2362b60101b279f2c23a10596763daca9012","subject":"Change \"Forbidden\" action to \"Disable\"","status":"MERGED","created":"2026-01-15 11:02:30.000000000","updated":"2026-01-16 05:07:27.000000000","submitted":"2026-01-16 05:07:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"973459","meta_rev_id":"c5052d8770dd733f22fe7a873411b61c8861a91f","_number":973459,"virtual_id_number":973459,"owner":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-16 05:07:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-01-16 02:30:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-01-16 02:30:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-15 12:07:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-16 02:30:48.000000000","updated_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"}],"messages":[{"id":"b659dbe76788e92289c54f0718645ef87f85766b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-15 11:02:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"23997cd824b641848e77b5246c93ee1245bdc456","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-15 11:14:49.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"49bbf78167070bb2e7939ad4d3fbadde5330a1e5","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-15 11:14:52.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"5ce7a86c4689f96763b4efe4eacd3d99f4684b98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-15 12:07:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6fae28082944dfbb5ac358837de7801\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/533e44ea16214b33a6e0dd687023c54a : SUCCESS in 6m 16s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/39033450260f4f86b6b7e8b23c3ec7a9 : SUCCESS in 9m 01s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/e8b14274165f4ce089a581ea6172730e : SUCCESS in 3m 44s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/54e5a1ef11c84140ad9d1f18b9a3b55b : SUCCESS in 7m 17s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/cd69d6cac65b4b55922fa53a72bea82e : SUCCESS in 50m 59s","accounts_in_message":[],"_revision_number":2},{"id":"e9db73d0137fdfcaa2e1b8b76144bdd9c4a529ff","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-01-16 02:30:48.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"30a207b99aeefa291a5e0698a73ca8504be6e7a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 02:31:29.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"529be4eb9328fa2c79103ae32c588db58903d575","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 05:07:27.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd1c03dff4a242408efee49f1db35e10\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21a9197b21804f9caf64f5461936ff2f : SUCCESS in 6m 31s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/956b3a2c421c444ead2d909626919f09 : SUCCESS in 7m 33s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/860e16e797db4ae39c42a69be9c02f5b : SUCCESS in 4m 48s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/a3ad382d1d17427ab46729dd3556c770 : SUCCESS in 5m 47s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/47fd4e70b2aa4bc7bd3654ba91e1d194 : SUCCESS in 31m 12s","accounts_in_message":[],"_revision_number":2},{"id":"c5052d8770dd733f22fe7a873411b61c8861a91f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 05:07:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d3933dbc52b5afb0bbd104ea7f6a430420555df5","revisions":{"c5d413db82ef77bb8f85a2176a492f0765b051ec":{"kind":"REWORK","_number":1,"created":"2026-01-15 11:02:30.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/59/973459/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/59/973459/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/59/973459/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/59/973459/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/59/973459/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/59/973459/1"}}},"commit":{"parents":[{"commit":"271032baab085cde5f85c5c1e9b1cffb6afcd4ce","subject":"Merge \"Correct various grammar mistakes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/271032baab085cde5f85c5c1e9b1cffb6afcd4ce"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-15 11:02:30.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-15 11:02:30.000000000","tz":0},"subject":"Change \"Forbidden\" action to \"Disable\"","message":"Change \"Forbidden\" action to \"Disable\"\n\nReplace \"forbidden\" action to reflect actual state\n more clearly and avoid misleading wording.\n\nChange-Id: Ic6aa2362b60101b279f2c23a10596763daca9012\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/c5d413db82ef77bb8f85a2176a492f0765b051ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/c5d413db82ef77bb8f85a2176a492f0765b051ec"}]},"branch":"refs/heads/master"},"d3933dbc52b5afb0bbd104ea7f6a430420555df5":{"kind":"REWORK","_number":2,"created":"2026-01-15 11:14:49.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/59/973459/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/59/973459/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/59/973459/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/59/973459/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/59/973459/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/59/973459/2"}}},"commit":{"parents":[{"commit":"271032baab085cde5f85c5c1e9b1cffb6afcd4ce","subject":"Merge \"Correct various grammar mistakes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/271032baab085cde5f85c5c1e9b1cffb6afcd4ce"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-15 11:02:30.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-15 11:14:44.000000000","tz":0},"subject":"Change \"Forbidden\" action to \"Disable\"","message":"Change \"Forbidden\" action to \"Disable\"\n\nReplace \"forbidden\" action to reflect actual state\n more clearly and avoid misleading wording.\n\nChange-Id: Ic6aa2362b60101b279f2c23a10596763daca9012\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d3933dbc52b5afb0bbd104ea7f6a430420555df5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d3933dbc52b5afb0bbd104ea7f6a430420555df5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
