)]}'
{"id":"openstack%2Fskyline-console~973514","triplet_id":"openstack%2Fskyline-console~master~If7bb85f64f57088bd3f3c19002295a6666200e5f","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-01-16 05:07:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If7bb85f64f57088bd3f3c19002295a6666200e5f","subject":"Modify \"Restricted Situation\" column","status":"MERGED","created":"2026-01-15 16:53:22.000000000","updated":"2026-01-16 05:08:55.000000000","submitted":"2026-01-16 05:07:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"973514","meta_rev_id":"d0b78a0fd98d4bad91f55389c056907bc98b1a31","_number":973514,"virtual_id_number":973514,"owner":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-16 05:07:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-01-16 02:31:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-01-16 02:31:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-15 17:36:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-16 02:31:07.000000000","updated_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"}],"messages":[{"id":"61e48c44882a9b6dd1d5a58b46ce3b2d874485ff","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-15 16:53:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"37825d59795eaa56b07a71a0a7959c5e20a1a480","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-15 16:54:26.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"e200a77fc539b24bb402b7555746232d8d7ab529","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-15 16:54:29.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"fcd049d08d395fdf14b5812b8194eeabfe9ba7ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-15 17:36:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/518a05da1ffd423f928cef0543dbe83f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2023803103ba46fa9cfd196db362dd03 : SUCCESS in 3m 22s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/b4922bbbe70149a19073c8ae9f34e7e1 : SUCCESS in 4m 30s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/a9ea1114165e40e48b98cbb2356d049a : SUCCESS in 9m 04s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/3fbe75a7324b48ad9f00dca48ed391ba : SUCCESS in 4m 46s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/4b7f2144bcfa4ccd823d71f2fddc0a97 : SUCCESS in 41m 32s","accounts_in_message":[],"_revision_number":2},{"id":"0b961e06c30dbdb4c8401edebffe809b0a3aad02","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-01-16 02:31:07.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e6397fc548657fce78ed6431a242f7fd37d2d4ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 02:31:56.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1f472c7f7cc9872128859dc3ed009c4a8d2239d2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 05:07:28.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b49adb3d57334c25871a896dfc8dd7a2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa42f96e87ed4f0dbc6783b499793a86 : SUCCESS in 6m 14s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/8518ff7841004a67b3e6840ab3808b46 : SUCCESS in 10m 35s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/4a6b6942fd3f49ebb344ffebb860cc4f : SUCCESS in 6m 36s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/ffae5fe1795f413da61e8a19ac859bb4 : SUCCESS in 3m 19s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/d9f3d744298941df9d0f419bba373e77 : SUCCESS in 38m 41s","accounts_in_message":[],"_revision_number":2},{"id":"fece75b46766e84ea8aa581d35acd69b0efad4f1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 05:07:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"d0b78a0fd98d4bad91f55389c056907bc98b1a31","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 05:08:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45683017f6c640f5a91586f74af75b41\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db11f9fd98724f36860f141c5996fff6 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ca5da651a93ec2495e4ed46fe187971f01fc5984","revisions":{"437f751d46084de78375b9a4d64a1ae6febb6945":{"kind":"REWORK","_number":1,"created":"2026-01-15 16:53:22.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/14/973514/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/14/973514/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/14/973514/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/14/973514/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/14/973514/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/14/973514/1"}}},"commit":{"parents":[{"commit":"271032baab085cde5f85c5c1e9b1cffb6afcd4ce","subject":"Merge \"Correct various grammar mistakes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/271032baab085cde5f85c5c1e9b1cffb6afcd4ce"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-15 16:53:22.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-15 16:53:22.000000000","tz":0},"subject":"Modify \"Restricted Situation\" column","message":"Modify \"Restricted Situation\" column\n\nSimplify  \"Restricted Situation\" column to \"Restriction\"\nfor more clarity.\n\nChange-Id: If7bb85f64f57088bd3f3c19002295a6666200e5f\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/437f751d46084de78375b9a4d64a1ae6febb6945"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/437f751d46084de78375b9a4d64a1ae6febb6945"}]},"branch":"refs/heads/master"},"ca5da651a93ec2495e4ed46fe187971f01fc5984":{"kind":"REWORK","_number":2,"created":"2026-01-15 16:54:26.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/14/973514/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/14/973514/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/14/973514/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/14/973514/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/14/973514/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/14/973514/2"}}},"commit":{"parents":[{"commit":"271032baab085cde5f85c5c1e9b1cffb6afcd4ce","subject":"Merge \"Correct various grammar mistakes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/271032baab085cde5f85c5c1e9b1cffb6afcd4ce"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-15 16:53:22.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-15 16:54:23.000000000","tz":0},"subject":"Modify \"Restricted Situation\" column","message":"Modify \"Restricted Situation\" column\n\nSimplify  \"Restricted Situation\" column to \"Restriction\"\nfor more clarity.\n\nChange-Id: If7bb85f64f57088bd3f3c19002295a6666200e5f\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ca5da651a93ec2495e4ed46fe187971f01fc5984"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ca5da651a93ec2495e4ed46fe187971f01fc5984"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
