)]}'
{"id":"openstack%2Fskyline-console~973518","triplet_id":"openstack%2Fskyline-console~master~I36a3c42fc90b9612f74f6df84aec1de6553a5b21","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-01-16 15:05:00.000000000","reason":"\u003cGERRIT_ACCOUNT_37632\u003e replied on the change","reason_account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-01-16 15:05:10.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I36a3c42fc90b9612f74f6df84aec1de6553a5b21","subject":"Remove \"System\" from \"System Roles\"","status":"ABANDONED","created":"2026-01-15 17:23:57.000000000","updated":"2026-01-16 15:05:10.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2bb0f65007c6e05804c08c0488668ad099553504","_number":973518,"virtual_id_number":973518,"owner":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-01-16 00:18:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":-1,"date":"2026-01-16 02:34:17.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-15 18:37:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-16 02:34:17.000000000","updated_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"}],"messages":[{"id":"bc595abd3e83e863a9e30d4115a12f055f48f150","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-15 17:23:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6472ee4f27a1bf640a6aa9151c3c9277a984696b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-15 17:25:09.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"cb1d97c937c9f2b30225b64eb03f47dac1fb360c","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-15 17:25:17.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"40a0fd1a32d70daea27102aa9d432823ed5a5487","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-15 17:28:00.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"b1f679f9f5714e7337c19e67af2dcf9ea8ad7c7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-15 18:37:07.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dc313ea6d3a348f8867413b43a902f72\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a1400b688bf47b4a48437c7e68ec942 : SUCCESS in 7m 51s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/9f67bda4bebb47499269cf60b57b6a3d : SUCCESS in 9m 31s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/5e7aa5bb93cc4627acefa5e8f7178bee : SUCCESS in 5m 14s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/8232521d29c64f38a19d8dfdb3710eef : SUCCESS in 6m 56s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/07cff582e6d945718695811e12928cdb : FAILURE in 1h 07m 30s","accounts_in_message":[],"_revision_number":3},{"id":"788d052da20a96c36f37301eaca4812678bd00c5","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-15 18:48:00.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7141e99d55f5e6ad21c88c1f577d3dc6c45041b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-15 19:41:14.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b832e4fa93964227afcd6d891fe30af2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a15febdff7e4a47953803632429019d : SUCCESS in 5m 12s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/8ea3ade0b19341c6a9200e6b06125d6b : SUCCESS in 10m 50s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/2633fde23cad472ab9b91603a63fd6f5 : SUCCESS in 4m 26s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/fc220c4884f64deeb3892bbfc6e15c47 : SUCCESS in 4m 27s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/6617a511f42d4726a2475f0cf4c8de9e : FAILURE in 52m 18s","accounts_in_message":[],"_revision_number":3},{"id":"90fb3d4a0c6469a984f80509ec09ed875966e472","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-15 23:40:13.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d3aa175279355117447c6655530cdc55ed6705e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-16 00:18:43.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bceec9b995a3427481c2689fc7b71e69\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/205f11c5cf9446b0ae8a0fb520517008 : SUCCESS in 6m 09s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/a8a576c8ff5e496ab6837a11aaf2f1bf : SUCCESS in 10m 46s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/4bff57db90e2456c883e3c86b48058c8 : SUCCESS in 8m 03s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/f7e19b55ba364ac195e879849c2d8d88 : SUCCESS in 4m 28s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/575b3cf890a34117a7d6711c382fa7e8 : SUCCESS in 37m 45s","accounts_in_message":[],"_revision_number":3},{"id":"a80ffa7f011696636547572ff7c41d494cdc0c59","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-01-16 02:34:17.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d0c966897c0b651444b04d51c1b88cc3b8f2c73e","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-16 15:05:00.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2bb0f65007c6e05804c08c0488668ad099553504","tag":"autogenerated:gerrit:abandon","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-16 15:05:10.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"cfe4263a006ffcc94620233f0bdb1348a6bb5c61","revisions":{"03248eb08d9a41f35cd0bc6cb58f42a6a409e5dc":{"kind":"REWORK","_number":1,"created":"2026-01-15 17:23:57.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/18/973518/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/18/973518/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/18/973518/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/18/973518/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/18/973518/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/18/973518/1"}}},"commit":{"parents":[{"commit":"271032baab085cde5f85c5c1e9b1cffb6afcd4ce","subject":"Merge \"Correct various grammar mistakes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/271032baab085cde5f85c5c1e9b1cffb6afcd4ce"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-15 17:23:57.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-15 17:23:57.000000000","tz":0},"subject":"Remove \"System\" from \"System Roles\"","message":"Remove \"System\" from \"System Roles\"\n\nGet rid of \"System\" from \"System Roles\" in \"Users\"\nsection to simplify UI, cause it brings not much sense.\n\nChange-Id: I36a3c42fc90b9612f74f6df84aec1de6553a5b21\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/03248eb08d9a41f35cd0bc6cb58f42a6a409e5dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/03248eb08d9a41f35cd0bc6cb58f42a6a409e5dc"}]},"branch":"refs/heads/master"},"7fa6471b6ed14d43e34b8e1dd9b2ade74c8e2519":{"kind":"REWORK","_number":2,"created":"2026-01-15 17:25:09.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/18/973518/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/18/973518/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/18/973518/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/18/973518/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/18/973518/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/18/973518/2"}}},"commit":{"parents":[{"commit":"271032baab085cde5f85c5c1e9b1cffb6afcd4ce","subject":"Merge \"Correct various grammar mistakes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/271032baab085cde5f85c5c1e9b1cffb6afcd4ce"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-15 17:23:57.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-15 17:25:03.000000000","tz":0},"subject":"Remove \"System\" from \"System Roles\"","message":"Remove \"System\" from \"System Roles\"\n\nGet rid of \"System\" from \"System Roles\" in \"Users\"\nsection to simplify UI, cause it brings not much sense.\n\nChange-Id: I36a3c42fc90b9612f74f6df84aec1de6553a5b21\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7fa6471b6ed14d43e34b8e1dd9b2ade74c8e2519"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7fa6471b6ed14d43e34b8e1dd9b2ade74c8e2519"}]},"branch":"refs/heads/master"},"cfe4263a006ffcc94620233f0bdb1348a6bb5c61":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-01-15 17:28:00.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/18/973518/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/18/973518/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/18/973518/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/18/973518/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/18/973518/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/18/973518/3"}}},"commit":{"parents":[{"commit":"d71d8bd8f16de77a647e993f166d8bea9f27a655","subject":"Modify \"Affiliated Domain\" Field","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d71d8bd8f16de77a647e993f166d8bea9f27a655"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-15 17:23:57.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-15 17:28:00.000000000","tz":0},"subject":"Remove \"System\" from \"System Roles\"","message":"Remove \"System\" from \"System Roles\"\n\nGet rid of \"System\" from \"System Roles\" in \"Users\"\nsection to simplify UI, cause it brings not much sense.\n\nChange-Id: I36a3c42fc90b9612f74f6df84aec1de6553a5b21\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/cfe4263a006ffcc94620233f0bdb1348a6bb5c61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/cfe4263a006ffcc94620233f0bdb1348a6bb5c61"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
