)]}'
{"id":"openstack%2Fskyline-console~973738","triplet_id":"openstack%2Fskyline-console~master~I0ae56ee661e1c5eada277ac6f194406a81b5bf45","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-01-19 04:39:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0ae56ee661e1c5eada277ac6f194406a81b5bf45","subject":"Correct various grammar of formatting mistakes ","status":"MERGED","created":"2026-01-18 18:59:02.000000000","updated":"2026-01-19 04:40:18.000000000","submitted":"2026-01-19 04:39:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"973738","meta_rev_id":"c53b85db9164415a2792eb4b89885804893a2c69","_number":973738,"virtual_id_number":973738,"owner":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-19 04:39:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-01-19 03:46:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-01-19 03:46:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-18 20:17:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-19 03:46:11.000000000","updated_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"}],"messages":[{"id":"f1b8f004c62782e854a17e2f635a771d3a19b697","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-18 18:59:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"70d18af7d2fd6d690b11aa1e732af11327846532","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-18 20:17:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a3c1305751a4c8899060eecf3231bc2\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a53d8b99f3d44042b21dda59315cba16 : SUCCESS in 5m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/447aa47544854887abebd71175657727 : SUCCESS in 2m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f71cef79099945efb6b7beae23bce976 : SUCCESS in 5m 54s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/23e2cd7966ba49ee9d0c8589ba73659d : SUCCESS in 4m 09s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/de0c7b362975487d933a0b8945a81ad8 : SUCCESS in 8m 21s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/19fb5573632445fe88e50972fd5b8aed : SUCCESS in 4m 12s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/2a6fc19cd549477f9edc04a125905f3a : SUCCESS in 1h 09m 44s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/d1d5dba786ac482eacdddf06c6972c23 : SUCCESS in 49m 25s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/4cd80738440a418e9ceb75a79862affe : SUCCESS in 30m 38s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/3481ea7cb648452eb0ce83da0015f5e6 : SUCCESS in 32m 58s","accounts_in_message":[],"_revision_number":1},{"id":"0029a37131d908a209ec489aa89c1d22ddf27f2e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-18 20:18:13.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"883bfbd0504867791cdedb9c1e8f590e0cc8a889","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-18 21:24:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b62daf4aaf8f482db3c4c654e649cf53\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/031532f51a0f4b0cbf56473083b7c2d1 : SUCCESS in 6m 15s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/69374790402b4034acd24cbd68840b53 : SUCCESS in 8m 08s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/529a999a232a47d99e1c5639261f2241 : SUCCESS in 3m 18s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/5af794cc70424fada70924653843e7a8 : SUCCESS in 3m 10s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/2ba69d1ac75e4e6d91e4e84e42a7695f : SUCCESS in 1h 04m 27s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/d8694fb8db8942b2aa489dd693b87f12 : SUCCESS in 32m 29s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/b6ce7ef905db405fbf2ecf62e8cafaf3 : SUCCESS in 42m 15s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/b92713264ba9411aacc9bb491bbc2010 : SUCCESS in 57m 27s","accounts_in_message":[],"_revision_number":2},{"id":"67d24218ebde4c1c9d567587f56350ce0caf5ce8","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-18 21:24:51.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"0138ff91ed2a38435dae6d90057286485974ac4b","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-01-19 03:46:11.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"898a09dc18b2feffb29188aea48d7842b0cd5b57","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-01-19 03:46:17.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"748eebddd8e3c89f655b8048b15a9f8e9902c43f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 03:46:45.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d4850c1e9219e8063e4fa6ef3c5b842fb0448f2c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 04:39:30.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8708c383ecd3405080f665984ee7b1c9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f61c0c2262074a08a84bae9f5b633572 : SUCCESS in 6m 45s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/cbd717a25519423f81fbfc3f8e127d01 : SUCCESS in 6m 48s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/f952653ede2d4634861f032ee4129760 : SUCCESS in 7m 53s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/422dc4a1bb13437e9157a8e5e2cd5b4c : SUCCESS in 4m 51s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/78c1a36721db427cbc9cfae77eb525cb : SUCCESS in 52m 18s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/16732d0d588c4253905d336ca6c12edf : SUCCESS in 49m 24s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/789e57cc706a4e98abe9461f48688ff9 : SUCCESS in 30m 38s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/a66a5126c2e5447e9d0bbbc58913ebd8 : SUCCESS in 33m 51s","accounts_in_message":[],"_revision_number":2},{"id":"d4d59dc3f665600ed0f2bbbc146652f0a974a0ec","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 04:39:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"c53b85db9164415a2792eb4b89885804893a2c69","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 04:40:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3abf2f29c32f4bb5a49469cecd2bfa0c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc9ba4769e364d29a50731b1d5706145 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"dce66c80f636fdc59759e84cea44a58ada70ceee","revisions":{"f8a44a56967827c0dd71b54b191279d6318df4fc":{"kind":"REWORK","_number":1,"created":"2026-01-18 18:59:02.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/38/973738/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/38/973738/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/38/973738/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/38/973738/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/38/973738/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/38/973738/1"}}},"commit":{"parents":[{"commit":"6733040820533bc7f2360a2225af18d3ffcd1ee7","subject":"Merge \"Tags, server replace to instance\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6733040820533bc7f2360a2225af18d3ffcd1ee7"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-18 18:59:02.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-18 18:59:02.000000000","tz":0},"subject":"Correct various grammar of formatting mistakes ","message":"Correct various grammar of formatting mistakes \n\nChange-Id: I0ae56ee661e1c5eada277ac6f194406a81b5bf45\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f8a44a56967827c0dd71b54b191279d6318df4fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f8a44a56967827c0dd71b54b191279d6318df4fc"}]},"branch":"refs/heads/master"},"dce66c80f636fdc59759e84cea44a58ada70ceee":{"kind":"REWORK","_number":2,"created":"2026-01-18 20:18:13.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/38/973738/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/38/973738/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/38/973738/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/38/973738/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/38/973738/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/38/973738/2"}}},"commit":{"parents":[{"commit":"6733040820533bc7f2360a2225af18d3ffcd1ee7","subject":"Merge \"Tags, server replace to instance\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6733040820533bc7f2360a2225af18d3ffcd1ee7"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-18 18:59:02.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-18 20:02:16.000000000","tz":0},"subject":"Correct various grammar of formatting mistakes ","message":"Correct various grammar of formatting mistakes \n\nChange-Id: I0ae56ee661e1c5eada277ac6f194406a81b5bf45\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/dce66c80f636fdc59759e84cea44a58ada70ceee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/dce66c80f636fdc59759e84cea44a58ada70ceee"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
