)]}'
{"id":"openstack%2Fskyline-console~973742","triplet_id":"openstack%2Fskyline-console~master~I11c3e584b9ad2eaad718433e42e29af40121ffdc","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-01-20 08:24:43.000000000","reason":"\u003cGERRIT_ACCOUNT_37632\u003e replied on the change","reason_account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-01-20 01:14:43.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-01-20 11:04:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I11c3e584b9ad2eaad718433e42e29af40121ffdc","subject":"Same subnet with LB use Yes in UI","status":"MERGED","created":"2026-01-18 20:39:31.000000000","updated":"2026-01-20 11:05:14.000000000","submitted":"2026-01-20 11:04:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"973742","meta_rev_id":"b6e9108ee19449b3717482cf9cb18ba8f40dc0b9","_number":973742,"virtual_id_number":973742,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-20 11:04:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-01-20 01:14:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-01-20 01:14:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-18 20:41:08.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2026-01-18 20:41:08.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"},{"updated":"2026-01-18 21:45:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"02b1b66027e670a03894e54b66a0836fb572eff9","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-18 20:39:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c1cd7c065e0818d75554b103611147131636c7b7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-18 20:41:01.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"8e0791cf9201321fb65cfb82719b2d4719e50b45","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-18 20:41:08.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"4188d5bff4a5bc46e9561404d21a034a7bee20a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-18 21:45:40.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/04943c91899c4b048fde21c2dc05d101\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6275e72b321643bda994a8c6cfb90fb9 : SUCCESS in 3m 47s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/beb02bdfc910402bad8c889c853b9ad1 : SUCCESS in 4m 49s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/8b5d093826de4e1ca4fff78c110f09e9 : SUCCESS in 3m 17s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/38dd43f9be004e1883fc471692065c3e : SUCCESS in 3m 10s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/e1d9ecf18fc3430d884c86be4c4b3b06 : FAILURE in 50m 54s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/d8416ad24e464619b18eb5edc9f82684 : SUCCESS in 46m 16s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/7652ee3052d349c9b4c4dd51ebb82ffb : FAILURE in 1h 04m 07s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/ad2f60c468f247ad8873480615b3f05f : SUCCESS in 41m 21s","accounts_in_message":[],"_revision_number":2},{"id":"729135709006b2ca1935e35a800393eae03c31fe","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-19 07:34:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"aea46a5834131c970386e5068962f25ad03423c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 09:13:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19721d44c53947f4856eeb3f22c7a1e7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c07ff580c874474b76e44330ac923fc : SUCCESS in 3m 30s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/c25b0d657c2b4858ab24fc1727628899 : SUCCESS in 8m 21s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/7337f984f19b4843bb4b7056e403688a : SUCCESS in 4m 21s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/e53ed85377294a1a8f6b4fbbfdead7a0 : SUCCESS in 3m 46s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/1272ad5a191846fca13f721f16e4232b : SUCCESS in 1h 32m 30s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/f187b30b57a94b859b8314b41d86274a : SUCCESS in 32m 50s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/c4ec8b03c0504bcaa4947f379ebfa490 : SUCCESS in 47m 29s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/28d8513c144e46c78d196e6a0526201c : SUCCESS in 49m 01s","accounts_in_message":[],"_revision_number":2},{"id":"b4b86543bc63f303854727623f644578d30c2573","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-01-20 01:14:43.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d339cc6f6391a9d8fa239f3ecb5d3b06f6bb67f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 01:15:09.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5b66b7043eb3d385d127c2618ab7c6a222c6c18d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 02:41:14.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fce6e959313e4e779425580c767da9e3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26df9f27c8654b919fb9913d83c3d2e3 : SUCCESS in 5m 35s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/e90a38d146b244768d8fc8c81183e1a0 : SUCCESS in 6m 16s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/d631da2b7ad54861adc452716d20885c : SUCCESS in 7m 35s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/dcb1b7f9f41c47f3ae3f3b714744acff : SUCCESS in 7m 39s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/4a6a5e0d890f489aa16167d3cc07d32d : SUCCESS in 1h 20m 34s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/2615e33185404a8ca56bf705ce80baee : SUCCESS in 31m 12s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/889a245bc9314832bfbda73471f6cb29 : SUCCESS in 51m 23s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/6571ef75acaf4f52ad5cce6251df8429 : FAILURE in 11m 02s","accounts_in_message":[],"_revision_number":2},{"id":"aee2743f514ad8a9b0b32d6e9a35f7450a4d90b7","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-20 06:24:12.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3f9afd2bf7af94ae6faae04f7ca44979022b4fa1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 07:35:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5df484f71e054d898fa51a0898aff19b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f46124c1599b4c86a98e4cc0fdaa0182 : SUCCESS in 7m 03s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/b0541025025d4900b021dd56c5043c43 : SUCCESS in 4m 28s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/719ac99ff14a4324adfdc4289ab869f3 : SUCCESS in 6m 01s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/7b488138706a44a59122f1bce4f98434 : SUCCESS in 3m 19s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/e4633e4e112843979b17eab96dba89b7 : SUCCESS in 1h 05m 24s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/1f31489e46e6453a913acceb9d3439f3 : SUCCESS in 58m 39s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/6990e4b1769e4c3095b1002075daa3ab : FAILURE in 13m 33s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/1d2da098e9b240288e37c004ad59247f : SUCCESS in 39m 10s","accounts_in_message":[],"_revision_number":2},{"id":"1f6dbc807fcd5d7ad01cce4bbc1a65e18e7c10bd","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-20 08:24:43.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bc68fa0a7a4e2ea2fe0e2916b3003aed90af8818","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 09:50:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee3105e3062a46078339e25821bf5540\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb81881deb834b7490c828e536535ae6 : SUCCESS in 3m 52s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/4fea04f571ee4dd092fd6934c424e724 : SUCCESS in 10m 58s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/5e4e8607811846d1940a091179e59b35 : SUCCESS in 8m 08s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/c5f5efae2600491db5a2bc99234c0019 : SUCCESS in 7m 26s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/185ef87e928d45aaa89940c7a9e61a8b : SUCCESS in 1h 20m 14s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/7db5213a1fb54a44b0890fe9750eb51d : SUCCESS in 33m 33s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/3916f876429d4b719681fbed6ef21371 : SUCCESS in 43m 11s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/d81ff6951fba43e2a96ec5bfee0929a8 : SUCCESS in 50m 23s","accounts_in_message":[],"_revision_number":2},{"id":"85141b7a5364818b52c44aa14e3c2f32d9b519a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 09:51:14.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5290f1e4de2f7b76e4befc259a90ad53587add7a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 11:04:26.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3042530342c4c13a4b3ae39898019b0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8460d792335c4f9ca4d2ca853ca9829f : SUCCESS in 5m 23s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/9945896aafc44af493fb787681c6ff38 : SUCCESS in 5m 40s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/731612ba4b2e453daae67e95d9a8c923 : SUCCESS in 7m 21s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/e01869319e1f43a6bad79578702a2907 : SUCCESS in 7m 38s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/747bb380d4744600961a50cc2adbc6d4 : SUCCESS in 1h 11m 38s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/62b4dae6c2df4e5888ad5d65933ad627 : SUCCESS in 52m 20s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/9dc7c9078b7543499366756ea40e4b12 : SUCCESS in 1h 00m 33s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/73fc632ef0684e508c6518170c116f2b : SUCCESS in 33m 16s","accounts_in_message":[],"_revision_number":2},{"id":"e0031902f2812b0eab2f96b84ed064cc0dc04e6f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 11:04:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"b6e9108ee19449b3717482cf9cb18ba8f40dc0b9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 11:05:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b2448979a3a4ca7847d5c9b49917d4e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/012d87f1df454cbb8450efd67f32552c : SUCCESS in 39s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bb19450e5b195fa9416a0704b5819314e66777ec","revisions":{"85a0b5a73995c02d1bf4134e7a0cb6e582f908e6":{"kind":"REWORK","_number":1,"created":"2026-01-18 20:39:31.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/42/973742/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/42/973742/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/42/973742/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/42/973742/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/42/973742/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/42/973742/1"}}},"commit":{"parents":[{"commit":"6733040820533bc7f2360a2225af18d3ffcd1ee7","subject":"Merge \"Tags, server replace to instance\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6733040820533bc7f2360a2225af18d3ffcd1ee7"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-18 20:39:31.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-18 20:39:31.000000000","tz":0},"subject":"Same subnet with LB use Yes in UI","message":"Same subnet with LB use Yes in UI\n\nChange-Id: I11c3e584b9ad2eaad718433e42e29af40121ffdc\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/85a0b5a73995c02d1bf4134e7a0cb6e582f908e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/85a0b5a73995c02d1bf4134e7a0cb6e582f908e6"}]},"branch":"refs/heads/master"},"bb19450e5b195fa9416a0704b5819314e66777ec":{"kind":"REWORK","_number":2,"created":"2026-01-18 20:41:01.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/42/973742/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/42/973742/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/42/973742/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/42/973742/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/42/973742/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/42/973742/2"}}},"commit":{"parents":[{"commit":"6733040820533bc7f2360a2225af18d3ffcd1ee7","subject":"Merge \"Tags, server replace to instance\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6733040820533bc7f2360a2225af18d3ffcd1ee7"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-18 20:39:31.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-18 20:41:01.000000000","tz":0},"subject":"Same subnet with LB use Yes in UI","message":"Same subnet with LB use Yes in UI\n\nChange-Id: I11c3e584b9ad2eaad718433e42e29af40121ffdc\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/bb19450e5b195fa9416a0704b5819314e66777ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/bb19450e5b195fa9416a0704b5819314e66777ec"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
