)]}'
{"id":"openstack%2Fskyline-console~973743","triplet_id":"openstack%2Fskyline-console~master~Ibbc20992fa2d27865aceec517a949c89db817a08","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-01-19 09:54:21.000000000","reason":"removed on reply"},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-01-19 03:49:58.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-01-19 10:19:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibbc20992fa2d27865aceec517a949c89db817a08","subject":"Fix with writing RAM and CPU","status":"MERGED","created":"2026-01-18 20:47:58.000000000","updated":"2026-01-19 10:20:26.000000000","submitted":"2026-01-19 10:19:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"973743","meta_rev_id":"1b5104fd7914193bdaf9274a1e721e991d56b14d","_number":973743,"virtual_id_number":973743,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-19 10:19:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"recommended":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"all":[{"value":1,"date":"2026-01-19 09:54:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":2,"date":"2026-01-19 03:49:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":1,"date":"2026-01-19 03:49:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-18 20:53:05.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"},{"updated":"2026-01-18 20:53:05.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2026-01-18 22:34:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"250e94383b979779225027ba534f1c93e342580e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-18 20:47:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d541eea395f70ef779b6550852deaa644e6efad0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-18 20:51:06.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"c040245b08bf7dbf6ccbcd4a95c60e6ab0c5c5d2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-18 20:52:56.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"cf7b1bf429c765a3a53a1c81fafd7e17eae1ab01","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-18 20:53:05.000000000","message":"Patch Set 3:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"e9c3737cd961d32e27ae51815611c0371f6581c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-18 20:54:47.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"c9055a3e7aa74d59b84c80cbe2ef7fe36c189fb1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-18 20:56:37.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"a4471a4388720bac52073c5406fc956de9da11fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-18 21:04:13.000000000","message":"Patch Set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"6487816dfc931694c299d6373ddeea03c0ea2b50","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-18 21:04:28.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"2f215066ce871f862dbe744917cb579b4bac1b07","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-18 21:06:32.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"c5435072005af273549bc382f1af9423008ad42f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-18 22:34:45.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/629786833f7a4871a1b41e3595c544f4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5444d30fe7b14bea8b405f0aead51239 : SUCCESS in 6m 21s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/660d72f8094845cf8139802485956207 : SUCCESS in 6m 43s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/bf54a468287845b7985ad7ae867fd447 : SUCCESS in 8m 29s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/76c21bd621854a5b80edaa38123c965f : SUCCESS in 5m 06s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/d353aa397c214bc0bb7a6f1655481b17 : SUCCESS in 1h 20m 31s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/98491267273d46f7a53fb78fa338044b : SUCCESS in 58m 53s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/08050163903341dda2f89afe75a55ded : SUCCESS in 52m 45s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/d8dc74462e284d8a97aeee533ee3fa49 : SUCCESS in 48m 19s","accounts_in_message":[],"_revision_number":8},{"id":"d086d7672fae2d71086eb1bd4efdfce1836da3c4","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-01-19 03:49:58.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"62afc5915d975d177bede36d5eee091b2f34db85","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 03:50:34.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"428b969d80f38db819b02b3f20aa3e298b54ae3e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 04:51:48.000000000","message":"Patch Set 8: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/aaf5fea56d974ad4a4f574a4538f0d22\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ecb9c8825374333aeb42c5bcf7a90e9 : SUCCESS in 5m 58s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/40e849a2d472428aab270e148f4b3865 : SUCCESS in 8m 26s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/58519772d7294d1e8fe04cbba0487095 : SUCCESS in 7m 09s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/bc7729d01d0b48e9ac0ed67afcf4075b : SUCCESS in 5m 12s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/0f721e7810fc432fa565995463360480 : FAILURE in 53m 52s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/12721682251441c49e16bc88e281c910 : SUCCESS in 51m 30s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/6465443acdb8499fa599c5bcbe4583b7 : SUCCESS in 53m 21s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/150c47e028c146e1b1cd1d0bab80a4fa : FAILURE in 10m 18s","accounts_in_message":[],"_revision_number":8},{"id":"f4dae9013f5203f4dde67c4b7b6df3bd982e5de6","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-19 07:26:52.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"561101c82f719b4e9d9fac8cc5480b2ab5aaac83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 08:56:59.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ffd06d94585944cc8292f518af0febc3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee771e921e404e5daa1601c25b15d0fe : SUCCESS in 8m 13s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/56b04e2b0ecf4e1f936a1bbe6fbc97d7 : SUCCESS in 4m 29s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/321fc7466be340f3a96db3cd11b5d8ad : SUCCESS in 4m 20s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/ca951de0c47c46f193349c0b20f79adb : SUCCESS in 3m 24s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/c2fa3df9b179489ba91d53e14b709668 : SUCCESS in 1h 24m 17s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/7f02a4e7fa9c443ab4a432a2663d7445 : SUCCESS in 59m 08s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/ece7a452ac2b4852b3f3c3c7aa58e4f2 : SUCCESS in 51m 18s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/635ce403eb8b4b5f9b075fb914a7deaa : SUCCESS in 52m 11s","accounts_in_message":[],"_revision_number":8},{"id":"0fb8c4616b280b23d10ce32549af952cfcb49c50","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 08:57:29.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"9a9b61b7bf9506d6e671cf710b29307e8e057e0c","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-19 09:54:21.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"245b9a840b1ea78d847e8956b440625cc44024a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 10:19:00.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4cb4b820fb82482da859b259a2985a4a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1677b0d1ffa14ca18651d14a8f0e37c7 : SUCCESS in 5m 52s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/5198ab0ec354451ebaae8e73b04fbddb : SUCCESS in 12m 31s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/84693f913d1e4d87b2aff27559efe6ea : SUCCESS in 6m 47s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/87d359e7f65742eab17815dc6ec39206 : SUCCESS in 4m 35s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/707db8f7e06c446889c7a16b10588517 : SUCCESS in 1h 20m 24s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/efa364b2bc504700abdc4b763b606069 : SUCCESS in 58m 20s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/e00f5b627c50430bbcaf01a705eca8a5 : SUCCESS in 33m 23s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/da99338ee14146e1a3c56c34a1fb40d4 : SUCCESS in 30m 51s","accounts_in_message":[],"_revision_number":8},{"id":"bbe24c3241b7c8b03bb626c18004b1e5c91c607c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 10:19:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"1b5104fd7914193bdaf9274a1e721e991d56b14d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 10:20:26.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/437ec3448d51445ba626c3037c762313\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a1e7856f5c24a8eaf8b4a151b665084 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"b0dad6943cfdf7575c33db14a48fb928566f7058","revisions":{"fc4b7ea2eedc001d158b03d6c61a048186fe2755":{"kind":"REWORK","_number":1,"created":"2026-01-18 20:47:58.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/43/973743/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/43/973743/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/973743/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/973743/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/973743/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/43/973743/1"}}},"commit":{"parents":[{"commit":"6733040820533bc7f2360a2225af18d3ffcd1ee7","subject":"Merge \"Tags, server replace to instance\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6733040820533bc7f2360a2225af18d3ffcd1ee7"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-18 20:47:58.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-18 20:47:58.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: Ibbc20992fa2d27865aceec517a949c89db817a08\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/fc4b7ea2eedc001d158b03d6c61a048186fe2755"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/fc4b7ea2eedc001d158b03d6c61a048186fe2755"}]},"branch":"refs/heads/master"},"af0ff02f4f9659336433f79386ddadc7e85c4f2f":{"kind":"REWORK","_number":2,"created":"2026-01-18 20:51:06.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/43/973743/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/43/973743/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/973743/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/973743/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/973743/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/43/973743/2"}}},"commit":{"parents":[{"commit":"6733040820533bc7f2360a2225af18d3ffcd1ee7","subject":"Merge \"Tags, server replace to instance\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6733040820533bc7f2360a2225af18d3ffcd1ee7"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-18 20:47:58.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-18 20:51:06.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: Ibbc20992fa2d27865aceec517a949c89db817a08\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/af0ff02f4f9659336433f79386ddadc7e85c4f2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/af0ff02f4f9659336433f79386ddadc7e85c4f2f"}]},"branch":"refs/heads/master"},"0fd53d0068dc64a53b084f4494d770d2606608b8":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-01-18 20:52:56.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/43/973743/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/43/973743/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/973743/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/973743/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/973743/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/43/973743/3"}}},"commit":{"parents":[{"commit":"6733040820533bc7f2360a2225af18d3ffcd1ee7","subject":"Merge \"Tags, server replace to instance\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6733040820533bc7f2360a2225af18d3ffcd1ee7"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-18 20:47:58.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-18 20:52:56.000000000","tz":0},"subject":"Fix with writing RAM and CPU","message":"Fix with writing RAM and CPU\n\nChange-Id: Ibbc20992fa2d27865aceec517a949c89db817a08\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/0fd53d0068dc64a53b084f4494d770d2606608b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/0fd53d0068dc64a53b084f4494d770d2606608b8"}]},"branch":"refs/heads/master","description":"Edit commit message"},"ddca830f720493eb96fd8e7b0cd8b4be8d55f6cb":{"kind":"REWORK","_number":4,"created":"2026-01-18 20:54:47.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/43/973743/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/43/973743/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/973743/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/973743/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/973743/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/43/973743/4"}}},"commit":{"parents":[{"commit":"6733040820533bc7f2360a2225af18d3ffcd1ee7","subject":"Merge \"Tags, server replace to instance\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6733040820533bc7f2360a2225af18d3ffcd1ee7"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-18 20:47:58.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-18 20:54:46.000000000","tz":0},"subject":"Fix with writing RAM and CPU","message":"Fix with writing RAM and CPU\n\nChange-Id: Ibbc20992fa2d27865aceec517a949c89db817a08\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ddca830f720493eb96fd8e7b0cd8b4be8d55f6cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ddca830f720493eb96fd8e7b0cd8b4be8d55f6cb"}]},"branch":"refs/heads/master"},"af234aa7d9d40a58f2190945eb89a860de6ca346":{"kind":"REWORK","_number":5,"created":"2026-01-18 20:56:37.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/43/973743/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/43/973743/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/973743/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/973743/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/973743/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/43/973743/5"}}},"commit":{"parents":[{"commit":"6733040820533bc7f2360a2225af18d3ffcd1ee7","subject":"Merge \"Tags, server replace to instance\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6733040820533bc7f2360a2225af18d3ffcd1ee7"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-18 20:47:58.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-18 20:56:37.000000000","tz":0},"subject":"Fix with writing RAM and CPU","message":"Fix with writing RAM and CPU\n\nChange-Id: Ibbc20992fa2d27865aceec517a949c89db817a08\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/af234aa7d9d40a58f2190945eb89a860de6ca346"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/af234aa7d9d40a58f2190945eb89a860de6ca346"}]},"branch":"refs/heads/master"},"415c34c946b0d386e6de651ca48c529fb4e0c418":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2026-01-18 21:04:13.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/43/973743/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/43/973743/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/973743/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/973743/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/973743/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/43/973743/6"}}},"commit":{"parents":[{"commit":"6733040820533bc7f2360a2225af18d3ffcd1ee7","subject":"Merge \"Tags, server replace to instance\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6733040820533bc7f2360a2225af18d3ffcd1ee7"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-18 20:47:58.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-18 21:04:13.000000000","tz":0},"subject":"Fix with writing RAM, CPU and IP","message":"Fix with writing RAM, CPU and IP\n\nChange-Id: Ibbc20992fa2d27865aceec517a949c89db817a08\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/415c34c946b0d386e6de651ca48c529fb4e0c418"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/415c34c946b0d386e6de651ca48c529fb4e0c418"}]},"branch":"refs/heads/master","description":"Edit commit message"},"86058d11b320b5c095be504faa789ca346eabf62":{"kind":"REWORK","_number":7,"created":"2026-01-18 21:04:28.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/43/973743/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/43/973743/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/973743/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/973743/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/973743/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/43/973743/7"}}},"commit":{"parents":[{"commit":"6733040820533bc7f2360a2225af18d3ffcd1ee7","subject":"Merge \"Tags, server replace to instance\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6733040820533bc7f2360a2225af18d3ffcd1ee7"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-18 20:47:58.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-18 21:04:28.000000000","tz":0},"subject":"Fix with writing RAM, CPU and IP","message":"Fix with writing RAM, CPU and IP\n\nChange-Id: Ibbc20992fa2d27865aceec517a949c89db817a08\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/86058d11b320b5c095be504faa789ca346eabf62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/86058d11b320b5c095be504faa789ca346eabf62"}]},"branch":"refs/heads/master"},"b0dad6943cfdf7575c33db14a48fb928566f7058":{"kind":"REWORK","_number":8,"created":"2026-01-18 21:06:32.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/43/973743/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/43/973743/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/973743/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/973743/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/43/973743/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/43/973743/8"}}},"commit":{"parents":[{"commit":"6733040820533bc7f2360a2225af18d3ffcd1ee7","subject":"Merge \"Tags, server replace to instance\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6733040820533bc7f2360a2225af18d3ffcd1ee7"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-18 20:47:58.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-18 21:06:32.000000000","tz":0},"subject":"Fix with writing RAM and CPU","message":"Fix with writing RAM and CPU\n\nChange-Id: Ibbc20992fa2d27865aceec517a949c89db817a08\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b0dad6943cfdf7575c33db14a48fb928566f7058"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b0dad6943cfdf7575c33db14a48fb928566f7058"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
