)]}'
{"id":"openstack%2Fskyline-console~973744","triplet_id":"openstack%2Fskyline-console~master~I0c5b68000cd75d57d6e3dee8ab33a9da0be90f22","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-01-20 02:59:02.000000000","reason":"Change was submitted"},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-01-20 01:14:58.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-01-20 02:59:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0c5b68000cd75d57d6e3dee8ab33a9da0be90f22","subject":"Confirm form, \"confirm\" button update to OK","status":"MERGED","created":"2026-01-18 21:13:00.000000000","updated":"2026-01-20 03:00:12.000000000","submitted":"2026-01-20 02:59:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"973744","meta_rev_id":"1d1980d5e42e3d2f0cf1b5054448187e26eb7a0c","_number":973744,"virtual_id_number":973744,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-20 02:59:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-01-20 01:14:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-01-20 01:14:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-18 21:15:17.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"},{"updated":"2026-01-18 21:15:25.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2026-01-18 22:10:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"354bd464c7a921e11c352b60695cceedc849f59d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-18 21:13:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bb87adbd74bfc527ac807bf8e58667ec84c21f15","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-18 21:14:34.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"fac75ba3ed1dd55e2890210a9d3e8043b7802046","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-18 21:15:01.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"c79e3b39eebe6d972c25fadb90c766cca6e9eb7c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-18 21:15:05.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"be0d88e9f3e1040d503e4d99aa9f39007a381ae0","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-18 21:15:17.000000000","message":"Patch Set 4:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":4},{"id":"c7a93671e10ad36da090b56cbbebda94af2e822e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-18 22:10:52.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ab1490c138ad49e78e12c2fdb8861a07\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fe3113a9f9b489f938c814cef174040 : SUCCESS in 5m 19s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/5fa3f591661a49daa51aa2523d0b7df1 : SUCCESS in 6m 46s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/bbbdf9ec107a4e41a9110e5d45d7cbf6 : SUCCESS in 7m 41s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/9a41f601ac0944c59f6c795843a43e58 : SUCCESS in 7m 33s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/3aee5328805348cdb52174e70b6832f9 : FAILURE in 53m 00s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/b2641bed8924442bac983b67f8b50bee : SUCCESS in 33m 02s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/7ce6d3b929384f4397f91b7d426f5e86 : SUCCESS in 51m 41s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/89edcdd7a9b24f529d96186bcb7f29fe : SUCCESS in 52m 35s","accounts_in_message":[],"_revision_number":4},{"id":"e788f7ad6080c2ec6aad3633d66cf4f7137fe1e1","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-19 09:15:06.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8a7c0b2ae7332cadf7916dad8426397c32cc5e6b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-19 09:15:41.000000000","message":"Patch Set 5: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"3f46e88e9cec8c696a8a65693b2fe1de5b339793","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 10:34:41.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf47e43113bd436cb9eb95b9cdececf2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce4c9c67dfc241fe9b1b4866869f9a78 : SUCCESS in 3m 21s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/4b6565a21b3c486b8630f51a29240677 : SUCCESS in 10m 45s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/a1031d15d2fe4065aa72d35066689419 : SUCCESS in 3m 19s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/eb6d0568433b43d29e8d54e0b142878d : SUCCESS in 4m 39s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/8b6d7a2ef9bd496f964ea98b626d192d : SUCCESS in 51m 04s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/c748acfb4bde4ff591107519b2c96e4d : SUCCESS in 1h 03m 34s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/8e2904c39b6a4725809ae1c7afaa3eba : SUCCESS in 28m 25s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/b84b818b71c24d2e80f6253ef429b99d : SUCCESS in 49m 21s","accounts_in_message":[],"_revision_number":5},{"id":"0d759698239ce6465941a3cbd66b3ae76bfd8186","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-01-20 01:14:58.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"1590ea440977a07e03c7c8eaf1585482cf710d61","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 01:15:28.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"872a5f725f0d8aa7849a44455f86cfc994e9c9fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 02:59:02.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b944b785bf246378b26486dbf755f4c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/825d810875de446b930cad849f8c65a8 : SUCCESS in 6m 25s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/f72ba2123a79476fa1b840744716bc69 : SUCCESS in 10m 46s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/be67300d4bc74cbf8e095ac426276cf8 : SUCCESS in 7m 19s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/965e3d7958f240c4b470364445043aab : SUCCESS in 7m 24s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/9a3e01a85fe4489ba20e05183c7f29d2 : SUCCESS in 1h 27m 53s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/69a04ad94b5248e6ada72db5a1994a52 : SUCCESS in 1h 01m 08s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/707af97d4368412f8a3775c776f70ee5 : SUCCESS in 38m 22s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/aaffd7b46cb04443ad4b3ab1ce605059 : SUCCESS in 38m 31s","accounts_in_message":[],"_revision_number":5},{"id":"22b3ab8fc80b3d624058220eccf63d0a82a875be","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 02:59:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"1d1980d5e42e3d2f0cf1b5054448187e26eb7a0c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 03:00:12.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4861583899a344aca208d2c6a85875e8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2d2d6c4a069432782bd4ece7315e193 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"e429f5b78143e55c850573c3bd860a90959f7da1","revisions":{"fcc8c3b383a399cb47ffd1621a6d601888105b21":{"kind":"REWORK","_number":1,"created":"2026-01-18 21:13:00.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/44/973744/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/44/973744/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/44/973744/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/44/973744/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/44/973744/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/44/973744/1"}}},"commit":{"parents":[{"commit":"6733040820533bc7f2360a2225af18d3ffcd1ee7","subject":"Merge \"Tags, server replace to instance\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6733040820533bc7f2360a2225af18d3ffcd1ee7"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-18 21:13:00.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-18 21:13:00.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: I0c5b68000cd75d57d6e3dee8ab33a9da0be90f22\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/fcc8c3b383a399cb47ffd1621a6d601888105b21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/fcc8c3b383a399cb47ffd1621a6d601888105b21"}]},"branch":"refs/heads/master"},"92a9812f73f9d880da6eb3d5c1dcfc3817e64dac":{"kind":"REWORK","_number":2,"created":"2026-01-18 21:14:34.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/44/973744/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/44/973744/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/44/973744/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/44/973744/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/44/973744/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/44/973744/2"}}},"commit":{"parents":[{"commit":"6733040820533bc7f2360a2225af18d3ffcd1ee7","subject":"Merge \"Tags, server replace to instance\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6733040820533bc7f2360a2225af18d3ffcd1ee7"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-18 21:13:00.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-18 21:14:34.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: I0c5b68000cd75d57d6e3dee8ab33a9da0be90f22\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/92a9812f73f9d880da6eb3d5c1dcfc3817e64dac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/92a9812f73f9d880da6eb3d5c1dcfc3817e64dac"}]},"branch":"refs/heads/master"},"231d71e140e6a0b93213442148fb6b2b73279fdf":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-01-18 21:15:01.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/44/973744/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/44/973744/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/44/973744/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/44/973744/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/44/973744/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/44/973744/3"}}},"commit":{"parents":[{"commit":"6733040820533bc7f2360a2225af18d3ffcd1ee7","subject":"Merge \"Tags, server replace to instance\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6733040820533bc7f2360a2225af18d3ffcd1ee7"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-18 21:13:00.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-18 21:15:01.000000000","tz":0},"subject":"Confirm form, button confirm update to OK","message":"Confirm form, button confirm update to OK\n\n\nChange-Id: I0c5b68000cd75d57d6e3dee8ab33a9da0be90f22\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/231d71e140e6a0b93213442148fb6b2b73279fdf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/231d71e140e6a0b93213442148fb6b2b73279fdf"}]},"branch":"refs/heads/master","description":"Edit commit message"},"54d509a9dd3b76ceb597c7693378a0a928065401":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2026-01-18 21:15:05.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/44/973744/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/44/973744/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/44/973744/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/44/973744/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/44/973744/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/44/973744/4"}}},"commit":{"parents":[{"commit":"6733040820533bc7f2360a2225af18d3ffcd1ee7","subject":"Merge \"Tags, server replace to instance\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6733040820533bc7f2360a2225af18d3ffcd1ee7"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-18 21:13:00.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-18 21:15:05.000000000","tz":0},"subject":"Confirm form, button confirm update to OK","message":"Confirm form, button confirm update to OK\n\nChange-Id: I0c5b68000cd75d57d6e3dee8ab33a9da0be90f22\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/54d509a9dd3b76ceb597c7693378a0a928065401"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/54d509a9dd3b76ceb597c7693378a0a928065401"}]},"branch":"refs/heads/master","description":"Edit commit message"},"e429f5b78143e55c850573c3bd860a90959f7da1":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2026-01-19 09:15:41.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/44/973744/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/44/973744/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/44/973744/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/44/973744/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/44/973744/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/44/973744/5"}}},"commit":{"parents":[{"commit":"6733040820533bc7f2360a2225af18d3ffcd1ee7","subject":"Merge \"Tags, server replace to instance\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6733040820533bc7f2360a2225af18d3ffcd1ee7"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-18 21:13:00.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-19 09:15:41.000000000","tz":0},"subject":"Confirm form, \"confirm\" button update to OK","message":"Confirm form, \"confirm\" button update to OK\n\nChange-Id: I0c5b68000cd75d57d6e3dee8ab33a9da0be90f22\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e429f5b78143e55c850573c3bd860a90959f7da1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e429f5b78143e55c850573c3bd860a90959f7da1"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
