)]}'
{"id":"openstack%2Fskyline-console~973807","triplet_id":"openstack%2Fskyline-console~master~I7fece0cee41a2f860eb4ca0cb1cdae1ab95858b9","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-01-20 02:59:12.000000000","reason":"Change was submitted"},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-01-20 01:20:29.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-01-20 02:59:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7fece0cee41a2f860eb4ca0cb1cdae1ab95858b9","subject":"Fix for VPN","status":"MERGED","created":"2026-01-19 12:01:26.000000000","updated":"2026-01-20 02:59:12.000000000","submitted":"2026-01-20 02:59:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"973807","meta_rev_id":"f90e2d5e1376b5a91a867ee8e7ccc277ceb0edbc","_number":973807,"virtual_id_number":973807,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-20 02:59:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-01-20 01:20:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-01-20 01:20:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-19 12:06:02.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2026-01-19 12:06:02.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"},{"updated":"2026-01-19 12:42:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a91662e6410df76486080cb8a5ad777a4df912f1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-19 12:01:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6a5ecf70f6a55818833bf6fc84952f211339f70d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-19 12:02:30.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"50df0847002f9144e10ddcda69a37cd07884cf41","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-19 12:03:30.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"e334f083efb4e8ed067196e1ce66a6a44d5b246f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-19 12:04:27.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"fe86dcaefc58ac2e6acea71793e27e1994775c17","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-19 12:05:54.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"2dae626fd0526268e2e0d1c9a891bf8848446ab1","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-19 12:06:02.000000000","message":"Patch Set 5:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":5},{"id":"9f2741f43509f6cec256c924461167184d97f18b","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-19 12:11:53.000000000","message":"Patch Set 5:\n\n(5 comments)","accounts_in_message":[],"_revision_number":5},{"id":"8c6d6adaaba88c01d825b5647b55c1fb6094a74d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 12:42:05.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4dc17a048a84e7c9531bc126dd33c7a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22fa5cc0390d4a90a8b3b0d6be4d772c : SUCCESS in 6m 49s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/44976a4078bd44478cf9a3bd3089ddc9 : SUCCESS in 10m 46s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/d0b636d8365b4837a729a5ab14eb4334 : SUCCESS in 7m 40s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/2beae1fd8d65407e819790c194794ef1 : SUCCESS in 7m 20s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/2da472d084ed48db869e4fbb22d353bf : SUCCESS in 35m 16s","accounts_in_message":[],"_revision_number":5},{"id":"d765ff027df1f4f14cb7a18c2429fdd042191724","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-19 19:39:00.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"f207731f9213100e68f1cdc19543b3b68c3c94d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-19 19:39:16.000000000","message":"Patch Set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"3c968a9226cd1862cfafd8d642c179c361509a91","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-19 19:39:19.000000000","message":"Patch Set 7:\n\n(5 comments)","accounts_in_message":[],"_revision_number":7},{"id":"c6cb0592f244c5dbceb27c221dbff72cf0b536a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 20:26:49.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d000c66d2cc4ee6b15d9664b97aad1b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb4a04e87b214168b4fe062a731909cd : SUCCESS in 5m 35s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/b57fac64447e4a5186624b127e5af73c : SUCCESS in 4m 13s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/814405d6d224408784166ae8f731f541 : SUCCESS in 3m 17s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/a90ba78f942f42f28d35d38757cf89e9 : SUCCESS in 6m 44s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/da63bc00bd544593878fda65f14afc06 : SUCCESS in 46m 09s","accounts_in_message":[],"_revision_number":7},{"id":"217650fc6ef73e73c524b559c1ae257bb08dd526","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-01-20 01:20:29.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"dcd2df04dba649ff65538e8de57e0bd8696401d2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 01:21:04.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"0950ecd68b093e44cd36280cd0a99860870ec431","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 02:59:12.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50afa13d69544da98e94c3151f21829a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ef8108ee791479f8ce8d0e40561981f : SUCCESS in 3m 24s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/9f9f54323b1c420cadfd54156eb00e01 : SUCCESS in 4m 32s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/43a62f566e604bacb0944aad1ee8a4bd : SUCCESS in 6m 08s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/59aaf94a8e7041f0ad88c70b10c57f7c : SUCCESS in 5m 35s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/2500212f288f41d0bc156e9fe8019e6a : SUCCESS in 58m 09s","accounts_in_message":[],"_revision_number":7},{"id":"f90e2d5e1376b5a91a867ee8e7ccc277ceb0edbc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 02:59:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"f87b2adf86b6a35e6637179843e32189e28578b0","revisions":{"ff75dc59c9b68fe26e2f76c44602f36c2ca65147":{"kind":"REWORK","_number":1,"created":"2026-01-19 12:01:26.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/07/973807/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/07/973807/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973807/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973807/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973807/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/07/973807/1"}}},"commit":{"parents":[{"commit":"eb22f1592fd54c7faeb8363bbdd1fc5b77fa2b52","subject":"Merge \"Fix with writing RAM and CPU\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/eb22f1592fd54c7faeb8363bbdd1fc5b77fa2b52"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-19 12:01:26.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-19 12:01:26.000000000","tz":0},"subject":"Fix for VPN","message":"Fix for VPN\n\ntrivial fix\n\nChange-Id: I7fece0cee41a2f860eb4ca0cb1cdae1ab95858b9\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ff75dc59c9b68fe26e2f76c44602f36c2ca65147"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ff75dc59c9b68fe26e2f76c44602f36c2ca65147"}]},"branch":"refs/heads/master"},"afebd83df5f71bc55647c0c3347c4567e5a96381":{"kind":"REWORK","_number":2,"created":"2026-01-19 12:02:30.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/07/973807/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/07/973807/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973807/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973807/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973807/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/07/973807/2"}}},"commit":{"parents":[{"commit":"eb22f1592fd54c7faeb8363bbdd1fc5b77fa2b52","subject":"Merge \"Fix with writing RAM and CPU\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/eb22f1592fd54c7faeb8363bbdd1fc5b77fa2b52"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-19 12:01:26.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-19 12:02:29.000000000","tz":0},"subject":"Fix for VPN","message":"Fix for VPN\n\ntrivial fix\n\nChange-Id: I7fece0cee41a2f860eb4ca0cb1cdae1ab95858b9\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/afebd83df5f71bc55647c0c3347c4567e5a96381"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/afebd83df5f71bc55647c0c3347c4567e5a96381"}]},"branch":"refs/heads/master"},"ee042d3f1d7422338c800ccdca8eda478ca76974":{"kind":"REWORK","_number":3,"created":"2026-01-19 12:03:30.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/07/973807/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/07/973807/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973807/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973807/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973807/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/07/973807/3"}}},"commit":{"parents":[{"commit":"eb22f1592fd54c7faeb8363bbdd1fc5b77fa2b52","subject":"Merge \"Fix with writing RAM and CPU\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/eb22f1592fd54c7faeb8363bbdd1fc5b77fa2b52"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-19 12:01:26.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-19 12:03:30.000000000","tz":0},"subject":"Fix for VPN","message":"Fix for VPN\n\ntrivial fix\n\nChange-Id: I7fece0cee41a2f860eb4ca0cb1cdae1ab95858b9\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ee042d3f1d7422338c800ccdca8eda478ca76974"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ee042d3f1d7422338c800ccdca8eda478ca76974"}]},"branch":"refs/heads/master"},"2973102edf3c97170d11a53747535557560c7827":{"kind":"REWORK","_number":4,"created":"2026-01-19 12:04:27.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/07/973807/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/07/973807/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973807/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973807/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973807/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/07/973807/4"}}},"commit":{"parents":[{"commit":"eb22f1592fd54c7faeb8363bbdd1fc5b77fa2b52","subject":"Merge \"Fix with writing RAM and CPU\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/eb22f1592fd54c7faeb8363bbdd1fc5b77fa2b52"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-19 12:01:26.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-19 12:04:27.000000000","tz":0},"subject":"Fix for VPN","message":"Fix for VPN\n\ntrivial fix\n\nChange-Id: I7fece0cee41a2f860eb4ca0cb1cdae1ab95858b9\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/2973102edf3c97170d11a53747535557560c7827"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/2973102edf3c97170d11a53747535557560c7827"}]},"branch":"refs/heads/master"},"a9c47b9a5299092a1fc99dcd4eafa423c3074008":{"kind":"REWORK","_number":5,"created":"2026-01-19 12:05:54.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/07/973807/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/07/973807/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973807/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973807/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973807/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/07/973807/5"}}},"commit":{"parents":[{"commit":"eb22f1592fd54c7faeb8363bbdd1fc5b77fa2b52","subject":"Merge \"Fix with writing RAM and CPU\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/eb22f1592fd54c7faeb8363bbdd1fc5b77fa2b52"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-19 12:01:26.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-19 12:05:53.000000000","tz":0},"subject":"Fix for VPN","message":"Fix for VPN\n\ntrivial fix\n\nChange-Id: I7fece0cee41a2f860eb4ca0cb1cdae1ab95858b9\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a9c47b9a5299092a1fc99dcd4eafa423c3074008"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a9c47b9a5299092a1fc99dcd4eafa423c3074008"}]},"branch":"refs/heads/master"},"07df056a0030df35fd4b90a7008d69fce08a285c":{"kind":"REWORK","_number":6,"created":"2026-01-19 19:39:00.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/07/973807/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/07/973807/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973807/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973807/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973807/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/07/973807/6"}}},"commit":{"parents":[{"commit":"eb22f1592fd54c7faeb8363bbdd1fc5b77fa2b52","subject":"Merge \"Fix with writing RAM and CPU\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/eb22f1592fd54c7faeb8363bbdd1fc5b77fa2b52"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-19 12:01:26.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-19 19:38:56.000000000","tz":0},"subject":"Fix for VPN","message":"Fix for VPN\n\ntrivial fix\n\nChange-Id: I7fece0cee41a2f860eb4ca0cb1cdae1ab95858b9\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/07df056a0030df35fd4b90a7008d69fce08a285c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/07df056a0030df35fd4b90a7008d69fce08a285c"}]},"branch":"refs/heads/master"},"f87b2adf86b6a35e6637179843e32189e28578b0":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2026-01-19 19:39:16.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/07/973807/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/07/973807/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973807/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973807/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/07/973807/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/07/973807/7"}}},"commit":{"parents":[{"commit":"eb22f1592fd54c7faeb8363bbdd1fc5b77fa2b52","subject":"Merge \"Fix with writing RAM and CPU\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/eb22f1592fd54c7faeb8363bbdd1fc5b77fa2b52"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-19 12:01:26.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-19 19:39:16.000000000","tz":0},"subject":"Fix for VPN","message":"Fix for VPN\n\ntrivial fix\n\nChange-Id: I7fece0cee41a2f860eb4ca0cb1cdae1ab95858b9\nCo-authored-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f87b2adf86b6a35e6637179843e32189e28578b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f87b2adf86b6a35e6637179843e32189e28578b0"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
