)]}'
{"id":"openstack%2Fskyline-console~973848","triplet_id":"openstack%2Fskyline-console~master~I8a0b924f07d3d564e09caa24872687f7da80e64b","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-01-20 02:59:11.000000000","reason":"Change was submitted"},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-01-20 01:19:03.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}},"hashtags":[],"change_id":"I8a0b924f07d3d564e09caa24872687f7da80e64b","subject":"Rename \"Console\" Page to \"Project\" Page","status":"MERGED","created":"2026-01-19 15:32:48.000000000","updated":"2026-01-20 02:59:11.000000000","submitted":"2026-01-20 02:59:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"973848","meta_rev_id":"d51fb3f38cde5893b2c44893ac1f2ab1d00631c7","_number":973848,"virtual_id_number":973848,"owner":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-20 02:59:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-01-20 01:19:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-01-19 19:37:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-01-20 01:19:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-19 16:56:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-19 19:37:57.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-01-19 19:37:57.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"}],"messages":[{"id":"2db729f7d5f52d7d42c40d178717b7454f0dabf7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-19 15:32:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"31715a960a3627949c4b1ccd58a58bae982ce657","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-19 15:47:46.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"b45bd5b6e89c488de5bb2cbf9f74e254df45c925","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-19 15:47:54.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"a5d8eca54a807d77d6b4d8ec2da8cee87f987293","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-19 16:56:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e15d8cb1faf14f0599ad3829a688d205\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c1fc068b3a947fea92e1ad782fa36f7 : SUCCESS in 3m 19s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/a03663c7ff274fe88e307189ef9c7cf9 : SUCCESS in 10m 57s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/cf3f4b43c3824036be04ff8a076c66d1 : SUCCESS in 3m 28s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/6660f86eec5547238df96e5d1488372b : SUCCESS in 4m 39s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/aef3c09c41d34d268b042513bdaa94a5 : SUCCESS in 1h 07m 21s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/54e93e10ddd442e1bd75cee404ae040f : SUCCESS in 34m 32s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/2b61b2ade438452fa09b1c081089507e : SUCCESS in 54m 24s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/575a490ffbaa4431bc304881632b91f6 : SUCCESS in 31m 20s","accounts_in_message":[],"_revision_number":2},{"id":"73872950b8f7f807e6f585c25ac74e7acf4ca54c","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-19 19:37:57.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d77811f63e107b9f8fb6cd2adc28e283581f5ae8","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-01-20 01:19:03.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d30e3e995eafbd535e1b415fa32b8d9b6da7dd5c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 01:19:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"844cdd251ab62db0860066f421084012b5fee2fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 02:59:10.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2e451005cc4410ebb46f0e7a53f683a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/129d1101b33c4f9db79b996f941d9a44 : SUCCESS in 5m 18s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/f35ea11c173e425490bc3994c55139a7 : SUCCESS in 12m 30s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/82cfbbe96e4c461c88b0462f88c3c84e : SUCCESS in 5m 25s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/8647e025955a4338b34cc0f51163a261 : SUCCESS in 7m 40s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/5c3620285fd34847ac890b5ebc7ca193 : SUCCESS in 1h 06m 23s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/a79d081cf86545ab887e8172b8c62c73 : SUCCESS in 30m 55s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/1c0e5dce3a3b45b39acca61155f18d40 : SUCCESS in 40m 00s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/3e744753599e4e5c94ac111c90766123 : SUCCESS in 52m 07s","accounts_in_message":[],"_revision_number":2},{"id":"d51fb3f38cde5893b2c44893ac1f2ab1d00631c7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 02:59:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e9aaf8af6905c6bca357f4fca6d2293e48479be4","revisions":{"ccf7a6c12b53c6ae5729a3dbe631283df9fa3e50":{"kind":"REWORK","_number":1,"created":"2026-01-19 15:32:48.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/48/973848/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/48/973848/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/48/973848/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/48/973848/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/48/973848/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/48/973848/1"}}},"commit":{"parents":[{"commit":"eb22f1592fd54c7faeb8363bbdd1fc5b77fa2b52","subject":"Merge \"Fix with writing RAM and CPU\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/eb22f1592fd54c7faeb8363bbdd1fc5b77fa2b52"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-19 15:32:48.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-19 15:32:48.000000000","tz":0},"subject":"Rename \"Console\" Page to \"Project\" Page","message":"Rename \"Console\" Page to \"Project\" Page\n\nThe term Console has historically caused confusion, as it does not clearly represent a Keystone project term and is often associated with VM access. Using Project will \nimprove clarity and reduce user misunderstandings.\n\nChange-Id: I8a0b924f07d3d564e09caa24872687f7da80e64b\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ccf7a6c12b53c6ae5729a3dbe631283df9fa3e50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ccf7a6c12b53c6ae5729a3dbe631283df9fa3e50"}]},"branch":"refs/heads/master"},"e9aaf8af6905c6bca357f4fca6d2293e48479be4":{"kind":"REWORK","_number":2,"created":"2026-01-19 15:47:46.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/48/973848/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/48/973848/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/48/973848/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/48/973848/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/48/973848/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/48/973848/2"}}},"commit":{"parents":[{"commit":"eb22f1592fd54c7faeb8363bbdd1fc5b77fa2b52","subject":"Merge \"Fix with writing RAM and CPU\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/eb22f1592fd54c7faeb8363bbdd1fc5b77fa2b52"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-19 15:32:48.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-19 15:45:15.000000000","tz":0},"subject":"Rename \"Console\" Page to \"Project\" Page","message":"Rename \"Console\" Page to \"Project\" Page\n\nThe term Console has historically caused confusion, as it does not clearly represent a Keystone project term and is often associated with VM access. Using Project will \nimprove clarity and reduce user misunderstandings.\n\nChange-Id: I8a0b924f07d3d564e09caa24872687f7da80e64b\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e9aaf8af6905c6bca357f4fca6d2293e48479be4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e9aaf8af6905c6bca357f4fca6d2293e48479be4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
