)]}'
{"id":"openstack%2Fskyline-console~973903","triplet_id":"openstack%2Fskyline-console~master~I9beee6c567dd110110d310138fd92adb1c8a26f4","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-01-20 07:53:24.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}},"hashtags":[],"change_id":"I9beee6c567dd110110d310138fd92adb1c8a26f4","subject":"feat: Support default hidden columns","status":"MERGED","created":"2026-01-20 03:52:09.000000000","updated":"2026-01-20 08:59:28.000000000","submitted":"2026-01-20 08:58:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"973903","meta_rev_id":"f07b3b784f29b6ceb2391d2be77778bdcd7f863b","_number":973903,"virtual_id_number":973903,"owner":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-20 08:58:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"recommended":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"all":[{"value":1,"date":"2026-01-20 07:25:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":2,"date":"2026-01-20 07:53:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":1,"date":"2026-01-20 07:53:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-20 04:43:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-20 07:25:57.000000000","updated_by":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"}],"messages":[{"id":"98dfc240b45be952612eb6838763cf547fbbce86","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-01-20 03:52:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a7539a4d545327c14db3f28457362166220e132e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 04:43:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3758697f97e642df98bc385d0b970059\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/206e59b89550476686e7a2b3b6085151 : SUCCESS in 1m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2aa358ea45df4f3ea8353f9ead13e682 : SUCCESS in 3m 22s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/ed428e36185448e981d2c089b660e0fa : SUCCESS in 4m 48s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/f71311bd2438489d85cc48452ec48def : SUCCESS in 3m 22s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/d4bdf03ce9584b61baeac04d95fdbc7e : SUCCESS in 3m 24s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/a4a033234c944e8bbb42bfcfa6e9f3a4 : SUCCESS in 50m 40s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/e556d3e469f4426a818c79fd9f6dd90f : SUCCESS in 32m 40s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/0b463d6532fa4a93912aa31fa3bdb7ce : SUCCESS in 32m 59s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/c344c6fe34da4ac592ff2f135cfcd6ea : SUCCESS in 32m 38s","accounts_in_message":[],"_revision_number":1},{"id":"bafa30e7d9b18623b9c00bc245a0c21baf4e0830","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-01-20 06:05:35.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"79acc0cd504b085944056d4984dc42c5ec07aa3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 07:07:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1acd315511b42459e98c328c5730f54\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4707d9b2e9f940968618f88b7f105d39 : SUCCESS in 2m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e283e643f96c402680994326e353be1a : SUCCESS in 8m 12s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/68e91d1b8a92455da8d64ace6dfbb2a3 : SUCCESS in 11m 00s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/f593ab6d071c418280a4769eecd1b50d : SUCCESS in 7m 31s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/71a4cd34a4e04438bab2c18f7dc710d6 : SUCCESS in 3m 15s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/fe3db77d97be41fc8a1b8031d56a4ad8 : SUCCESS in 49m 44s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/de599e0b8d924826a28584e37b14dcbf : SUCCESS in 31m 40s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/f55edb4595d3413c8b69642e0d4c711e : SUCCESS in 30m 49s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/43721fba842e494ab61eff9f30cc82a9 : SUCCESS in 59m 36s","accounts_in_message":[],"_revision_number":2},{"id":"853498f59dd6d0c8157127b2658f2087c261890d","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-20 07:25:57.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"177950964bb09208b6367dd7d0b56cf73619daf6","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-01-20 07:53:24.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f0d779dd6c3f1edc744974b2c307c74eafc01de2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 07:53:52.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f2fd6a8b3bf92b021f2c1c0750fde89aa5d37b9b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 08:58:35.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/263fd41f365848c3ae1c5872d17d5f85\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/629a7c79c9164a1cb732da110564ee76 : SUCCESS in 1m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a3d4cc24945469d87b4638b87ceb0a3 : SUCCESS in 4m 00s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/0731e0cb9b31430583a1e02904cb0832 : SUCCESS in 7m 33s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/f426ef52f7c24103a64937677ce9bbe6 : SUCCESS in 4m 39s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/c7d352aa2ce64912ae6d0e44438a0a05 : SUCCESS in 6m 41s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/6d8c207231d945afbdb8fecd0dc5eb2f : SUCCESS in 48m 57s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/39ca5b9a8fa3480881a7cfdde540d174 : SUCCESS in 1h 00m 17s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/da374dfa24374d03b29040dfca3abbd2 : SUCCESS in 53m 46s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/35f00f8d8589481cb5c2750f297715b4 : SUCCESS in 50m 28s","accounts_in_message":[],"_revision_number":2},{"id":"be135f1ca38b316f318e87fb0b194b7a6c3fad8f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 08:58:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"f07b3b784f29b6ceb2391d2be77778bdcd7f863b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 08:59:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c900c1a49b36478cbc6df16c23e2767c\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0a3d071f7d4e43faacccc0b0d3469f18 : SUCCESS in 38s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9d7e8bad1024bd5b85e23de6c6090f6 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fafd1366e2383b4ccee09be3b32f4ba7dc50878c","revisions":{"508d10124b8e783406ba7db24f52f1795802ab66":{"kind":"REWORK","_number":1,"created":"2026-01-20 03:52:09.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/03/973903/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/03/973903/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/03/973903/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/03/973903/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/03/973903/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/03/973903/1"}}},"commit":{"parents":[{"commit":"d5a45145061c38e87c1b6a677bf5f9f55f66e8df","subject":"Merge \"Add tooltip for External Gateway option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d5a45145061c38e87c1b6a677bf5f9f55f66e8df"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@algoblu.com","date":"2026-01-20 03:50:05.000000000","tz":480},"committer":{"name":"zhangjingwei","email":"zhang.jingwei@algoblu.com","date":"2026-01-20 03:51:01.000000000","tz":480},"subject":"feat: Support default hidden columns","message":"feat: Support default hidden columns\n\nSet  and  on table columns to hide them by default on initial table load.\n\nChange-Id: I9beee6c567dd110110d310138fd92adb1c8a26f4\nSigned-off-by: Jingwei Zhang \u003czhang.jingwei@algoblu.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/508d10124b8e783406ba7db24f52f1795802ab66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/508d10124b8e783406ba7db24f52f1795802ab66"}]},"branch":"refs/heads/master"},"fafd1366e2383b4ccee09be3b32f4ba7dc50878c":{"kind":"REWORK","_number":2,"created":"2026-01-20 06:05:35.000000000","uploader":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"ref":"refs/changes/03/973903/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/03/973903/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/03/973903/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/03/973903/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/03/973903/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/03/973903/2"}}},"commit":{"parents":[{"commit":"d5a45145061c38e87c1b6a677bf5f9f55f66e8df","subject":"Merge \"Add tooltip for External Gateway option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d5a45145061c38e87c1b6a677bf5f9f55f66e8df"}]}],"author":{"name":"zhangjingwei","email":"zhang.jingwei@algoblu.com","date":"2026-01-20 03:50:05.000000000","tz":480},"committer":{"name":"zhangjingwei","email":"zhang.jingwei@algoblu.com","date":"2026-01-20 06:04:57.000000000","tz":480},"subject":"feat: Support default hidden columns","message":"feat: Support default hidden columns\n\nSet  and  on table columns to hide them by default on initial table load.\n\nChange-Id: I9beee6c567dd110110d310138fd92adb1c8a26f4\nSigned-off-by: Jingwei Zhang \u003czhang.jingwei@algoblu.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/fafd1366e2383b4ccee09be3b32f4ba7dc50878c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/fafd1366e2383b4ccee09be3b32f4ba7dc50878c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
