)]}'
{"id":"openstack%2Fskyline-console~973958","triplet_id":"openstack%2Fskyline-console~master~Id781e0edbdb7f275349161bb1da4bed4cb3c4947","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-01-21 07:12:30.000000000","reason":"\u003cGERRIT_ACCOUNT_37632\u003e replied on the change","reason_account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-01-21 01:53:59.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}},"hashtags":[],"change_id":"Id781e0edbdb7f275349161bb1da4bed4cb3c4947","subject":"Fix spacing after colon in Server Groups label","status":"MERGED","created":"2026-01-20 16:36:03.000000000","updated":"2026-01-21 08:40:33.000000000","submitted":"2026-01-21 08:39:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"973958","meta_rev_id":"42a94e8346dd92bbe242fc36d8acbdbab49440ce","_number":973958,"virtual_id_number":973958,"owner":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-21 08:39:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-01-21 01:53:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-01-20 20:05:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-01-21 01:53:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-20 17:43:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-20 20:05:07.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-01-20 20:05:07.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"}],"messages":[{"id":"5a790512162935b4ff0d6d3b6a18df1287dac382","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-20 16:36:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"92bcb3c25dcb74115d875ee4d9f743f5c1b9a983","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-20 16:39:22.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"9b53d7606f86a57f575c7e3fdd5165ea3b768759","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 17:43:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4601999164c45f4acaa3734eac623bb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26353b2e124f4c9c8599bfa288dd057d : SUCCESS in 6m 28s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/e4612e0037b2475bad718ed299639906 : SUCCESS in 5m 06s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/7a38563f233b4db19933522444a541a0 : SUCCESS in 3m 14s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/f9b7dadc7e644bba9f4564680aba6d1a : SUCCESS in 3m 55s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/67c516b8512a499192fbad45dd66893e : SUCCESS in 55m 34s","accounts_in_message":[],"_revision_number":2},{"id":"f05f5dd7fca6531ece7ce8870c63ffe1f9ab190b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-20 18:21:33.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"448dd24506a324f57c8c5ac0ef05949f65b5eac9","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-20 18:21:35.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":3},{"id":"24ddd689e745443cb5c645dbeae95a1e9705c3a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 19:29:11.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da62314b89424cf78e0e7cf0435712c1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89a3ec104cd540469c23c62c6260edc8 : SUCCESS in 4m 38s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/eee1eb70b654490f91a97a69da070dbe : SUCCESS in 6m 46s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/bcee6880953e420ab0cc795787a10f10 : SUCCESS in 4m 23s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/d7c6d2f5e5294921b674ce18b515a5d7 : SUCCESS in 7m 40s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/144385ea6a424b14b386b5f2299733ff : SUCCESS in 59m 23s","accounts_in_message":[],"_revision_number":3},{"id":"3bb100b8c8587ed447fec25a10214401595478af","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-20 20:05:07.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"79d32651bc024c68aa5c1c2f48082ff9809886d6","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-01-21 01:53:59.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"016f1900bb930b5d794c210419bf68713ee34787","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 01:54:29.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f8d6d5153ead510140bbcbd07822af779c463fd4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 03:29:37.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/035a87ad4cf64dfa98bc733425fe4bf1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31d0c9177a0d4b98ac3cf6ac91716b72 : SUCCESS in 3m 52s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/3b3016d6ac06432f89903aeb58cd2b56 : FAILURE in 3m 21s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/dd36c4a23c0043b9a1a84466689107ab : SUCCESS in 8m 16s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/2d2bf6bc50014148a2270953fb0cc3b1 : SUCCESS in 7m 45s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/1dab4b4f170e4e769f2a3fece392a016 : SUCCESS in 44m 27s","accounts_in_message":[],"_revision_number":3},{"id":"bae2df8d7933cd1b8655c4464e803322166a0128","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-01-21 03:58:48.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9798ffe169fb18d5737021e65db920613d8eda2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 04:42:42.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/558d87a8827742d29a8072b700160c8e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cee83cbd6b946b186aac72a96fc9e1c : SUCCESS in 3m 58s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/1b2be3c2d30e4cec941453652d586c9e : SUCCESS in 4m 47s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/e81b0fe98e15426c86ea37397bd0c61d : SUCCESS in 3m 15s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/6256c9c35b3147fe8650c1d82c541cd5 : SUCCESS in 3m 14s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/7f81ba146ccc4beebe4b81ba434a1095 : FAILURE in 42m 56s","accounts_in_message":[],"_revision_number":3},{"id":"a5d3f5f93366bd01f0aa8360d5374fdb04f92ef7","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-21 07:12:30.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d76ce065b4d90384cee9a2410e67dc02e322f1ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 08:08:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d36beac26de94567bda072800ace979e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/135fd202e43042a59ccdc7166473f00f : SUCCESS in 6m 36s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/c8109049a9074f679482812910c63def : SUCCESS in 10m 18s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/a74b0caa8eb440ca835edb7c34d734a5 : SUCCESS in 3m 09s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/a5df045e730f41f39fc81c6c69d858d7 : SUCCESS in 7m 17s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/6e2b8496c7664318a64c092988b05be3 : SUCCESS in 49m 00s","accounts_in_message":[],"_revision_number":3},{"id":"285333bb4e582cbb6a8c863de391d512d3dcbe24","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 08:08:25.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f2a8faafac517744fe7f2de119f80c875e435d79","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 08:39:40.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03c654597fa7465ab57df5c7092b31f9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5cf74618f7a43ef9a5cd981bd4363d8 : SUCCESS in 7m 17s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/d950ad1d36b74bd0abcfacec909aaa20 : SUCCESS in 11m 15s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/9c37eb787a734852befde2d4fe81c184 : SUCCESS in 3m 12s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/e55254f067b84c17a793ef8cf0c4cc44 : SUCCESS in 7m 28s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/a0d2fa39ae364700abae4b34a6af3237 : SUCCESS in 30m 56s","accounts_in_message":[],"_revision_number":3},{"id":"a8ca80ed4cfb2af42db5699df91e0c52909e6c2b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 08:39:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"42a94e8346dd92bbe242fc36d8acbdbab49440ce","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 08:40:33.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01e5a0c019d74136b2c7e3a61f609cad\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6c19ea800ad49c48cd961c5ff86df06 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ae1eb3987e2a58c09d565457964b73f9f3bd00b6","revisions":{"c327865395bdeffdc47672919827a74a53685680":{"kind":"REWORK","_number":1,"created":"2026-01-20 16:36:03.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/58/973958/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/58/973958/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/58/973958/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/58/973958/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/58/973958/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/58/973958/1"}}},"commit":{"parents":[{"commit":"826a5f7666cab045b0cf84fc99a49a7590cf468b","subject":"Merge \"Same subnet with LB use Yes in UI\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/826a5f7666cab045b0cf84fc99a49a7590cf468b"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-20 16:36:03.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-20 16:36:03.000000000","tz":0},"subject":"WIP:","message":"WIP:\n\nChange-Id: Id781e0edbdb7f275349161bb1da4bed4cb3c4947\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/c327865395bdeffdc47672919827a74a53685680"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/c327865395bdeffdc47672919827a74a53685680"}]},"branch":"refs/heads/master"},"809371fffd8b5d2280e4f4ba6bf309ab72fae780":{"kind":"REWORK","_number":2,"created":"2026-01-20 16:39:22.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/58/973958/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/58/973958/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/58/973958/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/58/973958/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/58/973958/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/58/973958/2"}}},"commit":{"parents":[{"commit":"826a5f7666cab045b0cf84fc99a49a7590cf468b","subject":"Merge \"Same subnet with LB use Yes in UI\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/826a5f7666cab045b0cf84fc99a49a7590cf468b"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-20 16:36:03.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-20 16:38:09.000000000","tz":0},"subject":"WIP:","message":"WIP:\n\nChange-Id: Id781e0edbdb7f275349161bb1da4bed4cb3c4947\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/809371fffd8b5d2280e4f4ba6bf309ab72fae780"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/809371fffd8b5d2280e4f4ba6bf309ab72fae780"}]},"branch":"refs/heads/master"},"ae1eb3987e2a58c09d565457964b73f9f3bd00b6":{"kind":"REWORK","_number":3,"created":"2026-01-20 18:21:33.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/58/973958/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/58/973958/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/58/973958/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/58/973958/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/58/973958/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/58/973958/3"}}},"commit":{"parents":[{"commit":"826a5f7666cab045b0cf84fc99a49a7590cf468b","subject":"Merge \"Same subnet with LB use Yes in UI\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/826a5f7666cab045b0cf84fc99a49a7590cf468b"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-20 16:36:03.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-20 18:21:30.000000000","tz":0},"subject":"Fix spacing after colon in Server Groups label","message":"Fix spacing after colon in Server Groups label\n\nAdded an explicit space after the colon in the\n“Members of Each Group” label to ensure correct\nformatting in the UI.\n\nChange-Id: Id781e0edbdb7f275349161bb1da4bed4cb3c4947\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ae1eb3987e2a58c09d565457964b73f9f3bd00b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ae1eb3987e2a58c09d565457964b73f9f3bd00b6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
