)]}'
{"id":"openstack%2Fskyline-console~974014","triplet_id":"openstack%2Fskyline-console~master~Icec1cd1bf8219860d6ae8afa039113dfb1efa8f9","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-01-26 21:07:06.000000000","reason":"removed on reply"},"28706":{"account":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"last_update":"2026-01-28 05:14:39.000000000","reason":"Change was submitted"},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-01-28 03:35:25.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-01-28 05:14:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icec1cd1bf8219860d6ae8afa039113dfb1efa8f9","subject":"Remove ironic-inspector","status":"MERGED","created":"2026-01-20 22:04:05.000000000","updated":"2026-01-28 05:14:39.000000000","submitted":"2026-01-28 05:14:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"974014","meta_rev_id":"818db6f5c837315222e614926a1a8ac2fbfee5d7","_number":974014,"virtual_id_number":974014,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-28 05:14:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"recommended":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"all":[{"value":1,"date":"2026-01-26 21:07:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":2,"date":"2026-01-28 03:35:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":1,"date":"2026-01-28 03:35:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-20 23:34:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-26 15:58:10.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2026-01-26 15:58:10.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"REVIEWER"},{"updated":"2026-01-26 15:58:10.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"}],"messages":[{"id":"61b867e75b22aa2c5f71772c9964faad242a3f64","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-20 22:04:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"21cb71cd0e03bf09cd81ca61b90b9e027f01c0d5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-20 22:06:36.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"c67dc2bdad6e5a7218e6066c2f7f6f07f0d911f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 23:34:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/398dd4b044174b3693a994828ce8e340\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb3d9854e60d4166abd55395fa88a563 : SUCCESS in 7m 23s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/dc8faca13fca4a89a0e8f6b78263383b : SUCCESS in 11m 14s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/7ebff94abd2f4cb48c236f440dc34098 : SUCCESS in 3m 40s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/4831724b3abe456fadfb3bb18343dafa : SUCCESS in 3m 48s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/edb48c4065a94d3387348acc1037ff34 : SUCCESS in 1h 24m 57s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/4305e42b164f49d0837aebb885ad0cf7 : SUCCESS in 59m 30s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/923674d674fa4c3ead2125875368898e : SUCCESS in 49m 17s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/d03910ac3f7c4289b4816fbdfb50c116 : SUCCESS in 51m 02s","accounts_in_message":[],"_revision_number":2},{"id":"babaa4085ceb29e0430f5c6a383eb038ee60936e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-26 14:49:30.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"769cf1365d3d013ca6daf3ffe106122677acf6d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 15:52:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ec1b016588e41d5adff5038c8daf7ce\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/593a01cdcb2a46f5972382c7bd8009cf : SUCCESS in 6m 03s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/5cd95c23ebe14607b28a316885152474 : SUCCESS in 10m 42s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/0e0b2be9ef204a898cf86d44c968b112 : SUCCESS in 3m 25s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/8a0d17a50eab40b9859a124c9ba49591 : SUCCESS in 6m 49s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/8d61267c4a244483bfcec482a62c93cb : SUCCESS in 50m 59s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/9a23f59061204a0e895dc373d14ba8e3 : SUCCESS in 1h 00m 32s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/da65589182ac417fa9cf7994d1a7c287 : SUCCESS in 34m 52s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/c61b8775028647468411806b03900846 : SUCCESS in 44m 08s","accounts_in_message":[],"_revision_number":3},{"id":"61da319d168b1d780cd5b18967bf7d637c03e899","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-26 15:58:10.000000000","message":"Patch Set 3:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"52da2a49fb06daf3726e61c8d415de972364c3a2","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-26 21:07:06.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"9277b4f1c0f2835891527bc98fee2a2559e90078","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-01-28 03:35:25.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"48652f63034073d1c234558ecdf9d54e5806d270","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 03:36:06.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"6defa04ea15c005a8cb2240a72a543d090d4d1c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 05:14:39.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5100659afb624fa58113db446e214cbf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03b78ef9dc384d48924dfc9e63f6ff80 : SUCCESS in 6m 53s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/9fa0c52a6ee04cd8974163703ee83814 : SUCCESS in 11m 34s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/03977fe9735143dca97ecc858b332d8d : SUCCESS in 7m 39s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/847867d5e69447eeb75f3ccd67239321 : SUCCESS in 5m 37s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/a4728f0c7fed46b9ac6f8dfe989730e7 : SUCCESS in 1h 21m 34s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/2f1030c6513a4138bc43dc48062dcf00 : SUCCESS in 44m 49s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/816336a0f3ae44c2bed1dd29561851f9 : SUCCESS in 37m 29s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/54a9fea63dcb482db639f7223c5ed66c : SUCCESS in 51m 17s","accounts_in_message":[],"_revision_number":3},{"id":"818db6f5c837315222e614926a1a8ac2fbfee5d7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 05:14:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"79776a94f8142a82b51af159bf45abd5039c81a1","revisions":{"b8d0fc4ed1a56dd46c61a2b9cae400adc0f6f7f0":{"kind":"REWORK","_number":1,"created":"2026-01-20 22:04:05.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/14/974014/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/14/974014/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/14/974014/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/14/974014/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/14/974014/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/14/974014/1"}}},"commit":{"parents":[{"commit":"826a5f7666cab045b0cf84fc99a49a7590cf468b","subject":"Merge \"Same subnet with LB use Yes in UI\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/826a5f7666cab045b0cf84fc99a49a7590cf468b"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-20 22:04:05.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-20 22:04:05.000000000","tz":0},"subject":"Remove ironic-inspector","message":"Remove ironic-inspector\n\nRetired in 2026.1 - I0caabd1ddeeada471531af7093d24a5add10ca5b\n\nChange-Id: Icec1cd1bf8219860d6ae8afa039113dfb1efa8f9\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b8d0fc4ed1a56dd46c61a2b9cae400adc0f6f7f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b8d0fc4ed1a56dd46c61a2b9cae400adc0f6f7f0"}]},"branch":"refs/heads/master"},"58737e7b5a7d6278048f32edaa6065e5856833ec":{"kind":"REWORK","_number":2,"created":"2026-01-20 22:06:36.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/14/974014/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/14/974014/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/14/974014/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/14/974014/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/14/974014/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/14/974014/2"}}},"commit":{"parents":[{"commit":"826a5f7666cab045b0cf84fc99a49a7590cf468b","subject":"Merge \"Same subnet with LB use Yes in UI\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/826a5f7666cab045b0cf84fc99a49a7590cf468b"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-20 22:04:05.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-20 22:06:33.000000000","tz":0},"subject":"Remove ironic-inspector","message":"Remove ironic-inspector\n\nRetired in 2026.1 - I0caabd1ddeeada471531af7093d24a5add10ca5b\n\nChange-Id: Icec1cd1bf8219860d6ae8afa039113dfb1efa8f9\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/58737e7b5a7d6278048f32edaa6065e5856833ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/58737e7b5a7d6278048f32edaa6065e5856833ec"}]},"branch":"refs/heads/master"},"79776a94f8142a82b51af159bf45abd5039c81a1":{"kind":"REWORK","_number":3,"created":"2026-01-26 14:49:30.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/14/974014/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/14/974014/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/14/974014/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/14/974014/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/14/974014/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/14/974014/3"}}},"commit":{"parents":[{"commit":"826a5f7666cab045b0cf84fc99a49a7590cf468b","subject":"Merge \"Same subnet with LB use Yes in UI\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/826a5f7666cab045b0cf84fc99a49a7590cf468b"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-20 22:04:05.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-26 14:49:29.000000000","tz":0},"subject":"Remove ironic-inspector","message":"Remove ironic-inspector\n\nRetired in 2026.1 - I0caabd1ddeeada471531af7093d24a5add10ca5b\n\nChange-Id: Icec1cd1bf8219860d6ae8afa039113dfb1efa8f9\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/79776a94f8142a82b51af159bf45abd5039c81a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/79776a94f8142a82b51af159bf45abd5039c81a1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
