)]}'
{"id":"openstack%2Fskyline-console~974017","triplet_id":"openstack%2Fskyline-console~master~I6078cad1b25c132a57228e42b5d32da42c59a583","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-01-22 03:53:30.000000000","reason":"Change was submitted"},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-01-22 02:18:32.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}},"hashtags":[],"change_id":"I6078cad1b25c132a57228e42b5d32da42c59a583","subject":"Adjust network detail card width","status":"MERGED","created":"2026-01-20 23:06:21.000000000","updated":"2026-01-22 03:54:20.000000000","submitted":"2026-01-22 03:53:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"974017","meta_rev_id":"4f24fdb8133f19acf86da37ac9697627265d0858","_number":974017,"virtual_id_number":974017,"owner":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-22 03:53:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-01-22 02:18:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-01-21 20:34:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-01-22 02:18:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-21 00:20:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-21 20:34:30.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-01-21 20:34:30.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"}],"messages":[{"id":"4c756e968505401b895ed3fc43f2b417d373f59b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-20 23:06:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1b3359bf55713ff9b8e662527262e9d19d54f2ad","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-20 23:09:59.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"665eb72ad7a497732b9a6c934244ea85767508e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 00:20:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a331bdfcc984849b1b638e0c3a642db\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69e13b660a5647db98c26c9efedc1ed8 : SUCCESS in 6m 29s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/4b8dad21db214c74b0f6d723e5a50612 : SUCCESS in 11m 56s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/e89922247f3d413da2046b29d1aa3118 : SUCCESS in 4m 34s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/f73facedaae1415bb26c580d7ada5c42 : SUCCESS in 7m 28s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/b56db247d50c4970846cf1010f10cade : SUCCESS in 1h 09m 47s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/02de7391e1704a50a50ce4d90f048eba : SUCCESS in 56m 08s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/deb3db681e414db592d842b6ddc6e656 : SUCCESS in 49m 40s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/f3c72fae49b343b9b4f1b39d39c20efa : SUCCESS in 53m 05s","accounts_in_message":[],"_revision_number":2},{"id":"643b5e6785780d7a2f132825441775a1e64d2373","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-21 11:07:12.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"f3065520a1041ae106874908c2f1e9ecfbcec27f","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-21 11:07:15.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":3},{"id":"93f0062ddcff7749c5293b7d884091cfd06a6393","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 12:21:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2ea94bb78974b0b8f3e16e8b7810075\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b6e4c9672a64cbba2125cda1649ee8d : SUCCESS in 6m 54s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/5069613ac57f415389f6521586e38db9 : SUCCESS in 11m 03s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/b9138f32ff434471ac26fd14336e6a8a : SUCCESS in 7m 54s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/94dd0a6304dd4e02930b5ad4343fa905 : SUCCESS in 7m 34s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/98622cb62f814fa3aabb9cd0251890cf : SUCCESS in 1h 12m 55s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/0927933f02514911b1aa05cdd2e563dd : SUCCESS in 59m 44s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/6db32080f9db49449f065d3580f6d675 : SUCCESS in 37m 12s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/ef2c546412864feb9508e5f66c1efae7 : SUCCESS in 48m 59s","accounts_in_message":[],"_revision_number":3},{"id":"84ae057f54ba53927ec88a484d5a555e1d7a2538","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-21 20:34:30.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6de0dc273efdea9e70bd4d3f49bcdc6b8ab4a707","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-01-22 02:18:32.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ce4cda385272fe4a7c90073433eb0f4f2f2c156d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 02:19:02.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"863d14c3e896c7be2b8ec26fa26c8522b1b6bae4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 03:53:30.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b43c402d59ab487997fa4c9b83f15de2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee4a19f4c23f406d877d23e6d90d4c79 : SUCCESS in 6m 06s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/f5c762837c734431b79f4e3e9bc66a3d : SUCCESS in 6m 11s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/c9a80be953914ae397c772fbfac2b3be : SUCCESS in 8m 04s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/8f065e6ac78d43009a1fe29bcfd86dde : SUCCESS in 8m 29s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/ccfcc1024b1b471e9c843371ce54070f : SUCCESS in 1h 20m 29s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/42351a4946d94383af9f780739c13c16 : SUCCESS in 33m 12s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/71d9352f6de44ba883a7313022a6085f : SUCCESS in 30m 43s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/ec20f97fde4b4f57a58ec9e58f56f696 : SUCCESS in 48m 10s","accounts_in_message":[],"_revision_number":3},{"id":"6a4c2e404b426dc7558113a520209b4e3c2b6daf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 03:53:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"4f24fdb8133f19acf86da37ac9697627265d0858","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 03:54:20.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1da8840269846309e6621b0ac80aead\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2abce8698bc4d6a97da30605132d947 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"063df13e46a5c05cce6ac3d01ba1000250c3b681","revisions":{"f9e8bb5235b85ffebc82df595f711ef9d6570961":{"kind":"REWORK","_number":1,"created":"2026-01-20 23:06:21.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/17/974017/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/17/974017/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/17/974017/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/17/974017/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/17/974017/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/17/974017/1"}}},"commit":{"parents":[{"commit":"826a5f7666cab045b0cf84fc99a49a7590cf468b","subject":"Merge \"Same subnet with LB use Yes in UI\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/826a5f7666cab045b0cf84fc99a49a7590cf468b"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-20 23:06:21.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-20 23:06:21.000000000","tz":0},"subject":"Adjust network detail card width","message":"Adjust network detail card width\n\nExpand left-side width to ensure Russian\nlabels fit on a single line.\n\nChange-Id: I6078cad1b25c132a57228e42b5d32da42c59a583\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f9e8bb5235b85ffebc82df595f711ef9d6570961"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f9e8bb5235b85ffebc82df595f711ef9d6570961"}]},"branch":"refs/heads/master"},"e4ccc4c5583efb2edd740381ca7b033e90ca9805":{"kind":"REWORK","_number":2,"created":"2026-01-20 23:09:59.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/17/974017/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/17/974017/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/17/974017/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/17/974017/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/17/974017/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/17/974017/2"}}},"commit":{"parents":[{"commit":"826a5f7666cab045b0cf84fc99a49a7590cf468b","subject":"Merge \"Same subnet with LB use Yes in UI\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/826a5f7666cab045b0cf84fc99a49a7590cf468b"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-20 23:06:21.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-20 23:09:53.000000000","tz":0},"subject":"Adjust network detail card width","message":"Adjust network detail card width\n\nExpand left-side width to ensure Russian\nlabels fit on a single line.\n\nChange-Id: I6078cad1b25c132a57228e42b5d32da42c59a583\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e4ccc4c5583efb2edd740381ca7b033e90ca9805"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e4ccc4c5583efb2edd740381ca7b033e90ca9805"}]},"branch":"refs/heads/master"},"063df13e46a5c05cce6ac3d01ba1000250c3b681":{"kind":"REWORK","_number":3,"created":"2026-01-21 11:07:12.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/17/974017/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/17/974017/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/17/974017/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/17/974017/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/17/974017/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/17/974017/3"}}},"commit":{"parents":[{"commit":"826a5f7666cab045b0cf84fc99a49a7590cf468b","subject":"Merge \"Same subnet with LB use Yes in UI\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/826a5f7666cab045b0cf84fc99a49a7590cf468b"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-20 23:06:21.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-21 11:07:05.000000000","tz":0},"subject":"Adjust network detail card width","message":"Adjust network detail card width\n\nExpand left-side width to ensure Russian\nlabels fit on a single line.\n\nChange-Id: I6078cad1b25c132a57228e42b5d32da42c59a583\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/063df13e46a5c05cce6ac3d01ba1000250c3b681"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/063df13e46a5c05cce6ac3d01ba1000250c3b681"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
