)]}'
{"id":"openstack%2Fskyline-console~974255","triplet_id":"openstack%2Fskyline-console~master~I76c9bd512e891d9e2acca42180e84ceb3367ffe7","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-01-27 07:20:13.000000000","reason":"\u003cGERRIT_ACCOUNT_37632\u003e replied on the change","reason_account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-01-27 02:38:34.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-01-26 09:37:09.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"I76c9bd512e891d9e2acca42180e84ceb3367ffe7","subject":"Hide extra columns in image selection","status":"MERGED","created":"2026-01-21 21:22:48.000000000","updated":"2026-01-27 09:05:14.000000000","submitted":"2026-01-27 09:04:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"974255","meta_rev_id":"93aeb3d68d4d39036352419d6a56e349d9d5cb78","_number":974255,"virtual_id_number":974255,"owner":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-27 09:04:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-01-27 02:38:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-01-26 09:37:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-01-27 02:38:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-21 22:50:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-22 11:06:41.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-01-22 11:06:41.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"}],"messages":[{"id":"76d5b69b231a37217baf7b751fa226dca4bb53f7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-21 21:22:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dd64014f67bc238fefffaa5ac8d65c49c61ae963","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-21 21:26:02.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"c57546a21f68edf67e0ae687b5047f386fa499a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 22:50:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e5c25bc4f984066b35b1bd6bb0d394d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b082ed3d39ef4f1b92c958395bc162a0 : SUCCESS in 6m 53s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/83bef153158749ac982786a9d63c9f12 : SUCCESS in 11m 00s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/7d539d430de74cc9ab19c1f4518542f0 : SUCCESS in 7m 45s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/c598d52c038344ed80fecfc3d92552e3 : SUCCESS in 7m 20s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/574a84c53da04b63a88b5a28278eae79 : SUCCESS in 1h 22m 31s","accounts_in_message":[],"_revision_number":2},{"id":"9658403c903754f875c0a9736444d2b1744c7b71","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-21 23:51:42.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"f90e02368bf709256250d6c590c2dc6a6d287511","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-21 23:51:44.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":3},{"id":"6f7768b1855b5290f6381901095e777e885ea4b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 00:47:58.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/61849b1ff7cd44189463bc345ed4146a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0d57550f92d4192a5d6ba7f7a816745 : SUCCESS in 6m 46s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/994ee4506bc146c898421b58b58a978a : SUCCESS in 4m 10s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/ea0aba11a89846b8a9878fc855b73e52 : SUCCESS in 6m 10s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/1b65bd72d2dc4085b8fc5b287377948c : SUCCESS in 8m 20s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/3a7e56d1d727445898a30f082375ff9a : FAILURE in 55m 30s","accounts_in_message":[],"_revision_number":3},{"id":"545b2c7e8c480a7bbb89fddbc24edce37be95599","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-22 08:57:10.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"bb98c51bdb70526c98875e6b8dec90f844d59fe9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 10:53:50.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1a15e48196742928b4b830978e0323b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f89b21f5d51c4c24a4fc77fe9485cd40 : SUCCESS in 6m 17s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/ab36fbbf4031419aa9efec830a5924ec : SUCCESS in 13m 40s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/846b47d7f1034a78aba5600a2c7a4783 : SUCCESS in 7m 42s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/59d1f6e6ae714ca4820528764abae07f : SUCCESS in 3m 23s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/56bb8794a9c94044bb97cb2accd1066f : SUCCESS in 1h 35m 18s","accounts_in_message":[],"_revision_number":3},{"id":"bb264619ec464e092a3f86cb65fe65d76fbdde5a","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-22 11:06:41.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"038e61e380269a0ca9c05ed96a4881b240ecd7be","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-01-23 02:15:35.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a7877321622f4aa79ba204276d7ea6c491d0b900","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-23 07:22:19.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1e0ff88949259b452af5c5b1c491ef38d9b12d66","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-23 07:22:22.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"ff653e0dc52b6c862285fe9ec828bacb29e6f193","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-23 07:29:26.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"3c893508ef397aaf4c68c517e599a81e011a357d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 08:46:36.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18274de4689b4bd89e52b716d3d83cf5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/674df58e669a4aa7aa028863e64ddd68 : SUCCESS in 4m 06s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/8855b773769c4a989e9f1c4100dca88f : SUCCESS in 8m 33s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/f6d98a42997a43969314172253761d94 : SUCCESS in 3m 40s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/3d0609422c7a417ba35e7a8f1dd22173 : SUCCESS in 5m 59s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/7d136d72814047e09b7cd93d3ff4cd1e : SUCCESS in 1h 22m 26s","accounts_in_message":[],"_revision_number":4},{"id":"60dd04928cfc82dd31787934e43981d34ab89d78","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-01-26 02:38:49.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ad23d8d19c021f7b8129d727edf9af2ac49e09e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-26 09:28:06.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"eaa0069b3750b1d604f54a6a23c71fbd8f2d9327","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-26 09:28:14.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"2bbf34b34c882f3257b4e263fb41c77a455dd316","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-26 09:37:09.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"b536cbd236fb7f9b18e6473ab9ecb1be8fbb68ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 10:21:23.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed75f5be3db7434dbd6815f481b93fce\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d21f3e91a364dd89339e09a7ebff956 : SUCCESS in 6m 12s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/fc35f2b9a70347959c18dc54dcb57d4f : SUCCESS in 11m 03s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/357ea02782754fe89f9e822ec50a6a58 : SUCCESS in 7m 37s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/ab408d148e5d44bfb99d732ecb58c22a : SUCCESS in 3m 55s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/75e3a058435b4757b7ffad3669fdfd64 : SUCCESS in 52m 25s","accounts_in_message":[],"_revision_number":5},{"id":"b238b1e43134f2c688492ef539f6f4fd77e2b00b","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-01-27 02:38:34.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"7bab765fd4d12acfca6f16d81bb7b4a11733f11b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 02:39:06.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"8c51326ad2a2e8751ad3885f786a231d18d2d9e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 03:51:54.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2f46718db9184fc293d4a37dbf46f77e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d00e74d19e55403e814956a8ea17f70a : SUCCESS in 5m 58s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/71ae855878e44ef197302486a766b5d9 : SUCCESS in 9m 22s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/e972d94b29b446df95f1ecfca02fa61c : SUCCESS in 4m 36s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/510411a069b54976bff81f3464508563 : SUCCESS in 5m 36s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/4f0e59cafab9475e8e2f8b10e6b19186 : FAILURE in 1h 08m 55s","accounts_in_message":[],"_revision_number":5},{"id":"3bc27bbdfb3d1592dd85695abb2a68a791800889","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-27 07:20:13.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"dba1b3d01551c1eb97a4584fbd9dbda5d4e02897","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 08:10:00.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d14c7558277642a0ae6d2ea031d316d7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a85f44f75b894e29a1435fe02d331aaf : SUCCESS in 7m 30s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/2adf0998b97d4bef8a496a956ca1e396 : SUCCESS in 6m 21s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/1c2acbe9f24f47c28315849559d5f00c : SUCCESS in 5m 02s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/e2fea76f1c5848fc8c425ba4e04b6e2b : SUCCESS in 7m 17s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/7c64550ba62a4178ba3007738042fc42 : SUCCESS in 47m 03s","accounts_in_message":[],"_revision_number":5},{"id":"39e8e23fe114ef8783cc3cf414e4b9d973eb9369","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 08:10:23.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"048369dcfea3959214c81806e5e36a12b7e7ebdd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 09:04:17.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e60199c377040b4a19490b3d62115be\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47cae924d49a4ea2b50dd34b5dd5d9ef : SUCCESS in 3m 22s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/9606e4af1c78418895cd898430393a51 : SUCCESS in 6m 59s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/82ec67ab8ea44bed88d2f8bca7f51d5c : SUCCESS in 7m 32s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/1854ec3a06ff4a3d8cf1ba005cee164a : SUCCESS in 4m 16s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/c0927306bd564b7ab9ba03aa3dc9c01f : SUCCESS in 53m 48s","accounts_in_message":[],"_revision_number":5},{"id":"bd6a4626119cbd2b96e1c0297a862adebcc05186","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 09:04:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"93aeb3d68d4d39036352419d6a56e349d9d5cb78","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 09:05:14.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9970c473a16e41198e5e4ecd8cc6f42f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93741bfbe6f44fd3bfcd3c23cfcc2b13 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"797527cb66dbc92998837a8555378559d46a91ae","revisions":{"330746fcbc51877afce25be1b4c4a61357742d17":{"kind":"REWORK","_number":1,"created":"2026-01-21 21:22:48.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/55/974255/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/55/974255/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/974255/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/974255/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/974255/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/55/974255/1"}}},"commit":{"parents":[{"commit":"d216a6adafe54e84eac73fc498f86f2e142f146f","subject":"Merge \"Remove CoreOS image type and add AlmaLinux and Rocky Linux\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d216a6adafe54e84eac73fc498f86f2e142f146f"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-21 21:22:48.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-21 21:22:48.000000000","tz":0},"subject":"Hide Image Columns","message":"Hide Image Columns\n\nChange-Id: I76c9bd512e891d9e2acca42180e84ceb3367ffe7\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/330746fcbc51877afce25be1b4c4a61357742d17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/330746fcbc51877afce25be1b4c4a61357742d17"}]},"branch":"refs/heads/master"},"9877ea888657529cf36a465c22abf65b8c184785":{"kind":"REWORK","_number":2,"created":"2026-01-21 21:26:02.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/55/974255/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/55/974255/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/974255/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/974255/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/974255/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/55/974255/2"}}},"commit":{"parents":[{"commit":"d216a6adafe54e84eac73fc498f86f2e142f146f","subject":"Merge \"Remove CoreOS image type and add AlmaLinux and Rocky Linux\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d216a6adafe54e84eac73fc498f86f2e142f146f"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-21 21:22:48.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-21 21:25:50.000000000","tz":0},"subject":"Hide Image Columns","message":"Hide Image Columns\n\nChange-Id: I76c9bd512e891d9e2acca42180e84ceb3367ffe7\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/9877ea888657529cf36a465c22abf65b8c184785"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/9877ea888657529cf36a465c22abf65b8c184785"}]},"branch":"refs/heads/master"},"33ae764be4bbf3e1624747bd2a53f21e3a2ac807":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-01-21 23:51:42.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/55/974255/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/55/974255/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/974255/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/974255/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/974255/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/55/974255/3"}}},"commit":{"parents":[{"commit":"d216a6adafe54e84eac73fc498f86f2e142f146f","subject":"Merge \"Remove CoreOS image type and add AlmaLinux and Rocky Linux\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d216a6adafe54e84eac73fc498f86f2e142f146f"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-21 21:22:48.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-21 23:51:42.000000000","tz":0},"subject":"Hide extra columns in image selection","message":"Hide extra columns in image selection\n\nHide Min System Disk, Min Memory, and Format\ncolumns to make the image selection view\ncleaner and easier.\n\nChange-Id: I76c9bd512e891d9e2acca42180e84ceb3367ffe7\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/33ae764be4bbf3e1624747bd2a53f21e3a2ac807"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/33ae764be4bbf3e1624747bd2a53f21e3a2ac807"}]},"branch":"refs/heads/master","description":"Edit commit message"},"a2315429db659b0e2fec6d7e73e95b8154965567":{"kind":"REWORK","_number":4,"created":"2026-01-23 07:22:22.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/55/974255/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/55/974255/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/974255/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/974255/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/974255/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/55/974255/4"}}},"commit":{"parents":[{"commit":"d216a6adafe54e84eac73fc498f86f2e142f146f","subject":"Merge \"Remove CoreOS image type and add AlmaLinux and Rocky Linux\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d216a6adafe54e84eac73fc498f86f2e142f146f"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-21 21:22:48.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-23 07:21:44.000000000","tz":0},"subject":"Hide extra columns in image selection","message":"Hide extra columns in image selection\n\nHide Min System Disk, Min Memory, and Format\ncolumns to make the image selection view\ncleaner and easier.\n\nChange-Id: I76c9bd512e891d9e2acca42180e84ceb3367ffe7\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a2315429db659b0e2fec6d7e73e95b8154965567"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a2315429db659b0e2fec6d7e73e95b8154965567"}]},"branch":"refs/heads/master"},"797527cb66dbc92998837a8555378559d46a91ae":{"kind":"REWORK","_number":5,"created":"2026-01-26 09:28:06.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/55/974255/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/55/974255/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/974255/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/974255/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/974255/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/55/974255/5"}}},"commit":{"parents":[{"commit":"d216a6adafe54e84eac73fc498f86f2e142f146f","subject":"Merge \"Remove CoreOS image type and add AlmaLinux and Rocky Linux\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d216a6adafe54e84eac73fc498f86f2e142f146f"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-21 21:22:48.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-26 09:27:49.000000000","tz":0},"subject":"Hide extra columns in image selection","message":"Hide extra columns in image selection\n\nHide Min System Disk, Min Memory, and Format\ncolumns to make the image selection view\ncleaner and easier.\n\nChange-Id: I76c9bd512e891d9e2acca42180e84ceb3367ffe7\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/797527cb66dbc92998837a8555378559d46a91ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/797527cb66dbc92998837a8555378559d46a91ae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
