)]}'
{"id":"openstack%2Fskyline-console~974315","triplet_id":"openstack%2Fskyline-console~master~I2c0769273106be0d942f0fba177ee5ce035760bf","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-01-28 05:14:35.000000000","reason":"Change was submitted"},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-01-28 02:46:19.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-01-27 21:10:30.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"I2c0769273106be0d942f0fba177ee5ce035760bf","subject":"Filter OS tabs by available images when creating instance","status":"MERGED","created":"2026-01-22 12:51:02.000000000","updated":"2026-01-28 05:16:33.000000000","submitted":"2026-01-28 05:14:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"974315","meta_rev_id":"a6200cfead37d764dc65fe40c0a2ffbc404020a9","_number":974315,"virtual_id_number":974315,"owner":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-28 05:14:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-01-28 02:46:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-01-27 21:10:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-01-28 02:46:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-22 15:05:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-23 13:53:10.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-01-23 13:53:10.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"}],"messages":[{"id":"a1785a89338e75aec3c76130b1d99f3f9aa8dd23","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-22 12:51:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fe4e174bc7537e8b8e4d7ec7311d4145bd86056d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-22 13:00:47.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"2bc66e98e3a85c9dab2d0462a2c6ba526e92f15c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-22 13:01:02.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"ac9bdbea159d27475c94e3095b81705575de4930","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 15:05:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce5759353c924f598751adc63ca20312\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/029f26762f1d42e1b5031063a63d357e : SUCCESS in 6m 56s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/94a91bb38a0f4e55837b5e9cc3510fae : SUCCESS in 5m 28s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/b496fd2ef2a640ac83a70ad49772497b : SUCCESS in 4m 07s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/73c8f9375f04461ea1a8ab3107c968c8 : SUCCESS in 7m 51s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/b3aebef5a359474c992449ab3734bec4 : SUCCESS in 1h 26m 05s","accounts_in_message":[],"_revision_number":3},{"id":"99b8aee5d5c29da43bc6465a027dee8e475bc882","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-22 19:18:30.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"aa6a9701bcfc29cd7a0a27c84c5fac1543c81cb7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 20:53:21.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/707e026c8c5344ff8dd4591e556aa1bf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d7cb58b489a4e388d4b36f902fd2bb2 : SUCCESS in 5m 44s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/6c11b71ed5e042f48e25186822b9352a : SUCCESS in 4m 09s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/557bf18874564d17b7adf45c21cbced3 : SUCCESS in 3m 47s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/2784cdf1590342b9a6d6968c5059c3a7 : SUCCESS in 8m 50s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/d18b7fc43a4b4674a74973ba0875f870 : SUCCESS in 1h 26m 33s","accounts_in_message":[],"_revision_number":4},{"id":"50d3ae29117ee6be987141338441ca097a8bc22e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-22 21:31:20.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"0bd82b9f4d66af3b221e9073ec63b0a0aaa92da1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 22:35:12.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c71ea1e09e364530978868e78d07abc1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06204b76671e4096b68542fa5f8bc080 : SUCCESS in 7m 30s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/1061c176247746538cbf770b867e2832 : SUCCESS in 9m 52s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/1bf924b3c19f4b2ea076ac3bc642c53a : SUCCESS in 6m 47s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/138b908b2db54d50857f3dc4b895485b : SUCCESS in 8m 55s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/c70643551aa14da9bf7f0f11629427f5 : SUCCESS in 1h 03m 02s","accounts_in_message":[],"_revision_number":5},{"id":"2928515fe27b982d4611fdf6d63f54ec5c535806","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-22 23:23:50.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"75b1279b6f5762113bc0cf4b404e415d5a1d3d79","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-22 23:25:24.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"dafe75c2c30b8c2113b2ca50aefecf06d85b6bf5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 00:49:35.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f24e7f82feb3406ea102e4bfc3287923\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfc5fdeb4c944df9970811480a4d2b08 : SUCCESS in 7m 23s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/e89462b99ddd4f2aa4614d7eda961d9e : SUCCESS in 11m 03s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/3a16f04179984a0e89e9c01a735f4c97 : SUCCESS in 8m 02s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/57a8195bc9ba470db9ba6243f0679d07 : SUCCESS in 7m 33s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/47d79c9ef7db497c8d5010d58b070c18 : SUCCESS in 1h 20m 16s","accounts_in_message":[],"_revision_number":7},{"id":"414d514f96172456af1e346c6ec67301f6954431","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-23 09:22:16.000000000","message":"Patch Set 8: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"b93fac1db535e3fa4fb5ca06267c83f0a7a627c9","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-23 09:22:36.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":8},{"id":"204366ce3ab4ae138dbb89e5eb9b4ec4385a47fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 10:48:46.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a0c02fcab4a4374ab6a79ebc59137b9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfc5e89c768b4ec7b5321fddf96f03a3 : SUCCESS in 6m 29s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/a16412d9e0c84ae7b76b76776fb4b5db : SUCCESS in 9m 09s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/db40ea907a0546f1928a80ea87c3a4f4 : SUCCESS in 7m 47s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/ad0a2e8c4c394026934b8026577fffd7 : SUCCESS in 3m 13s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/6ba7e467cf9b4863bbbcb5fc01eadcc3 : SUCCESS in 1h 21m 01s","accounts_in_message":[],"_revision_number":8},{"id":"7b8be30ca8f02fc5f42253770454bfc7ee3d4a5f","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-23 13:53:10.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"4eb2189570e2cf09229f15dad2db7fe611f79ac1","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-01-26 02:34:56.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"ecdc1ad780a44db0c55d6659280435ddf4c5816c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-26 10:04:12.000000000","message":"Patch Set 9: Published edit on patch set 8.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"47452e1682fc63a1c5183c414b1f14bada8d98e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 11:17:58.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/909ce89ec6ad44d587bc36897c8af612\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0bda8217ef84933b0600913e732bb13 : SUCCESS in 7m 19s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/c236908fee0a42e8b235475d91f67bc6 : SUCCESS in 10m 38s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/8eb10b0ee86a4ce3b6ba18fdf661541d : SUCCESS in 7m 21s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/9d11e56e01584ce4a1784cd0a1a1b5dc : SUCCESS in 4m 44s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/9ebb65d960224714b9e15dfd433e9c24 : SUCCESS in 1h 11m 06s","accounts_in_message":[],"_revision_number":9},{"id":"4032fc6ed5cbf4432bdf6c2e9883c610b7366529","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-26 12:28:17.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"d8ef67b1c08c8c5e65e555b78aa95a21a2ae3b55","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-01-27 02:56:06.000000000","message":"Patch Set 9: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"88e59d766be9c99fe308703ec120a842d7d1e4b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-27 08:25:51.000000000","message":"Patch Set 10: Published edit on patch set 9.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"061207712bdc1143d73626db60eb6820b47d9c15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 09:20:44.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b76fb9328fbc4156b83431b376fbb4e1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8a4d83e364b44938385fde62f863829 : SUCCESS in 4m 19s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/7a6c6eb0ce5c4aeb99ddd00e4a9f6e97 : SUCCESS in 5m 30s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/d87387625cd44bf395b2e43030c7fa40 : SUCCESS in 7m 02s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/7faf266b5e344f17b71a35008ea12239 : SUCCESS in 8m 01s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/519a64ab75e54f3483c7929604ec6e82 : FAILURE in 53m 39s","accounts_in_message":[],"_revision_number":10},{"id":"c28dd645cb0a49a72aa9d70a6a31a25fdf4e391a","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-27 12:31:36.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"773c26683f6e28d7160a4b88e494e0fa56cf7cf1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 13:48:37.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7075680d9174a56b33b0920f0580a61\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fcd48b0387754ddb92fa8dfd33b9ea62 : SUCCESS in 3m 19s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/0739aa401feb4220bbf07829e4682c54 : SUCCESS in 7m 10s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/7a740045c7db4c6c9a6297443b331957 : SUCCESS in 7m 00s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/892482f39cd64a2bac1e9fcad2a7e5d2 : SUCCESS in 7m 28s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/768c8625b1e9444dac2947ba16264cef : SUCCESS in 1h 15m 06s","accounts_in_message":[],"_revision_number":10},{"id":"9a5608fdd2f30e255ec0c4d6fa27d54ac9c6009d","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-27 19:28:25.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"449047195d725a05255c59d0f55ae2a5eb16d763","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-27 21:10:30.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"c0fb8e8cd5f9a34ed28424e8bdef30987012232a","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-01-28 02:46:19.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"de623fcf3c6140e2485dc7314d42836108c38fec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 02:46:53.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"ece4b9f2abd43fcefaa90a9b19584d033897fd23","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 05:14:35.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e78c4e494144f95996f0f77e0c9f766\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ee4c5bf14ae48ada2ac958c7a5d37df : SUCCESS in 6m 11s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/169afd8d50b4481b96bc9efbcc6de86f : SUCCESS in 5m 19s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/1fd6cbdfe05e4673b06f7f4b9f790835 : SUCCESS in 9m 01s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/6de2c38ba65d4cfea091d16392cd350c : SUCCESS in 7m 37s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/d39099cb2bce4c698a65a0e65d57bc1d : SUCCESS in 1h 23m 00s","accounts_in_message":[],"_revision_number":10},{"id":"e7c8dae73ba99d724ac249d61a3f851e59a3cfa3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 05:14:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"a6200cfead37d764dc65fe40c0a2ffbc404020a9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 05:16:33.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9fda9ce92cba448c8317f611653fceab\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fedbe916719d4f4e804d3997dcf44564 : SUCCESS in 1m 44s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"61279e74d2d2b3886b575663ee912658f5b9bc8b","revisions":{"6605f7d15df2b9dfe4bc747212e6def66b3fa6eb":{"kind":"REWORK","_number":1,"created":"2026-01-22 12:51:02.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/15/974315/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/15/974315/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/974315/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/974315/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/974315/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/15/974315/1"}}},"commit":{"parents":[{"commit":"f829bf22d5c5b3b98717fb0e4b50c29c8e41ee60","subject":"Merge \"Update: Instance detail view full log opening on url and re-fetches log data on page reload.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f829bf22d5c5b3b98717fb0e4b50c29c8e41ee60"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-22 12:51:02.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-22 12:51:02.000000000","tz":0},"subject":"Instance Base Detail","message":"Instance Base Detail\n\nChange-Id: I2c0769273106be0d942f0fba177ee5ce035760bf\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6605f7d15df2b9dfe4bc747212e6def66b3fa6eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6605f7d15df2b9dfe4bc747212e6def66b3fa6eb"}]},"branch":"refs/heads/master"},"e8e785c77b55b710475c74b88a39c75278d749ce":{"kind":"REWORK","_number":2,"created":"2026-01-22 13:00:47.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/15/974315/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/15/974315/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/974315/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/974315/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/974315/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/15/974315/2"}}},"commit":{"parents":[{"commit":"f829bf22d5c5b3b98717fb0e4b50c29c8e41ee60","subject":"Merge \"Update: Instance detail view full log opening on url and re-fetches log data on page reload.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f829bf22d5c5b3b98717fb0e4b50c29c8e41ee60"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-22 12:51:02.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-22 13:00:47.000000000","tz":0},"subject":"Instance Base Detail","message":"Instance Base Detail\n\nChange-Id: I2c0769273106be0d942f0fba177ee5ce035760bf\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e8e785c77b55b710475c74b88a39c75278d749ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e8e785c77b55b710475c74b88a39c75278d749ce"}]},"branch":"refs/heads/master"},"494feca4d0841eac7abfe4bea31e41ccf1e7a640":{"kind":"REWORK","_number":3,"created":"2026-01-22 13:01:02.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/15/974315/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/15/974315/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/974315/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/974315/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/974315/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/15/974315/3"}}},"commit":{"parents":[{"commit":"f829bf22d5c5b3b98717fb0e4b50c29c8e41ee60","subject":"Merge \"Update: Instance detail view full log opening on url and re-fetches log data on page reload.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f829bf22d5c5b3b98717fb0e4b50c29c8e41ee60"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-22 12:51:02.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-22 13:01:02.000000000","tz":0},"subject":"Instance Base Detail","message":"Instance Base Detail\n\nChange-Id: I2c0769273106be0d942f0fba177ee5ce035760bf\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/494feca4d0841eac7abfe4bea31e41ccf1e7a640"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/494feca4d0841eac7abfe4bea31e41ccf1e7a640"}]},"branch":"refs/heads/master"},"12abe9bb68951a51d1ac018202a3b5f505f6e006":{"kind":"REWORK","_number":4,"created":"2026-01-22 19:18:30.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/15/974315/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/15/974315/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/974315/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/974315/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/974315/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/15/974315/4"}}},"commit":{"parents":[{"commit":"f829bf22d5c5b3b98717fb0e4b50c29c8e41ee60","subject":"Merge \"Update: Instance detail view full log opening on url and re-fetches log data on page reload.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f829bf22d5c5b3b98717fb0e4b50c29c8e41ee60"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-22 12:51:02.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-22 19:18:30.000000000","tz":0},"subject":"Instance Base Detail","message":"Instance Base Detail\n\nChange-Id: I2c0769273106be0d942f0fba177ee5ce035760bf\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/12abe9bb68951a51d1ac018202a3b5f505f6e006"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/12abe9bb68951a51d1ac018202a3b5f505f6e006"}]},"branch":"refs/heads/master"},"8e81d335519c9cab3b781b85d96f3cfa59607fbd":{"kind":"REWORK","_number":5,"created":"2026-01-22 21:31:20.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/15/974315/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/15/974315/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/974315/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/974315/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/974315/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/15/974315/5"}}},"commit":{"parents":[{"commit":"f829bf22d5c5b3b98717fb0e4b50c29c8e41ee60","subject":"Merge \"Update: Instance detail view full log opening on url and re-fetches log data on page reload.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f829bf22d5c5b3b98717fb0e4b50c29c8e41ee60"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-22 12:51:02.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-22 21:31:20.000000000","tz":0},"subject":"Instance Base Detail","message":"Instance Base Detail\n\nChange-Id: I2c0769273106be0d942f0fba177ee5ce035760bf\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/8e81d335519c9cab3b781b85d96f3cfa59607fbd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/8e81d335519c9cab3b781b85d96f3cfa59607fbd"}]},"branch":"refs/heads/master"},"ee9264f59854de304f908c3be658020893abc071":{"kind":"REWORK","_number":6,"created":"2026-01-22 23:23:50.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/15/974315/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/15/974315/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/974315/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/974315/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/974315/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/15/974315/6"}}},"commit":{"parents":[{"commit":"f829bf22d5c5b3b98717fb0e4b50c29c8e41ee60","subject":"Merge \"Update: Instance detail view full log opening on url and re-fetches log data on page reload.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f829bf22d5c5b3b98717fb0e4b50c29c8e41ee60"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-22 12:51:02.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-22 23:23:41.000000000","tz":0},"subject":"Instance Base Detail","message":"Instance Base Detail\n\nChange-Id: I2c0769273106be0d942f0fba177ee5ce035760bf\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ee9264f59854de304f908c3be658020893abc071"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ee9264f59854de304f908c3be658020893abc071"}]},"branch":"refs/heads/master"},"23990bfac36efe655d16bd5108ef901112df9188":{"kind":"REWORK","_number":7,"created":"2026-01-22 23:25:24.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/15/974315/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/15/974315/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/974315/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/974315/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/974315/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/15/974315/7"}}},"commit":{"parents":[{"commit":"f829bf22d5c5b3b98717fb0e4b50c29c8e41ee60","subject":"Merge \"Update: Instance detail view full log opening on url and re-fetches log data on page reload.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f829bf22d5c5b3b98717fb0e4b50c29c8e41ee60"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-22 12:51:02.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-22 23:25:01.000000000","tz":0},"subject":"Instance Base Detail","message":"Instance Base Detail\n\nChange-Id: I2c0769273106be0d942f0fba177ee5ce035760bf\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/23990bfac36efe655d16bd5108ef901112df9188"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/23990bfac36efe655d16bd5108ef901112df9188"}]},"branch":"refs/heads/master"},"cc3fa592c25c9477e1799d861a2eebd8e7ceb499":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2026-01-23 09:22:16.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/15/974315/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/15/974315/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/974315/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/974315/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/974315/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/15/974315/8"}}},"commit":{"parents":[{"commit":"f829bf22d5c5b3b98717fb0e4b50c29c8e41ee60","subject":"Merge \"Update: Instance detail view full log opening on url and re-fetches log data on page reload.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f829bf22d5c5b3b98717fb0e4b50c29c8e41ee60"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-22 12:51:02.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-23 09:22:16.000000000","tz":0},"subject":"Filter OS tabs by available images when creating instance","message":"Filter OS tabs by available images when creating instance\n\nInstance creation OS selector now shows only operating\nsystem tabs that have at least one active image available,\nhiding empty OS distributions.\n\nChange-Id: I2c0769273106be0d942f0fba177ee5ce035760bf\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/cc3fa592c25c9477e1799d861a2eebd8e7ceb499"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/cc3fa592c25c9477e1799d861a2eebd8e7ceb499"}]},"branch":"refs/heads/master","description":"Edit commit message"},"0591537f6a350ee53607a37400bf3ec0e2ba05a0":{"kind":"REWORK","_number":9,"created":"2026-01-26 10:04:12.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/15/974315/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/15/974315/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/974315/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/974315/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/974315/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/15/974315/9"}}},"commit":{"parents":[{"commit":"f829bf22d5c5b3b98717fb0e4b50c29c8e41ee60","subject":"Merge \"Update: Instance detail view full log opening on url and re-fetches log data on page reload.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f829bf22d5c5b3b98717fb0e4b50c29c8e41ee60"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-22 12:51:02.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-26 10:02:57.000000000","tz":0},"subject":"Filter OS tabs by available images when creating instance","message":"Filter OS tabs by available images when creating instance\n\nInstance creation OS selector now shows only operating\nsystem tabs that have at least one active image available,\nhiding empty OS distributions.\n\nChange-Id: I2c0769273106be0d942f0fba177ee5ce035760bf\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/0591537f6a350ee53607a37400bf3ec0e2ba05a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/0591537f6a350ee53607a37400bf3ec0e2ba05a0"}]},"branch":"refs/heads/master"},"61279e74d2d2b3886b575663ee912658f5b9bc8b":{"kind":"REWORK","_number":10,"created":"2026-01-27 08:25:51.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/15/974315/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/15/974315/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/974315/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/974315/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/15/974315/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/15/974315/10"}}},"commit":{"parents":[{"commit":"f829bf22d5c5b3b98717fb0e4b50c29c8e41ee60","subject":"Merge \"Update: Instance detail view full log opening on url and re-fetches log data on page reload.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f829bf22d5c5b3b98717fb0e4b50c29c8e41ee60"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-22 12:51:02.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-27 08:25:39.000000000","tz":0},"subject":"Filter OS tabs by available images when creating instance","message":"Filter OS tabs by available images when creating instance\n\nInstance creation OS selector now shows only operating\nsystem tabs that have at least one active image available,\nhiding empty OS distributions.\n\nChange-Id: I2c0769273106be0d942f0fba177ee5ce035760bf\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/61279e74d2d2b3886b575663ee912658f5b9bc8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/61279e74d2d2b3886b575663ee912658f5b9bc8b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
