)]}'
{"id":"openstack%2Fskyline-console~974316","triplet_id":"openstack%2Fskyline-console~master~Ieb48a8819b89105d08c4a91a8fe4e35bfdb75610","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-01-23 03:41:15.000000000","reason":"Change was submitted"},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-01-23 02:18:28.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}},"hashtags":[],"change_id":"Ieb48a8819b89105d08c4a91a8fe4e35bfdb75610","subject":"Fix OS display names in image selector","status":"MERGED","created":"2026-01-22 13:15:19.000000000","updated":"2026-01-23 03:43:06.000000000","submitted":"2026-01-23 03:41:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"974316","meta_rev_id":"ea2122b051d54ee7ee1d33725fbd6fcfdf8b28fd","_number":974316,"virtual_id_number":974316,"owner":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-23 03:41:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-01-23 02:18:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-01-22 21:32:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-01-23 02:18:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-22 15:01:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-22 18:01:05.000000000","updated_by":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2026-01-22 21:32:22.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"7500c285d263272dc78a614ed25a482898bce0bc","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-22 13:15:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b2ea41bfb2ee67e8b011d45987268736daf178a0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-22 13:19:20.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"ea63b4b92127d50cdec759f2a76291f0a410bf35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 15:01:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6380c25b2d445568b3ccaae9be9e775\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c67463ab29d4705a5a615a0d283ed2c : SUCCESS in 3m 44s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/c1175c0e6aaf44c89485a307ef66e946 : SUCCESS in 11m 51s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/167fa649d55740b8a25579573ee9c0a3 : SUCCESS in 3m 31s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/ab542e3200754813b709ef1d17d26e9e : SUCCESS in 8m 36s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/11678c0a906144509f3e99b9446a599e : SUCCESS in 1h 32m 08s","accounts_in_message":[],"_revision_number":2},{"id":"d3a5e9e793764188b411482b937cc1e78e4d9536","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-22 15:37:17.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"fee673a072332de65bb35d253a2b7e2991c24e9e","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-22 15:37:27.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":3},{"id":"24ab6c02da75e5b1b593c57dff7af2c524938d0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 16:34:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff0ba105b2f84d6bafa590475db58cf7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a64f22c8b61e407f83d78c0ece866f4f : SUCCESS in 4m 36s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/37a5b4e2b3704137bc5edc9482b657d6 : SUCCESS in 5m 37s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/642b2ac1143041e4a923cdd4c316b9a3 : SUCCESS in 8m 34s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/92b9a42a03cc413eafcce76219119b0d : SUCCESS in 7m 33s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/3028ebe5721d4679bea5d62300ff2362 : SUCCESS in 55m 37s","accounts_in_message":[],"_revision_number":3},{"id":"f2c7440534e8cd6287e24bfa262e99a21e94178b","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-22 21:32:22.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"486da7175633b5dc5219f50c961325beeecdd164","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-01-23 02:18:28.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"307d4a57288f17e1367eec4b509ea49d5374affe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 02:19:00.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9ddf5f018b2144039065d58f52558628cd8b19e0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 03:41:15.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76b9021b42774dbf9f3f3cc151a05622\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96b51e57100248a791b5dda869138f4f : SUCCESS in 4m 13s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/dcc7ba7e2d384a3e924a224d69cde950 : SUCCESS in 7m 21s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/fadfd163f1954c48b8341157868bf22c : SUCCESS in 5m 20s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/cfbfd4592a09418883620da92029f80c : SUCCESS in 4m 37s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/cf16a1f89f2d4a46ad423ad9ee83e14e : SUCCESS in 1h 09m 27s","accounts_in_message":[],"_revision_number":3},{"id":"896739333a68c562b0bdad43b0cf1799c8cfcb41","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 03:41:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"ea2122b051d54ee7ee1d33725fbd6fcfdf8b28fd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 03:43:06.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a644ec53da1e4b1eb0c237b223b0714c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd58d26257fe4cb387120f5ce67db107 : SUCCESS in 1m 35s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8f8f7516b1bb60d31d4cedb64ebebdee2c88d3d2","revisions":{"c22389601bd36da34a2a94325f7429ea2a031b97":{"kind":"REWORK","_number":1,"created":"2026-01-22 13:15:19.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/16/974316/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/16/974316/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/16/974316/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/16/974316/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/16/974316/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/16/974316/1"}}},"commit":{"parents":[{"commit":"f829bf22d5c5b3b98717fb0e4b50c29c8e41ee60","subject":"Merge \"Update: Instance detail view full log opening on url and re-fetches log data on page reload.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f829bf22d5c5b3b98717fb0e4b50c29c8e41ee60"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-22 13:15:19.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-22 13:15:19.000000000","tz":0},"subject":"OS naming","message":"OS naming\n\nChange-Id: Ieb48a8819b89105d08c4a91a8fe4e35bfdb75610\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/c22389601bd36da34a2a94325f7429ea2a031b97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/c22389601bd36da34a2a94325f7429ea2a031b97"}]},"branch":"refs/heads/master"},"afb38c3685f4f61d3903154f2e0e1c70e94e76c0":{"kind":"REWORK","_number":2,"created":"2026-01-22 13:19:20.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/16/974316/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/16/974316/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/16/974316/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/16/974316/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/16/974316/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/16/974316/2"}}},"commit":{"parents":[{"commit":"f829bf22d5c5b3b98717fb0e4b50c29c8e41ee60","subject":"Merge \"Update: Instance detail view full log opening on url and re-fetches log data on page reload.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f829bf22d5c5b3b98717fb0e4b50c29c8e41ee60"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-22 13:15:19.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-22 13:18:56.000000000","tz":0},"subject":"OS naming","message":"OS naming\n\nChange-Id: Ieb48a8819b89105d08c4a91a8fe4e35bfdb75610\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/afb38c3685f4f61d3903154f2e0e1c70e94e76c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/afb38c3685f4f61d3903154f2e0e1c70e94e76c0"}]},"branch":"refs/heads/master"},"8f8f7516b1bb60d31d4cedb64ebebdee2c88d3d2":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-01-22 15:37:17.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/16/974316/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/16/974316/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/16/974316/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/16/974316/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/16/974316/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/16/974316/3"}}},"commit":{"parents":[{"commit":"f829bf22d5c5b3b98717fb0e4b50c29c8e41ee60","subject":"Merge \"Update: Instance detail view full log opening on url and re-fetches log data on page reload.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f829bf22d5c5b3b98717fb0e4b50c29c8e41ee60"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-22 13:15:19.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-22 15:37:17.000000000","tz":0},"subject":"Fix OS display names in image selector","message":"Fix OS display names in image selector\n\nCorrect OS names in the image selector to display proper\nbranding and full names (e.g. AlmaLinux, FreeBSD\nRocky Linux).\n\nChange-Id: Ieb48a8819b89105d08c4a91a8fe4e35bfdb75610\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/8f8f7516b1bb60d31d4cedb64ebebdee2c88d3d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/8f8f7516b1bb60d31d4cedb64ebebdee2c88d3d2"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
