)]}'
{"id":"openstack%2Fskyline-console~974323","triplet_id":"openstack%2Fskyline-console~master~I2039ca70edd1d1f1cde6593449fae5162037c0c3","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-01-23 04:13:07.000000000","reason":"Change was submitted"},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-01-23 02:23:04.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}},"hashtags":[],"change_id":"I2039ca70edd1d1f1cde6593449fae5162037c0c3","subject":"Fix notification word wrapping","status":"MERGED","created":"2026-01-22 14:12:41.000000000","updated":"2026-01-23 04:14:03.000000000","submitted":"2026-01-23 04:13:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"974323","meta_rev_id":"32166604757e4f4716c5e1654961ee0a8b1a4322","_number":974323,"virtual_id_number":974323,"owner":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-23 04:13:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-01-23 02:23:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-01-22 21:45:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-01-23 02:23:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-22 15:49:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-22 21:36:40.000000000","updated_by":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2026-01-22 21:45:57.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"e17d3e11c8d97336087fe191853a1877f580c823","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-22 14:12:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e8352ec41ad73fd2d0ffa00daa6ef8c9725a9652","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-22 14:14:15.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"6d0d555ede1df3a5d88a2f080848220156cbe91c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 15:49:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/608c84c3189f4d608f8f7b48285da1ff\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a17758bd50e6484bba7a6cf967cb9713 : SUCCESS in 7m 26s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/1d605d003c4044b7b97d8f49ac57efe1 : SUCCESS in 11m 37s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/d6ac046880db4297a020a0bd066b9aa4 : SUCCESS in 6m 14s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/0e38aa343ff1486e8c2dd4e88f75a1dc : SUCCESS in 4m 56s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/f3f73f23be8b4bf9af27e8b1c47ce775 : FAILURE in 54m 59s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/2575ec6e4e5042aca47c4f4778380319 : SUCCESS in 1h 02m 46s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/82cd1c02a4a041eea71d0f68ee126367 : SUCCESS in 53m 40s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/7ca43c0da2114539a08256f38d54b3bb : SUCCESS in 37m 38s","accounts_in_message":[],"_revision_number":2},{"id":"7819258217a02379de8ba843e7bc772c0001b34c","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-22 17:23:40.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f0b9c13bc5a1a89c173a22505d0fb14211b81106","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 18:42:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99e7c43f1a0d40e1beea983729abf6ce\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2080566fdd67421aaa8f5ae43339a36e : SUCCESS in 6m 13s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/b8f66855d76d461da4c15ecd9331bbe2 : SUCCESS in 4m 23s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/4a9e8a767ae74b3d8b3b37ed93b0cf0a : SUCCESS in 7m 59s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/dfc02f93edf64187985f50404ef7d16f : SUCCESS in 3m 14s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/83addae6fa87410b9eaa492fc96f01c8 : SUCCESS in 47m 56s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/b0f7b41c77ff4589a32e54020e1d2235 : SUCCESS in 1h 13m 23s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/fbedf38fb7804e56a05aba28089b850d : SUCCESS in 51m 55s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/53fe0a4c32e044339e6a1ecb42d7ceea : SUCCESS in 39m 24s","accounts_in_message":[],"_revision_number":2},{"id":"471cd90f59200920c59b0a6773fc10842942c3a3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-22 19:03:28.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"dd6707244eaaf9072e6eefd61b5ec0c546e4c684","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 20:32:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4fa28b567df04b1ea9032c2a4fe5f539\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4796b7282c34db7945b48b3ed023db8 : SUCCESS in 4m 01s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/820cdf2aaea84513992b4babca4a329a : SUCCESS in 11m 46s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/53ef4c21abe647a3acdcdf5ae748ef9c : SUCCESS in 5m 16s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/5e78c3d4497242329b7047757cb3edbc : SUCCESS in 3m 26s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/ff0335e14ab24d8aae242c48af0af41c : SUCCESS in 1h 22m 55s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/fad97ea7ecac4fa8bed1c30ac7b918a5 : SUCCESS in 1h 01m 43s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/b4de5e9317854ce9a8b6e168706815bd : SUCCESS in 50m 57s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/36df06e54f9f4bce95a6316716fe2ccd : SUCCESS in 53m 54s","accounts_in_message":[],"_revision_number":3},{"id":"3411d7ebff57b0bf96a42f434e731d80027595dc","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-22 21:36:28.000000000","message":"Patch Set 4: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"6ef1e3818e8e4f555c991ef14aeb4243807a819b","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-22 21:36:32.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":4},{"id":"f3aba4dc81b214b04cce6f8f4d0e36ab60d25729","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-22 21:45:57.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"02267c69c72cffc6c485cd7ea9924618cd3906bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 23:15:41.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7904185c09624d74ac62cd44d7ab22c3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c5f8f9a99e74367b7ff1519c5a7c4da : SUCCESS in 6m 28s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/31b183fb04cc4415a6984b3ba6714f2a : SUCCESS in 4m 26s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/fc177c69b98c46a1a912e07d9319de1e : SUCCESS in 3m 22s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/30e788ce904a4595bd321ebc04819bd7 : SUCCESS in 5m 54s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/e9c1b23f145748c5a0e581a1cc9d9e83 : SUCCESS in 1h 31m 02s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/1f4e5ed8e1a54294bef22a3e1f4a802a : SUCCESS in 59m 51s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/693036777ad2401db53de132f745b745 : SUCCESS in 55m 33s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/bf35f1ff834e47ca9aa1cbe286d6c56d : SUCCESS in 51m 29s","accounts_in_message":[],"_revision_number":4},{"id":"1355cb4009ced2cb4ad77853a378aa2762daf5d8","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-01-23 02:23:04.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"8b3f212e6f2003272854df6fa2f67a197e46d0f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 02:24:00.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"f875b1056ba53e26a60eb7f38fc06827c31f759d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 04:13:06.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90e7230b2fcf4dd7a0acde9f381934c6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/393fb11b53524552b4b237bbba9bb817 : SUCCESS in 4m 53s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/2fbca1dd4926461aa5cc05ab6e459085 : SUCCESS in 6m 20s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/e02db3a0c66248a9b61a298c74fd03e4 : SUCCESS in 6m 40s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/d7130c8b53564c1eab66ec3bdad869db : SUCCESS in 7m 41s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/60cd557c5f174ffd9fa9714ff013895b : SUCCESS in 53m 37s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/c8621c67b2d748b88de17a112df2e6fb : SUCCESS in 34m 14s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/fc57bf47738c436f9975f04e5fd35165 : SUCCESS in 32m 41s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/4df632e074b54aa3811e4c2a5f16574c : SUCCESS in 48m 39s","accounts_in_message":[],"_revision_number":4},{"id":"883f38383144c6e8a3f3196fc9192ea86d6bc2d2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 04:13:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"32166604757e4f4716c5e1654961ee0a8b1a4322","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 04:14:03.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f85551eee2a746c2906cae6de69d4dd2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf0c7475252b4fce9c4e7c9a99b25e49 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"0933a5d431b54ee38280f321efce21075b88a629","revisions":{"06fdbb36417c21a978ec643bc74adee60d81b95f":{"kind":"REWORK","_number":1,"created":"2026-01-22 14:12:41.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/23/974323/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/23/974323/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/23/974323/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/23/974323/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/23/974323/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/23/974323/1"}}},"commit":{"parents":[{"commit":"f829bf22d5c5b3b98717fb0e4b50c29c8e41ee60","subject":"Merge \"Update: Instance detail view full log opening on url and re-fetches log data on page reload.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f829bf22d5c5b3b98717fb0e4b50c29c8e41ee60"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-22 14:12:41.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-22 14:12:41.000000000","tz":0},"subject":"WIP","message":"WIP\n\nChange-Id: I2039ca70edd1d1f1cde6593449fae5162037c0c3\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/06fdbb36417c21a978ec643bc74adee60d81b95f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/06fdbb36417c21a978ec643bc74adee60d81b95f"}]},"branch":"refs/heads/master"},"1878125a061ffae3ceaa45954ddcff378ba6db19":{"kind":"REWORK","_number":2,"created":"2026-01-22 14:14:15.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/23/974323/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/23/974323/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/23/974323/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/23/974323/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/23/974323/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/23/974323/2"}}},"commit":{"parents":[{"commit":"f829bf22d5c5b3b98717fb0e4b50c29c8e41ee60","subject":"Merge \"Update: Instance detail view full log opening on url and re-fetches log data on page reload.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f829bf22d5c5b3b98717fb0e4b50c29c8e41ee60"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-22 14:12:41.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-22 14:14:08.000000000","tz":0},"subject":"WIP","message":"WIP\n\nChange-Id: I2039ca70edd1d1f1cde6593449fae5162037c0c3\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/1878125a061ffae3ceaa45954ddcff378ba6db19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/1878125a061ffae3ceaa45954ddcff378ba6db19"}]},"branch":"refs/heads/master"},"941b84d78998646ae3e45bc502278d5d6caffd02":{"kind":"REWORK","_number":3,"created":"2026-01-22 19:03:28.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/23/974323/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/23/974323/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/23/974323/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/23/974323/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/23/974323/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/23/974323/3"}}},"commit":{"parents":[{"commit":"f829bf22d5c5b3b98717fb0e4b50c29c8e41ee60","subject":"Merge \"Update: Instance detail view full log opening on url and re-fetches log data on page reload.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f829bf22d5c5b3b98717fb0e4b50c29c8e41ee60"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-22 14:12:41.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-22 19:03:28.000000000","tz":0},"subject":"WIP","message":"WIP\n\nChange-Id: I2039ca70edd1d1f1cde6593449fae5162037c0c3\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/941b84d78998646ae3e45bc502278d5d6caffd02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/941b84d78998646ae3e45bc502278d5d6caffd02"}]},"branch":"refs/heads/master"},"0933a5d431b54ee38280f321efce21075b88a629":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2026-01-22 21:36:28.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/23/974323/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/23/974323/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/23/974323/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/23/974323/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/23/974323/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/23/974323/4"}}},"commit":{"parents":[{"commit":"f829bf22d5c5b3b98717fb0e4b50c29c8e41ee60","subject":"Merge \"Update: Instance detail view full log opening on url and re-fetches log data on page reload.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f829bf22d5c5b3b98717fb0e4b50c29c8e41ee60"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-22 14:12:41.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-22 21:36:28.000000000","tz":0},"subject":"Fix notification word wrapping","message":"Fix notification word wrapping\n\nFix notification text wrapping where messages\nwere incorrectly split by individual characters\ninstead of whole words, improving readability of\ndifferent messages.\n\nChange-Id: I2039ca70edd1d1f1cde6593449fae5162037c0c3\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/0933a5d431b54ee38280f321efce21075b88a629"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/0933a5d431b54ee38280f321efce21075b88a629"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
