)]}'
{"id":"openstack%2Fskyline-console~974376","triplet_id":"openstack%2Fskyline-console~master~Ief74b807f0281ff13c4ed9de79fb48ff80149ebb","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-01-28 07:21:56.000000000","reason":"\u003cGERRIT_ACCOUNT_37632\u003e replied on the change","reason_account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-01-28 02:42:03.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-01-28 10:10:05.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"Ief74b807f0281ff13c4ed9de79fb48ff80149ebb","subject":"fix for Message Reason","status":"MERGED","created":"2026-01-22 21:39:52.000000000","updated":"2026-01-28 13:07:58.000000000","submitted":"2026-01-28 13:07:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"974376","meta_rev_id":"187e24771b414e4de47d11492008e468484e60ac","_number":974376,"virtual_id_number":974376,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-28 13:07:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":2,"date":"2026-01-28 02:42:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":1,"date":"2026-01-28 02:42:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-22 21:43:09.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"},{"updated":"2026-01-22 21:43:09.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2026-01-22 22:22:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4ec6ba0f50c57a3d149d87db7fe0c93a25049f47","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-22 21:39:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"89796a10b08714585bd205f3ebb2a077e89ec1e1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-22 21:42:25.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"d9bd0b1d1490cc2c5579954dd49899739f421868","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-22 21:42:29.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"d3b094123cc863340bdf880b9688aa381a66ab42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-22 22:22:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6729d3ba3bc14daf94f50757d0000a64\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b892122782a34b3bb96532e82f8f6aa0 : SUCCESS in 6m 58s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/4a154e858c594c1983f0b2f6fae3dede : SUCCESS in 6m 29s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/ff316779ae4a4f198ef757f73114a92f : SUCCESS in 6m 40s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/59383ec00f9a490eb21c2809912efeff : SUCCESS in 8m 17s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/017e288adeb44cc7bee247f256161fe1 : SUCCESS in 32m 06s","accounts_in_message":[],"_revision_number":2},{"id":"944c89f90a2b9bdace36385256f38c19e5e53fab","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-26 07:22:32.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"149ee47c4568ff00787174ba286205bbfb015d42","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-01-27 02:38:02.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4e84dddd132abe3c881cfd18f8f7041015911a26","tag":"autogenerated:gerrit:abandon","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-27 09:26:20.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2},{"id":"d94c0b0a84f1f83e7dd70cb2435bac6fe16c5670","tag":"autogenerated:gerrit:restore","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-27 10:56:55.000000000","message":"Restored","accounts_in_message":[],"_revision_number":2},{"id":"ce1731f9f79fec013af2b137060ed679abffacd7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-27 10:57:15.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"3dcc3d219b7afdb44d28110c93ece338ae8a62ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-27 10:58:25.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"900edf1e531346830e36d1fe70be4707fa40299a","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-27 10:58:33.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1b6b58ddf738edb6ccbfe6034a172650d9c1bf0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 12:46:37.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd4288e0bdf244cd8c9ae3030ccf0053\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3b686832ca64711bd590124c9a28476 : SUCCESS in 9m 09s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/d0a216ca4df14bfb9b2c93b2e6637dae : SUCCESS in 11m 35s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/3ce7309509024ce88965d7b490cd9d61 : SUCCESS in 7m 51s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/5ef39bb468d64548a7f016003d7e3492 : SUCCESS in 8m 15s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/66250f320e8f40e090a8d9a35e2e6483 : SUCCESS in 1h 37m 51s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/688e8f83542f4ef8b7ce8b1cb496dc2c : SUCCESS in 58m 49s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/bcd2855b62d141da8e80c9adcc9fa687 : SUCCESS in 52m 16s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/23d7db00c30341439d2cacdf474c147f : SUCCESS in 52m 42s","accounts_in_message":[],"_revision_number":4},{"id":"a6f954ba4b85ff8e0876daa96676b5794a81e5e1","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-01-28 02:42:03.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"e472fe3aec10a04d251d9a80bbf38e66c136b120","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 02:42:34.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"a3d1f6117f65817b068c9db7cde149f22ec3867e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 03:57:38.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c8fe18074d424f4db67fcab016f839cb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2fa498e72044f678cc6fb969e77580e : SUCCESS in 6m 35s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/4e76d713140c4535a6440a2a80df9aed : SUCCESS in 9m 27s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/413173f7a5b7486aba5d5e1d0ce86732 : SUCCESS in 4m 10s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/466a37c749604d5d8b5f7b221b350ab3 : SUCCESS in 7m 22s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/f5174e93d26743d3bc55e05e046d4052 : FAILURE in 53m 25s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/1a91ba2e5ce74fe28113b265ded6bb1f : SUCCESS in 1h 14m 21s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/eb6bee6461dc4d1698190dce3e7533fd : SUCCESS in 53m 46s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/41d960d69be54fd482d8ca7ed6121a2d : SUCCESS in 46m 17s","accounts_in_message":[],"_revision_number":4},{"id":"74ed79039aabe1fc0d48e2123426c4c42b9d1440","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-28 07:21:56.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"81733dc40cd38ba28d462a8b5f502695427f532d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 08:36:07.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d2bf9955f11416ab488a2ec5f288c22\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07f6996a5c784334a83b270b0ae20430 : SUCCESS in 6m 18s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/27e6d4a13d3d42328be2d6ae1a721886 : SUCCESS in 9m 03s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/e779a597e8034d1c88b31012e35ef940 : SUCCESS in 7m 37s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/efafb17b0a964c8f89ff06171fac3622 : SUCCESS in 7m 46s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/01eb2c3f9818408f83a2a4e53621e9d8 : SUCCESS in 1h 12m 09s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/700ff2a830dc4116b29845b056eba980 : SUCCESS in 51m 31s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/f055702a11184fa0aa8603dc90aff6a4 : SUCCESS in 53m 17s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/003f5499669f418e8ca40aef1d182ba8 : SUCCESS in 48m 41s","accounts_in_message":[],"_revision_number":4},{"id":"c84f40f3167b93e18777a46bb0ddee851d2611d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 08:36:36.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"dc75175e2d3191bae80530d476c6a13c927b19d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 10:08:49.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9603d740450741d686fc789b382a5de3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd8cf70cdd1c4f7db64034fa308ba248 : SUCCESS in 7m 01s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/3497c872514348789d9501f2cd09d8b3 : SUCCESS in 7m 22s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/c0adcc78ad494c95a70698a86352115f : SUCCESS in 5m 10s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/e58c9f1ae9294fa592e23a8b1e289f36 : SUCCESS in 7m 27s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/195da2118c4e42a096c3999fd70f0d52 : SUCCESS in 1h 20m 33s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/45b5c2a6a85047fe9c5d7a8af55df6ae : SUCCESS in 1h 31m 43s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/b1f908b05c3743baac400bd9f96e9576 : FAILURE in 44m 48s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/7ce9dbd2afcc48d9aa284db4e132f5ae : SUCCESS in 34m 59s","accounts_in_message":[],"_revision_number":4},{"id":"fbfec3a82cb76910f545bc55e3757a4a32ef2f78","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-28 10:10:05.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5ec7c60bd7fc7493d40bc93200061f53277f5115","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 11:39:49.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab7953b54dd44d20aa97b74de0543351\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4b4a5037496484a835b721ababa86d4 : SUCCESS in 6m 08s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/92c05a226f5e43b4beb0facd29e5fd7e : SUCCESS in 10m 52s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/5b396ed951c54eb7b6e54167c1d27609 : SUCCESS in 3m 54s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/b0675b34acde410c9eba5cbd7d87a599 : SUCCESS in 7m 49s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/b883b6568edd4e278e8c6f82ab423af7 : SUCCESS in 1h 24m 17s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/4578058f99f74031973314e5115e4957 : SUCCESS in 1h 00m 05s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/11030666035946e3967fcacceb114702 : SUCCESS in 53m 31s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/011359c557374c8faecd61c4a7e6acba : SUCCESS in 43m 43s","accounts_in_message":[],"_revision_number":4},{"id":"460b69d7984cec82af6e637ebc7743551c0c9482","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 11:40:19.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"94905fc5ce1de375683f81ea826123fd21067676","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 13:07:01.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7cd327f2724740f787a7f33b736246ed\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3c2e90764594d93a501fbc3a134d1be : SUCCESS in 6m 21s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/1c1f77fc2e084d859f28df605cdcda09 : SUCCESS in 10m 30s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/20557c73b2b94d9483bc635e89336421 : SUCCESS in 3m 51s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/a9fa465add094fe6a6ba14f9b6ac1474 : SUCCESS in 4m 44s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/9160ab43009d4e9b97d68c923846acbd : SUCCESS in 1h 21m 55s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/9a4ba6714d8c4170a0a8da23e3a3ee7d : SUCCESS in 1h 02m 27s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/103f49c4a7194c55b371dfc4b90b867a : SUCCESS in 51m 47s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/41046118f4874f3cbb4ceeee3742e797 : SUCCESS in 48m 53s","accounts_in_message":[],"_revision_number":4},{"id":"fcafda490a3d6e5ed245066223af2567da5f92af","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 13:07:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"187e24771b414e4de47d11492008e468484e60ac","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 13:07:58.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1ad78ffb124478c9f39ccaf74f8c82c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2971bfe199f746ed844f474e8a06cfd9 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"0615b5d6993228839bc4f87095b5fca7dbd33ea1","revisions":{"cf63b9c339303b66e376e191a1d65b9cf93413d2":{"kind":"REWORK","_number":1,"created":"2026-01-22 21:39:52.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/76/974376/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/76/974376/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/76/974376/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/76/974376/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/76/974376/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/76/974376/1"}}},"commit":{"parents":[{"commit":"f829bf22d5c5b3b98717fb0e4b50c29c8e41ee60","subject":"Merge \"Update: Instance detail view full log opening on url and re-fetches log data on page reload.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f829bf22d5c5b3b98717fb0e4b50c29c8e41ee60"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-22 21:39:52.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-22 21:39:52.000000000","tz":0},"subject":"Remote IP Prefix rename to IP Prefix","message":"Remote IP Prefix rename to IP Prefix\n\nlet\u0027s simplify the wording\n\nChange-Id: Ief74b807f0281ff13c4ed9de79fb48ff80149ebb\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/cf63b9c339303b66e376e191a1d65b9cf93413d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/cf63b9c339303b66e376e191a1d65b9cf93413d2"}]},"branch":"refs/heads/master"},"30c39033e348defc716902ca6ba832d49f1c1f98":{"kind":"REWORK","_number":2,"created":"2026-01-22 21:42:25.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/76/974376/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/76/974376/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/76/974376/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/76/974376/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/76/974376/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/76/974376/2"}}},"commit":{"parents":[{"commit":"f829bf22d5c5b3b98717fb0e4b50c29c8e41ee60","subject":"Merge \"Update: Instance detail view full log opening on url and re-fetches log data on page reload.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f829bf22d5c5b3b98717fb0e4b50c29c8e41ee60"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-22 21:39:52.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-22 21:42:25.000000000","tz":0},"subject":"Remote IP Prefix rename to IP Prefix","message":"Remote IP Prefix rename to IP Prefix\n\nlet\u0027s simplify the wording\n\nChange-Id: Ief74b807f0281ff13c4ed9de79fb48ff80149ebb\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/30c39033e348defc716902ca6ba832d49f1c1f98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/30c39033e348defc716902ca6ba832d49f1c1f98"}]},"branch":"refs/heads/master"},"6b8e9d7d2f70f4d7f257475daef0ce5d2e5db6de":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-01-27 10:57:15.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/76/974376/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/76/974376/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/76/974376/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/76/974376/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/76/974376/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/76/974376/3"}}},"commit":{"parents":[{"commit":"f829bf22d5c5b3b98717fb0e4b50c29c8e41ee60","subject":"Merge \"Update: Instance detail view full log opening on url and re-fetches log data on page reload.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f829bf22d5c5b3b98717fb0e4b50c29c8e41ee60"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-22 21:39:52.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-27 10:57:15.000000000","tz":0},"subject":"fix for Message Reason","message":"fix for Message Reason\n\nChange-Id: Ief74b807f0281ff13c4ed9de79fb48ff80149ebb\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6b8e9d7d2f70f4d7f257475daef0ce5d2e5db6de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6b8e9d7d2f70f4d7f257475daef0ce5d2e5db6de"}]},"branch":"refs/heads/master","description":"Edit commit message"},"0615b5d6993228839bc4f87095b5fca7dbd33ea1":{"kind":"REWORK","_number":4,"created":"2026-01-27 10:58:25.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/76/974376/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/76/974376/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/76/974376/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/76/974376/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/76/974376/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/76/974376/4"}}},"commit":{"parents":[{"commit":"f829bf22d5c5b3b98717fb0e4b50c29c8e41ee60","subject":"Merge \"Update: Instance detail view full log opening on url and re-fetches log data on page reload.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f829bf22d5c5b3b98717fb0e4b50c29c8e41ee60"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-22 21:39:52.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-27 10:58:25.000000000","tz":0},"subject":"fix for Message Reason","message":"fix for Message Reason\n\nChange-Id: Ief74b807f0281ff13c4ed9de79fb48ff80149ebb\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/0615b5d6993228839bc4f87095b5fca7dbd33ea1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/0615b5d6993228839bc4f87095b5fca7dbd33ea1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
