)]}'
{"id":"openstack%2Fskyline-console~974395","triplet_id":"openstack%2Fskyline-console~master~I211637cbd6a94672ce1d22d899f2746f4abf186d","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-01-24 14:39:14.000000000","reason":"removed on reply"},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-01-26 02:36:43.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-01-26 04:12:19.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I211637cbd6a94672ce1d22d899f2746f4abf186d","subject":"Backend rename to Back-End","status":"MERGED","created":"2026-01-23 07:36:59.000000000","updated":"2026-01-26 04:12:19.000000000","submitted":"2026-01-26 04:12:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"974395","meta_rev_id":"c411efcebc613babb39c03a0a83e4781e36199b7","_number":974395,"virtual_id_number":974395,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-26 04:12:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"recommended":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"all":[{"value":1,"date":"2026-01-24 14:39:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":2,"date":"2026-01-26 02:36:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":1,"date":"2026-01-26 02:36:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-23 07:38:55.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"},{"updated":"2026-01-23 07:38:55.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2026-01-23 09:09:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1b2497d0d221a16d8a02d3907f9d79e02ca30ec0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-23 07:36:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7e38b6c95597470e3a29c3f21dbf6c05fe34c6dc","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-23 07:38:47.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"0eddc913aedec719cf7c3f9375010eacc39152dc","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-23 07:38:55.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"238073d62ac57c62ac209b7da16ae0a54a366330","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-23 07:41:51.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"3f37a158145c775814be0d19e39eba89e837acab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 09:09:17.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6990438e11540739aeba98b52c9b558\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af4dbd30938a4dcdada606ad6c0be7f1 : SUCCESS in 6m 22s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/d407a826078e418f82f162186a4fdcd3 : SUCCESS in 12m 08s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/97d6c577fb274c25af8857cec66e3884 : SUCCESS in 7m 26s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/34cd86c63d124b8c8cb91de54cfad02e : SUCCESS in 7m 42s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/51ef3f020e0b4ab980da330e194c7349 : SUCCESS in 1h 22m 39s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/365189cbf57e4cf7a69d81fd78aaa1ba : SUCCESS in 52m 25s","accounts_in_message":[],"_revision_number":3},{"id":"f9f691e7879cb771caa06c6ecfd7a6e8c0e2976f","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-24 14:39:14.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"3127eb9000aa45386d3b26768ea39db067ba0cc3","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-01-26 02:36:43.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"76a01bd4dd50c4c2cb01ea5b14bf4df187325bb8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 02:37:22.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"25bc9a201c06903ba1c9f62051e52068ee2c1f4e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 04:12:19.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a19e2dc4a834ee8ac99f001b1d9340d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b927f6065ba4f9cad3d53abf22df63b : SUCCESS in 7m 06s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/bcb6c67cb692474ca1f1e17a3900653c : SUCCESS in 8m 59s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/5aeb531a434f4dd18abb95311cbada82 : SUCCESS in 4m 22s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/29ac71e52c1e486f983fc10b0846ab69 : SUCCESS in 7m 04s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/2828f406033c40b1a5896c51de35145c : SUCCESS in 1h 18m 30s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/d9ba432386084eabbe3195fd460d408c : SUCCESS in 50m 58s","accounts_in_message":[],"_revision_number":3},{"id":"c411efcebc613babb39c03a0a83e4781e36199b7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 04:12:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b265958ae1dd65babe4fb907475f6b467dd32741","revisions":{"173bc587daba75948111285e9cc82abb0880d097":{"kind":"REWORK","_number":1,"created":"2026-01-23 07:36:59.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/95/974395/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/95/974395/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/95/974395/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/95/974395/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/95/974395/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/95/974395/1"}}},"commit":{"parents":[{"commit":"7ca428991a488df316bf6ab05764c2859dd1d0d5","subject":"Merge \"Fix notification word wrapping\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7ca428991a488df316bf6ab05764c2859dd1d0d5"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-23 07:36:59.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-23 07:36:59.000000000","tz":0},"subject":"Backend rename to Back-End","message":"Backend rename to Back-End\n\ntrivial fix\n\nChange-Id: I211637cbd6a94672ce1d22d899f2746f4abf186d\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/173bc587daba75948111285e9cc82abb0880d097"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/173bc587daba75948111285e9cc82abb0880d097"}]},"branch":"refs/heads/master"},"f75bb79b464e57ea8cb5889859b8071e2243286f":{"kind":"REWORK","_number":2,"created":"2026-01-23 07:38:47.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/95/974395/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/95/974395/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/95/974395/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/95/974395/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/95/974395/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/95/974395/2"}}},"commit":{"parents":[{"commit":"7ca428991a488df316bf6ab05764c2859dd1d0d5","subject":"Merge \"Fix notification word wrapping\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7ca428991a488df316bf6ab05764c2859dd1d0d5"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-23 07:36:59.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-23 07:38:47.000000000","tz":0},"subject":"Backend rename to Back-End","message":"Backend rename to Back-End\n\ntrivial fix\n\nChange-Id: I211637cbd6a94672ce1d22d899f2746f4abf186d\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f75bb79b464e57ea8cb5889859b8071e2243286f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f75bb79b464e57ea8cb5889859b8071e2243286f"}]},"branch":"refs/heads/master"},"b265958ae1dd65babe4fb907475f6b467dd32741":{"kind":"REWORK","_number":3,"created":"2026-01-23 07:41:51.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/95/974395/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/95/974395/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/95/974395/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/95/974395/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/95/974395/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/95/974395/3"}}},"commit":{"parents":[{"commit":"7ca428991a488df316bf6ab05764c2859dd1d0d5","subject":"Merge \"Fix notification word wrapping\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7ca428991a488df316bf6ab05764c2859dd1d0d5"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-23 07:36:59.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-23 07:41:50.000000000","tz":0},"subject":"Backend rename to Back-End","message":"Backend rename to Back-End\n\ntrivial fix\n\nChange-Id: I211637cbd6a94672ce1d22d899f2746f4abf186d\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b265958ae1dd65babe4fb907475f6b467dd32741"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b265958ae1dd65babe4fb907475f6b467dd32741"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
