)]}'
{"id":"openstack%2Fskyline-console~974610","triplet_id":"openstack%2Fskyline-console~master~I46642f2f8269d30179f2e8345f2f5b6879f95f48","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-01-25 20:39:17.000000000","reason":"removed on reply"},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-01-26 02:23:12.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-01-26 04:12:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I46642f2f8269d30179f2e8345f2f5b6879f95f48","subject":"Logs - add phrase in language file","status":"MERGED","created":"2026-01-24 13:45:58.000000000","updated":"2026-01-26 04:12:59.000000000","submitted":"2026-01-26 04:12:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"974610","meta_rev_id":"816e15c60fa2ee0c468f121f50d33c1634a5a424","_number":974610,"virtual_id_number":974610,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-26 04:12:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"recommended":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"all":[{"value":1,"date":"2026-01-25 20:39:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":2,"date":"2026-01-26 02:23:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":1,"date":"2026-01-26 02:23:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-24 13:50:04.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2026-01-24 13:50:04.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"},{"updated":"2026-01-24 14:25:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"427fb7a80a0eab6b0bc6ee0d3967bb5815eb981a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-24 13:45:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4979b4360cd4b624ccb429b16234ce7e102fdb8e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-24 13:49:13.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"65515bfefeb8967e88c3a69732f1de9519586b35","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-24 13:49:52.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"193ab1143c044f393736551f4b61ab35b6cd31a7","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-24 13:50:04.000000000","message":"Patch Set 3:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"058b4346cb9b592f7030e01a4372bc2640ce61ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-24 13:50:55.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"106e67ae996ede751215c21b47552e71193b1e61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-24 14:25:39.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c03fa96861fd4f248059d1d958b04b57\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2d383f9a40f469e911eaba251d6db60 : SUCCESS in 6m 32s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/f33e1700c8d5427485e7eedd65b74ecc : FAILURE in 4m 19s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/36c9cc4a4cc8435a831e6b9a466a7690 : SUCCESS in 5m 56s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/f43620b6671b47e08dbf18d363ccf98d : SUCCESS in 7m 42s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/841630f74c2f4f32a31a81920dce896b : FAILURE in 33m 20s","accounts_in_message":[],"_revision_number":4},{"id":"c1b11853c9dc191a48ac70551479611fa33393a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-24 14:47:51.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"949ac4deff4428f5719906f59d73fd3be94ecce2","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-24 14:47:55.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":5},{"id":"d947fa939a4e92011ee7f4cc01da1b71a829d9d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-24 16:33:57.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ce373e7b95141dda41f4bbf9bea8b93\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ea6f6dec5ee4c79abc5d8c8d7ddc01c : SUCCESS in 6m 46s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/f214ab1fc4ce4efabb84baa9a19a0603 : SUCCESS in 9m 54s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/9a79e5d0839944828aae0b4a5cad26a1 : SUCCESS in 4m 14s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/49d6a9b4b5f349eb83d6354f3ed819a1 : SUCCESS in 7m 56s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/1085825ac7914449b7c3151b5dc15f98 : SUCCESS in 1h 28m 10s","accounts_in_message":[],"_revision_number":5},{"id":"e16655da14c5aedac5f94a5c47a5becccd3fe64b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-24 16:58:13.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"ea558822c93475fc9af5458eb00f2a603130c932","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-24 17:51:32.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c1e94c02d5554a239579a3cb873ce186\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29be93b184b144d6baddea28a080f20b : SUCCESS in 3m 21s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/19188c1382594279a9304fd2d0b1563f : SUCCESS in 5m 44s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/bb501957837b4b15aa0bb28374a3eb80 : SUCCESS in 7m 31s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/35d1bbbb68884166b3beee63529d2c70 : SUCCESS in 5m 56s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/51984b5e28be4cbdba5c835bae592ecd : FAILURE in 52m 18s","accounts_in_message":[],"_revision_number":6},{"id":"ad00970f0582b5f78b7bd239ff082c74eb4ee341","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-24 20:30:43.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"f02c4893fa1e20f0dfcc786f4b0152f761a94526","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-24 21:59:55.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4cc63eb4a54e4e8384d36456439ad37d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bef94908b5b2409c83816d205c0c50dc : SUCCESS in 3m 35s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/8c7b357659e8480d9a3cba931201e15d : SUCCESS in 8m 12s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/d1b800ee4ae3429a862213ff0d15ce2c : SUCCESS in 4m 52s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/3c0ab51475354aa186fdb6acb33b3042 : SUCCESS in 3m 58s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/3d24dc8cda2a43a18b8cbaec224c7021 : SUCCESS in 1h 23m 24s","accounts_in_message":[],"_revision_number":6},{"id":"1c3b77740dbab3e553b2e625d9e82c2c8a231c21","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-25 12:03:37.000000000","message":"Patch Set 6:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":6},{"id":"ae83344d53b3f3c09fa936f9e22c5a7f412f6fb4","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-25 20:39:17.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"439b0933ffbb7237bd03e8d1be9def553f838cc9","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-01-26 02:23:12.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"b85b8c9f2242ffcac79586de5f227bec479fe8ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 02:23:37.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"06520f40f5415af12638ec33725a8cbb35c5edbf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 04:12:04.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36dacd19eaf74015a4d714888314c698\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8d4552dff8d4c56894c1d82eaa0ebd4 : SUCCESS in 6m 56s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/d0f5b52848634942912c59487e4707a1 : SUCCESS in 8m 55s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/ba47b1cb6bc1418881eb7a0d6cbf9e4e : SUCCESS in 5m 13s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/74fdd328c7a245b5ba1706988df69993 : SUCCESS in 5m 11s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/d8b1e508382f40cc8a8f5ec92f4d181c : SUCCESS in 1h 24m 56s","accounts_in_message":[],"_revision_number":6},{"id":"494cdd5d82352cca78cbfe282151929d6e731094","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 04:12:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"816e15c60fa2ee0c468f121f50d33c1634a5a424","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 04:12:59.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/154ce85ec9ff4302b8409152af8db256\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae3d21a0d86145f6bf75bdb3efdbe8c3 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"50014286cf31476d0527c841543dd6e16142f761","revisions":{"42bd629f0aed57954cd98e6a231121932107b9f8":{"kind":"REWORK","_number":1,"created":"2026-01-24 13:45:58.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/10/974610/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/10/974610/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/10/974610/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/10/974610/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/10/974610/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/10/974610/1"}}},"commit":{"parents":[{"commit":"ead36ab260bbc41336096f379578d7f67eb4fcea","subject":"Merge \"Improve IP terminology for instances\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ead36ab260bbc41336096f379578d7f67eb4fcea"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-24 13:45:58.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-24 13:45:58.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: I46642f2f8269d30179f2e8345f2f5b6879f95f48\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/42bd629f0aed57954cd98e6a231121932107b9f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/42bd629f0aed57954cd98e6a231121932107b9f8"}]},"branch":"refs/heads/master"},"1ae2178dd83c1247edf9866bb17d77309a3664f4":{"kind":"REWORK","_number":2,"created":"2026-01-24 13:49:13.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/10/974610/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/10/974610/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/10/974610/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/10/974610/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/10/974610/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/10/974610/2"}}},"commit":{"parents":[{"commit":"ead36ab260bbc41336096f379578d7f67eb4fcea","subject":"Merge \"Improve IP terminology for instances\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ead36ab260bbc41336096f379578d7f67eb4fcea"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-24 13:45:58.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-24 13:49:13.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: I46642f2f8269d30179f2e8345f2f5b6879f95f48\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/1ae2178dd83c1247edf9866bb17d77309a3664f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/1ae2178dd83c1247edf9866bb17d77309a3664f4"}]},"branch":"refs/heads/master"},"b285dcdfa17996c90ca14803f657aeec2145ff9c":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-01-24 13:49:52.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/10/974610/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/10/974610/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/10/974610/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/10/974610/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/10/974610/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/10/974610/3"}}},"commit":{"parents":[{"commit":"ead36ab260bbc41336096f379578d7f67eb4fcea","subject":"Merge \"Improve IP terminology for instances\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ead36ab260bbc41336096f379578d7f67eb4fcea"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-24 13:45:58.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-24 13:49:52.000000000","tz":0},"subject":"Logs - add phrase in language file","message":"Logs - add phrase in language file\n\nChange-Id: I46642f2f8269d30179f2e8345f2f5b6879f95f48\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b285dcdfa17996c90ca14803f657aeec2145ff9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b285dcdfa17996c90ca14803f657aeec2145ff9c"}]},"branch":"refs/heads/master","description":"Edit commit message"},"aa4adc5b573bb15382cba5dd1cee83660d4b29fb":{"kind":"REWORK","_number":4,"created":"2026-01-24 13:50:55.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/10/974610/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/10/974610/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/10/974610/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/10/974610/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/10/974610/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/10/974610/4"}}},"commit":{"parents":[{"commit":"ead36ab260bbc41336096f379578d7f67eb4fcea","subject":"Merge \"Improve IP terminology for instances\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ead36ab260bbc41336096f379578d7f67eb4fcea"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-24 13:45:58.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-24 13:50:55.000000000","tz":0},"subject":"Logs - add phrase in language file","message":"Logs - add phrase in language file\n\nChange-Id: I46642f2f8269d30179f2e8345f2f5b6879f95f48\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/aa4adc5b573bb15382cba5dd1cee83660d4b29fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/aa4adc5b573bb15382cba5dd1cee83660d4b29fb"}]},"branch":"refs/heads/master"},"661aa664af092b15d2ee4df05223195c90ac88d2":{"kind":"REWORK","_number":5,"created":"2026-01-24 14:47:51.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/10/974610/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/10/974610/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/10/974610/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/10/974610/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/10/974610/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/10/974610/5"}}},"commit":{"parents":[{"commit":"ead36ab260bbc41336096f379578d7f67eb4fcea","subject":"Merge \"Improve IP terminology for instances\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ead36ab260bbc41336096f379578d7f67eb4fcea"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-24 13:45:58.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-24 14:47:51.000000000","tz":0},"subject":"Logs - add phrase in language file","message":"Logs - add phrase in language file\n\nChange-Id: I46642f2f8269d30179f2e8345f2f5b6879f95f48\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/661aa664af092b15d2ee4df05223195c90ac88d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/661aa664af092b15d2ee4df05223195c90ac88d2"}]},"branch":"refs/heads/master"},"50014286cf31476d0527c841543dd6e16142f761":{"kind":"REWORK","_number":6,"created":"2026-01-24 16:58:13.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/10/974610/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/10/974610/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/10/974610/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/10/974610/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/10/974610/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/10/974610/6"}}},"commit":{"parents":[{"commit":"ead36ab260bbc41336096f379578d7f67eb4fcea","subject":"Merge \"Improve IP terminology for instances\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ead36ab260bbc41336096f379578d7f67eb4fcea"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-24 13:45:58.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-24 16:58:13.000000000","tz":0},"subject":"Logs - add phrase in language file","message":"Logs - add phrase in language file\n\nChange-Id: I46642f2f8269d30179f2e8345f2f5b6879f95f48\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/50014286cf31476d0527c841543dd6e16142f761"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/50014286cf31476d0527c841543dd6e16142f761"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
