)]}'
{"id":"openstack%2Fskyline-console~974669","triplet_id":"openstack%2Fskyline-console~stable%2F2025.2~Idb3c3c1220b1fea3e032f1c6094f9a3700a97b66","project":"openstack/skyline-console","branch":"stable/2025.2","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-01-27 02:29:48.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"36709":{"account":{"_account_id":36709,"name":"Uwe Jäger","display_name":"Uwe","email":"uwe.jaeger@valiton.com","username":"uwej711"},"last_update":"2026-01-27 03:39:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idb3c3c1220b1fea3e032f1c6094f9a3700a97b66","subject":"fix: Delete Loadbalancer not possible, fixes 2102154","status":"MERGED","created":"2026-01-26 06:52:49.000000000","updated":"2026-01-27 03:40:07.000000000","submitted":"2026-01-27 03:39:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"974669","cherry_pick_of_change":974407,"cherry_pick_of_patch_set":2,"meta_rev_id":"6e3b17dbf210f97ded71100a6c1fd9127e96130e","_number":974669,"virtual_id_number":974669,"owner":{"_account_id":36709,"name":"Uwe Jäger","display_name":"Uwe","email":"uwe.jaeger@valiton.com","username":"uwej711"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-27 03:39:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-01-27 02:29:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-01-26 07:53:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-01-27 02:29:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-26 07:53:50.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-01-26 07:53:50.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2026-01-26 08:00:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"876c2c34b86ba4515d37f5315490473315791227","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36709,"name":"Uwe Jäger","display_name":"Uwe","email":"uwe.jaeger@valiton.com","username":"uwej711"},"date":"2026-01-26 06:52:49.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"2ac6f6137efca1a31830d1b87f1850c9a9e0e80d","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-26 07:53:50.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"18bcabf9d53f40751c3841568ca16e7842ae6cf7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 08:00:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3dfb6c87f7584e828df643ca8e815be7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9aafc9a51144e2390b7cb4c13bc2c0d : SUCCESS in 6m 03s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/de7ff38112764071a1232e50def8497d : SUCCESS in 12m 34s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/b1beeeb9166042b287eb79760c406026 : SUCCESS in 3m 42s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/83cdda2c1e694921a7237e5c1074fcf1 : SUCCESS in 5m 03s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/18a67fe2b53642e6910d2bbf9755eb38 : SUCCESS in 1h 02m 19s","accounts_in_message":[],"_revision_number":1},{"id":"37bcf1a20510c1b5b6257e059a7f0a07675cbcd6","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-01-27 02:29:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"30696a714e0fd06674e6d9d8dc26518a0c9708e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 02:30:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"231a23a5aab1baedb0e8864b60fce3aa67046dbe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 03:39:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/523f03b7c1994d4986feea3fbcab555e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63732a2e20884666be0af4208db8ddad : SUCCESS in 6m 02s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/5f035e28ce2446ef8253a56634729f52 : SUCCESS in 11m 32s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/0dd5387702764f9fbd5689039d3f0ebe : SUCCESS in 7m 32s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/a955bd1c34d7437fa7f1894ac20a36de : SUCCESS in 6m 44s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/6a20bf79d56b4a60a868de19a6583899 : SUCCESS in 1h 00m 45s","accounts_in_message":[],"_revision_number":1},{"id":"f286bfad726b9c923dde298ecf0981ffd3ed402b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 03:39:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6e3b17dbf210f97ded71100a6c1fd9127e96130e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 03:40:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a65b02b41fc4e42b4d57759339efdfa\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec7cdb0a9dc945aeb29992280b474a04 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"86a3ea1a5eff6f91e2fc707ef96947bcfe5da030","revisions":{"86a3ea1a5eff6f91e2fc707ef96947bcfe5da030":{"kind":"REWORK","_number":1,"created":"2026-01-26 06:52:49.000000000","uploader":{"_account_id":36709,"name":"Uwe Jäger","display_name":"Uwe","email":"uwe.jaeger@valiton.com","username":"uwej711"},"ref":"refs/changes/69/974669/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/69/974669/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/69/974669/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/69/974669/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/69/974669/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/69/974669/1"}}},"commit":{"parents":[{"commit":"a25965a66c1fe6db8a7350e9ed858bda71700e67","subject":"Fix project list for federated users","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a25965a66c1fe6db8a7350e9ed858bda71700e67"}]}],"author":{"name":"Uwe Jäger","email":"uwe.jaeger@valiton.com","date":"2026-01-23 10:11:27.000000000","tz":60},"committer":{"name":"Uwe Jäger","email":"uwe.jaeger@valiton.com","date":"2026-01-26 06:52:49.000000000","tz":0},"subject":"fix: Delete Loadbalancer not possible, fixes 2102154","message":"fix: Delete Loadbalancer not possible, fixes 2102154\n\nThis fixes https://bugs.launchpad.net/skyline-console/+bug/2102154 by\nmoving the cascade: true from the request body to url parameters.\n\nOtherwise nginx complains about the body being too large.\n\nCloses-Bug: #2102154\n\nChange-Id: Idb3c3c1220b1fea3e032f1c6094f9a3700a97b66\nSigned-off-by: Uwe Jäger \u003cuwe.jaeger@valiton.com\u003e\n(cherry picked from commit 8f8bc87fd33fc31d6a11cc3b26eb6898bcea003d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/86a3ea1a5eff6f91e2fc707ef96947bcfe5da030"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/86a3ea1a5eff6f91e2fc707ef96947bcfe5da030"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
