)]}'
{"id":"openstack%2Fskyline-console~974673","triplet_id":"openstack%2Fskyline-console~master~I24c6550a7a3e469f2f49d64e37d74248767018ee","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-01-27 07:19:27.000000000","reason":"\u003cGERRIT_ACCOUNT_37632\u003e replied on the change","reason_account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-01-27 02:43:08.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-01-27 09:42:18.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"I24c6550a7a3e469f2f49d64e37d74248767018ee","subject":"Hypervisor Detail - Members replace to Instances","status":"MERGED","created":"2026-01-26 08:39:24.000000000","updated":"2026-01-27 13:19:08.000000000","submitted":"2026-01-27 13:10:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"974673","meta_rev_id":"421e450923c230b86cf9461993fa494ac68b56c6","_number":974673,"virtual_id_number":974673,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-27 13:10:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-01-27 02:43:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-01-27 02:43:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-26 08:40:23.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2026-01-26 08:40:23.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"},{"updated":"2026-01-26 09:54:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f1acb5df62757100f35a211821280848f0924813","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-26 08:39:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7f24f6af610fea5b89e6adf094f7f2441f5c9a08","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-26 08:40:15.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"56ba64a63747b5c6599a43f256c0102951e0a738","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-26 08:40:23.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"97ca6dea4aa2b0c2d4fa23f8e9c4865a84a09395","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 09:54:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bacf6ee4732640fe884d5da40c2f3925\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f75db5868164dd385e3c92698291edf : SUCCESS in 6m 45s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/a739712e46264739aab88b56900411aa : SUCCESS in 10m 45s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/3e6acbedfd3445f7babb36bf70b2e29f : SUCCESS in 7m 25s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/2969b984837a4b4b9d77a219453ba8d0 : SUCCESS in 4m 30s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/9e8c2981af284acbb407cbcd7719d0dd : SUCCESS in 1h 12m 47s","accounts_in_message":[],"_revision_number":2},{"id":"bedb0a3cb6ba169f793ef7e4005c363f72e43518","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-01-27 02:43:08.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"af13487efaa711de38cb1ec41bca2dd4148fe7ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 02:43:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a880a6618d7244b082b7e818cccc1a9cc0727448","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 05:25:48.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/06b8e49b5af4407f87ace027ad2a9348\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ac0e1dfdb16490a84736a532d069a9c : SUCCESS in 5m 13s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/d919a03d8fd447b1a913b9d1199e4b86 : SUCCESS in 5m 05s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/322d73efc33749b89ffe456694b666c4 : SUCCESS in 5m 13s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/6ee2b098ff3b4d7aa218f2ac215126ac : SUCCESS in 4m 02s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/2c5385e677b34f47a93ff3f09db1b98f : FAILURE in 54m 00s","accounts_in_message":[],"_revision_number":2},{"id":"193820f71f27de25b74598415b35d330c7e2a41e","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-27 07:19:27.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"57bd15e145702473208e1ee981d2e18b27a20c76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 08:45:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4a9b9c99cfe4ec99364a4a68545e649\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/053332b4ac5341818f128977cdf93fcc : SUCCESS in 4m 24s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/a56cb8c1e14e45eba13d57bc549f685d : SUCCESS in 11m 26s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/b872d16d800b4d5bb6600bcbbdcdbb84 : SUCCESS in 3m 08s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/004ad320202e414a82f92d7dbb069709 : SUCCESS in 3m 11s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/05a77860d8934d1c9cee516f275fcdf8 : SUCCESS in 1h 20m 58s","accounts_in_message":[],"_revision_number":2},{"id":"2fb7b50cbe7e472fb5241a612a363e2503b3e242","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 08:47:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8d580c770ddfc37c35e0b3f2386fcdd9cfee1c70","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 09:41:56.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/edf5d7adfe154b71b7771787ae0f6b80\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3ed1ccb5cbc4103bbba84fbdfb817d4 : SUCCESS in 6m 18s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/bd3aedfdc0294c8a9051218ef5930307 : SUCCESS in 11m 18s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/7303d5296c354e1cb9e1cbe3b835aead : SUCCESS in 3m 27s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/d5369376db5844dc9b4cf5b8caa2a03a : SUCCESS in 4m 06s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/0b4844c3c8364f9c911a5f8ad326cd5e : FAILURE in 53m 56s","accounts_in_message":[],"_revision_number":2},{"id":"f1fde855aa47d4c0f90355e79d44979d8bbc7c66","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-27 09:42:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b6eb8857ed1f213cd0ff52f66e1d9d39519d9694","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 11:31:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85a444473bd2459fb49600635001b2a8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64c904a86df147a3bd634d9e18efd705 : SUCCESS in 6m 22s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/9bda56aee7aa4fd9865b7f137a8ed885 : SUCCESS in 12m 38s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/ffb858247b1449bb9a2c053e330094a0 : SUCCESS in 4m 58s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/1b30522fec83421893b84fa976e99d39 : SUCCESS in 3m 26s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/b5047530de2a4008a71b194604c67e38 : SUCCESS in 1h 35m 42s","accounts_in_message":[],"_revision_number":2},{"id":"bc806356c7c8b52b19ab5db7579c04c61be2e981","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 11:31:39.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"01d91c42e56ea32887e3ff34b2c81537dc1cb5bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 13:10:40.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/522583da0fea4fd9a5834eac6f65c008\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d801c10a6dc43149e969e84cb3df474 : SUCCESS in 6m 49s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/c50d31c76c5c4e27bf594eb3bcf5979b : SUCCESS in 13m 49s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/834b668170fa4d269aeb631b7c4b1a0b : SUCCESS in 7m 00s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/6c2a2d840e8743d5b23b73134bbecd8b : SUCCESS in 8m 14s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/09c29d9631a5473a9a68e73e818afd00 : SUCCESS in 1h 20m 48s","accounts_in_message":[],"_revision_number":2},{"id":"1da5f0d758c5e77174d32b0018ba68e92130b333","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 13:10:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"421e450923c230b86cf9461993fa494ac68b56c6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 13:19:08.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9cd71c779498413c862d6847351b9754\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc081a5892a94bfa9d2893e9cae6c422 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b3c63a7567cd7647461a22240074cb7be4146c90","revisions":{"bab321b9965d788e5a6e8a8b1a6edb3d8abb9b0a":{"kind":"REWORK","_number":1,"created":"2026-01-26 08:39:24.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/73/974673/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/73/974673/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/73/974673/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/73/974673/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/73/974673/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/73/974673/1"}}},"commit":{"parents":[{"commit":"e3d4e427eecddc57bbb5b7c0a740cc83475e0a2e","subject":"Merge \"docs: add missing pymysql dependency for database connectivity\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e3d4e427eecddc57bbb5b7c0a740cc83475e0a2e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-26 08:39:24.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-26 08:39:24.000000000","tz":0},"subject":"Hypervisor Detail - Members replace to Instances","message":"Hypervisor Detail - Members replace to Instances\n\nIn this tab showing instances.\n\nChange-Id: I24c6550a7a3e469f2f49d64e37d74248767018ee\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/bab321b9965d788e5a6e8a8b1a6edb3d8abb9b0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/bab321b9965d788e5a6e8a8b1a6edb3d8abb9b0a"}]},"branch":"refs/heads/master"},"b3c63a7567cd7647461a22240074cb7be4146c90":{"kind":"REWORK","_number":2,"created":"2026-01-26 08:40:15.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/73/974673/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/73/974673/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/73/974673/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/73/974673/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/73/974673/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/73/974673/2"}}},"commit":{"parents":[{"commit":"e3d4e427eecddc57bbb5b7c0a740cc83475e0a2e","subject":"Merge \"docs: add missing pymysql dependency for database connectivity\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e3d4e427eecddc57bbb5b7c0a740cc83475e0a2e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-26 08:39:24.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-26 08:40:15.000000000","tz":0},"subject":"Hypervisor Detail - Members replace to Instances","message":"Hypervisor Detail - Members replace to Instances\n\nIn this tab showing instances.\n\nChange-Id: I24c6550a7a3e469f2f49d64e37d74248767018ee\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b3c63a7567cd7647461a22240074cb7be4146c90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b3c63a7567cd7647461a22240074cb7be4146c90"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
