)]}'
{"id":"openstack%2Fskyline-console~974685","triplet_id":"openstack%2Fskyline-console~master~If442426ec559d93897190f29d2e039327acfa80a","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-01-27 07:19:54.000000000","reason":"\u003cGERRIT_ACCOUNT_37632\u003e replied on the change","reason_account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-01-27 02:49:28.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-01-27 11:07:46.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"If442426ec559d93897190f29d2e039327acfa80a","subject":"Remove no need spaces","status":"MERGED","created":"2026-01-26 09:39:34.000000000","updated":"2026-01-27 13:57:50.000000000","submitted":"2026-01-27 13:56:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"974685","meta_rev_id":"a91a3473ec0e3360e125e461a6c4ade839488e25","_number":974685,"virtual_id_number":974685,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-27 13:56:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-01-27 02:49:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-01-27 02:49:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-26 09:51:02.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"},{"updated":"2026-01-26 09:51:02.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2026-01-26 11:18:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7fef13da4eb1f8ba958d1b795d061139203a7de8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-26 09:39:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d84a627a4859ef1991fa275b2464a75cbf3afa32","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-26 09:44:43.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"43302e7e0f4520c0ed6e0a0cd48378f2a5733579","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-26 09:44:55.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"fbfd5832b05f8295a145d974d6775e2dfc0755e2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-26 09:48:08.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"e3974d7a1069a5b4b7dc0ae1fdf517a4aa6f5d74","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-26 09:50:38.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"31ccaa55b559a56afc0353a13e2dd09969d6936c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-26 09:50:44.000000000","message":"Patch Set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"1cc46a5f1359c9053b00e8174e7e106c967d98f0","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-26 09:51:02.000000000","message":"Patch Set 6:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":6},{"id":"f8464cb8b62c97958aaf1a0f0931f6c2f2bf2685","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-26 09:52:26.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"91de5081baa4559e7828531817b159a0ff1196c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 11:18:51.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c24ffe0b313348f4a5d55c30a2f0a687\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5df3b8f4fbe14411b2b1125e904d65a9 : SUCCESS in 8m 09s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/d9f2d4009ac046eab2b19ee5314bbf33 : SUCCESS in 9m 03s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/b689142323334b6e87704e5e45bf8478 : SUCCESS in 3m 25s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/0d3a7561c1bf4434915a2be88c0a794e : SUCCESS in 6m 11s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/032cc96273764eed8920be4dec89bd30 : SUCCESS in 1h 19m 58s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/72a74e0c41c74c0c99aec4facaa0dcbd : SUCCESS in 56m 10s","accounts_in_message":[],"_revision_number":7},{"id":"303f9d14233508b65cd38e2886b644f79bc43740","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-01-27 02:49:28.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"a5d9b464cf9db42fdf697b75e8320ada4acb4158","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 02:50:11.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"f58717617c2cdf1ca68befe6de34c0f700a825cc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 05:35:18.000000000","message":"Patch Set 7: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7656b699c3664c9583b0db53f47844d3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae571691aaf64f7c88e4a7d6b7b0d267 : SUCCESS in 4m 44s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/47cd5c0031aa44dea879cc87ec917dc2 : SUCCESS in 4m 13s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/527ca875c8ed408d8bab573b0dd930e0 : SUCCESS in 7m 24s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/286c87ded6554bcc99030cde09c5cbbf : SUCCESS in 7m 39s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/ced3203bf6b54866adc31397006a3304 : FAILURE in 53m 02s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/e263360d407543499facc5c788b8df85 : SUCCESS in 52m 47s","accounts_in_message":[],"_revision_number":7},{"id":"c51e097df3ec4d9d5b9f8f78f31f6d3dc65fa849","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-27 07:19:54.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"f2fcad783b4fd3512b4226def6864bc5b62cfb2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 08:54:07.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3815536bd91446e2b863da58a1c3d29d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3124f586e994e978c7292144ca0ee2d : SUCCESS in 6m 25s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/3a722348d528431a99d128f90f2e0bb5 : SUCCESS in 7m 19s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/ce3c5def1f484b0e9a37275bfee7b473 : SUCCESS in 8m 54s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/a387125e092247ebba9d764970a2b9c7 : SUCCESS in 4m 32s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/b87b7aa1c5154e01803c87c5219f3b0e : SUCCESS in 1h 24m 28s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/9d6c480e014749038773cbb74c04c3d2 : SUCCESS in 42m 25s","accounts_in_message":[],"_revision_number":7},{"id":"2f67f1773a5a3147c63aa1e53bdeac9e1ae136e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 08:54:35.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"a9c37c5402a1e38bf3c1cc1d95fc81bcb29fca61","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 11:06:06.000000000","message":"Patch Set 7: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4cf8d0ec33b1466eae729e09e0d94ccb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a09264eb62e9412e8cb1e5cf85aa8c5e : SUCCESS in 7m 53s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/eb60b24184094d5e86b06ea07578985d : SUCCESS in 11m 04s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/6580dba6ba7f45b183a306671a4529a2 : SUCCESS in 8m 13s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/0ad94557597149ae9a8693dce067ec5a : SUCCESS in 6m 34s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/8d7ab307b8244dbd98fc29486d0891b4 : SUCCESS in 1h 23m 41s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/c0df17cbd1144fe99185d1e3eb20c4d1 : FAILURE in 49m 50s","accounts_in_message":[],"_revision_number":7},{"id":"c4c6c0cecdb0a1f2f5c7bd6a0ac04b4cdd2dd6cc","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-27 11:07:46.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"efe35fd6fc03a67c9737ed440c7df4fdf25822fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 12:37:39.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7efdea202ad4e32894c235d39f6f28b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ceb3b338d7947da96c256f8f7a6c046 : SUCCESS in 7m 01s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/b0cc39ee9c8a4499aa88f2dd26968679 : SUCCESS in 11m 22s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/9d8fa522b4da48b89f32bcc9f8078980 : SUCCESS in 7m 39s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/bb25d5ef4fe44436805eeeab8cee4098 : SUCCESS in 7m 30s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/ca95f1be7c894e92881ee409d0f40706 : SUCCESS in 1h 23m 16s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/d85087ce50ed4f1db75b419e3bd8e87c : SUCCESS in 55m 13s","accounts_in_message":[],"_revision_number":7},{"id":"0c30ceeadf517e6798e4a95fa8e38208fb1a112a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 12:38:06.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"1f979c20b4573350ea8665f157fdc27e721469dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 13:56:52.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25495b56d5c74f86b506b2be6e41c402\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d863ad6193be4120ab5c08623cf9983d : SUCCESS in 6m 44s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/959dbed799a3442da621cfb09c1353fb : SUCCESS in 11m 51s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/ae2c4f9d3b5042e2b50e3ccf3b2105a4 : SUCCESS in 8m 22s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/382102028fc4482a97242b1d31c37e7c : SUCCESS in 3m 20s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/a93fbb43aea145e28952817b3eac3193 : SUCCESS in 1h 03m 01s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/22fcad8baa6c471495f806d9e2356413 : SUCCESS in 54m 04s","accounts_in_message":[],"_revision_number":7},{"id":"527c6b693528ab18c5c200e56729ef8361d67dbe","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 13:56:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"a91a3473ec0e3360e125e461a6c4ade839488e25","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 13:57:50.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0bedba6586c44c2186bc5109a679165c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2230ff937c6c48f380d6067e97bb9784 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"85c433fa52424a95414767d59bb09bdf4c096b49","revisions":{"3b244eb40ee516623f53c7b2f2391fd2874d0316":{"kind":"REWORK","_number":1,"created":"2026-01-26 09:39:34.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/85/974685/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/85/974685/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/85/974685/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/85/974685/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/85/974685/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/85/974685/1"}}},"commit":{"parents":[{"commit":"e3d4e427eecddc57bbb5b7c0a740cc83475e0a2e","subject":"Merge \"docs: add missing pymysql dependency for database connectivity\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e3d4e427eecddc57bbb5b7c0a740cc83475e0a2e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-26 09:39:34.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-26 09:39:34.000000000","tz":0},"subject":"Remove no need space","message":"Remove no need space\n\nChange-Id: If442426ec559d93897190f29d2e039327acfa80a\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/3b244eb40ee516623f53c7b2f2391fd2874d0316"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/3b244eb40ee516623f53c7b2f2391fd2874d0316"}]},"branch":"refs/heads/master"},"26770c4c85c5cf0fee5f5bc348231f3c12a35cd7":{"kind":"REWORK","_number":2,"created":"2026-01-26 09:44:43.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/85/974685/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/85/974685/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/85/974685/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/85/974685/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/85/974685/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/85/974685/2"}}},"commit":{"parents":[{"commit":"e3d4e427eecddc57bbb5b7c0a740cc83475e0a2e","subject":"Merge \"docs: add missing pymysql dependency for database connectivity\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e3d4e427eecddc57bbb5b7c0a740cc83475e0a2e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-26 09:39:34.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-26 09:44:43.000000000","tz":0},"subject":"Remove no need space","message":"Remove no need space\n\nChange-Id: If442426ec559d93897190f29d2e039327acfa80a\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/26770c4c85c5cf0fee5f5bc348231f3c12a35cd7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/26770c4c85c5cf0fee5f5bc348231f3c12a35cd7"}]},"branch":"refs/heads/master"},"4277de2343820a15feb3e5248b0fe3a863317f48":{"kind":"REWORK","_number":3,"created":"2026-01-26 09:44:55.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/85/974685/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/85/974685/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/85/974685/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/85/974685/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/85/974685/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/85/974685/3"}}},"commit":{"parents":[{"commit":"e3d4e427eecddc57bbb5b7c0a740cc83475e0a2e","subject":"Merge \"docs: add missing pymysql dependency for database connectivity\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e3d4e427eecddc57bbb5b7c0a740cc83475e0a2e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-26 09:39:34.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-26 09:44:55.000000000","tz":0},"subject":"Remove no need space","message":"Remove no need space\n\nChange-Id: If442426ec559d93897190f29d2e039327acfa80a\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/4277de2343820a15feb3e5248b0fe3a863317f48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/4277de2343820a15feb3e5248b0fe3a863317f48"}]},"branch":"refs/heads/master"},"639863530e0952ae819e87ea8f5e4d9a847dc015":{"kind":"REWORK","_number":4,"created":"2026-01-26 09:48:08.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/85/974685/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/85/974685/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/85/974685/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/85/974685/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/85/974685/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/85/974685/4"}}},"commit":{"parents":[{"commit":"e3d4e427eecddc57bbb5b7c0a740cc83475e0a2e","subject":"Merge \"docs: add missing pymysql dependency for database connectivity\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e3d4e427eecddc57bbb5b7c0a740cc83475e0a2e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-26 09:39:34.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-26 09:48:08.000000000","tz":0},"subject":"Remove no need space","message":"Remove no need space\n\nChange-Id: If442426ec559d93897190f29d2e039327acfa80a\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/639863530e0952ae819e87ea8f5e4d9a847dc015"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/639863530e0952ae819e87ea8f5e4d9a847dc015"}]},"branch":"refs/heads/master"},"d55c0b2167157775df4183397700b0a5d56a6ab3":{"kind":"REWORK","_number":5,"created":"2026-01-26 09:50:38.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/85/974685/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/85/974685/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/85/974685/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/85/974685/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/85/974685/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/85/974685/5"}}},"commit":{"parents":[{"commit":"e3d4e427eecddc57bbb5b7c0a740cc83475e0a2e","subject":"Merge \"docs: add missing pymysql dependency for database connectivity\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e3d4e427eecddc57bbb5b7c0a740cc83475e0a2e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-26 09:39:34.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-26 09:50:37.000000000","tz":0},"subject":"Remove no need space","message":"Remove no need space\n\nChange-Id: If442426ec559d93897190f29d2e039327acfa80a\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d55c0b2167157775df4183397700b0a5d56a6ab3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d55c0b2167157775df4183397700b0a5d56a6ab3"}]},"branch":"refs/heads/master"},"773a1799e7c0ec8ad408b3c9a4ecff8350639005":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2026-01-26 09:50:44.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/85/974685/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/85/974685/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/85/974685/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/85/974685/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/85/974685/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/85/974685/6"}}},"commit":{"parents":[{"commit":"e3d4e427eecddc57bbb5b7c0a740cc83475e0a2e","subject":"Merge \"docs: add missing pymysql dependency for database connectivity\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e3d4e427eecddc57bbb5b7c0a740cc83475e0a2e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-26 09:39:34.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-26 09:50:44.000000000","tz":0},"subject":"Remove no need spaces","message":"Remove no need spaces\n\nChange-Id: If442426ec559d93897190f29d2e039327acfa80a\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/773a1799e7c0ec8ad408b3c9a4ecff8350639005"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/773a1799e7c0ec8ad408b3c9a4ecff8350639005"}]},"branch":"refs/heads/master","description":"Edit commit message"},"85c433fa52424a95414767d59bb09bdf4c096b49":{"kind":"REWORK","_number":7,"created":"2026-01-26 09:52:26.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/85/974685/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/85/974685/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/85/974685/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/85/974685/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/85/974685/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/85/974685/7"}}},"commit":{"parents":[{"commit":"e3d4e427eecddc57bbb5b7c0a740cc83475e0a2e","subject":"Merge \"docs: add missing pymysql dependency for database connectivity\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e3d4e427eecddc57bbb5b7c0a740cc83475e0a2e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-26 09:39:34.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-26 09:52:26.000000000","tz":0},"subject":"Remove no need spaces","message":"Remove no need spaces\n\nChange-Id: If442426ec559d93897190f29d2e039327acfa80a\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/85c433fa52424a95414767d59bb09bdf4c096b49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/85c433fa52424a95414767d59bb09bdf4c096b49"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
