)]}'
{"id":"openstack%2Fskyline-console~974720","triplet_id":"openstack%2Fskyline-console~master~Ic09a19c0a57c04bb4034b8d27b5ac0967410a062","project":"openstack/skyline-console","branch":"master","topic":"reset","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-01-29 02:00:53.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"34517":{"account":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"last_update":"2026-01-29 04:10:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic09a19c0a57c04bb4034b8d27b5ac0967410a062","subject":"feat: Add instance reset state functionality","status":"MERGED","created":"2026-01-26 12:39:23.000000000","updated":"2026-01-29 04:11:18.000000000","submitted":"2026-01-29 04:10:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"974720-reset","meta_rev_id":"bea9c0d752a0245bbb28ae86154c1dc93b08133c","_number":974720,"virtual_id_number":974720,"owner":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-29 04:10:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-01-29 02:00:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-01-29 02:00:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-26 14:12:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-26 14:37:44.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2026-01-26 14:37:44.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"CC"},{"updated":"2026-01-26 14:37:44.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"}],"messages":[{"id":"9736d4106af056d592cd2a9f6b7d6573264ad591","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"date":"2026-01-26 12:39:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b92adfcc24f62280e98d563fd897225991bdc2ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 14:12:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e675fce6f8343eb9e19cc25e607bc72\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/031dee97519f4fcb99021066541c0a73 : SUCCESS in 7m 09s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/8591acde4fe34d6b85199fe44d4c43ab : SUCCESS in 11m 28s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/6ebae73c32be42a9824e68e31ac0523d : SUCCESS in 3m 43s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/791cd5bbc28d48cbaaaa666b25460724 : SUCCESS in 4m 14s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/84269c8dd4c74249ae2362bd3595b7d3 : SUCCESS in 1h 23m 12s","accounts_in_message":[],"_revision_number":1},{"id":"16016783f1b50658111b6b84e233f4fab0b05d49","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-01-27 03:13:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b8eaca692b6acc56c5893c92d11b70b972409d27","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"date":"2026-01-27 05:25:16.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"0f51f39a66164d2faaf923e9420386c0d2f3a6fb","author":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"date":"2026-01-27 05:25:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c01a5ebd02eeb24f592069ad614d65d06e8498fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 06:52:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d64175863034d1589a960bde89f065a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9827f7ae0a3445acb50355c197846529 : SUCCESS in 6m 09s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/85c1fe60e7054c18a3d23e8b93e351b3 : SUCCESS in 11m 15s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/7a99ad7015414becb0b19c9e528cf0f4 : SUCCESS in 4m 45s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/f1d0913c97b14c5584a1fb0c0ad19b18 : SUCCESS in 6m 22s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/b25c912df81d4ea3904c612e31ce4328 : SUCCESS in 1h 22m 26s","accounts_in_message":[],"_revision_number":2},{"id":"c842691c5848b99a236a378af2183af4216cd791","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-01-28 03:34:47.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"66397d63d79afee277199565f4e303cad8ffe61c","author":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"date":"2026-01-28 07:41:32.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0a85fc089e5ff1cabbf97c461dac81b58200eade","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"date":"2026-01-28 09:05:18.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"c2e7e0f4ed9e7142c57f533a832469f55242d4a1","author":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"date":"2026-01-28 09:07:51.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"dff41ceb7137c68b5b47dc4021930b1df5e90388","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 10:01:37.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ff3bf93185dc4c9786a4d5442977c830\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82389cb04639439590a72911a760acfd : SUCCESS in 3m 23s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/5206133b182a47b7a1091a4191b0fe5a : SUCCESS in 5m 26s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/0bf68fe7367e4bada4b310321896cde4 : SUCCESS in 3m 09s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/0ef0e4028de644e399047426318badd5 : SUCCESS in 3m 15s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/ebdb38521a0849e9a4cc2c3eac8c91c1 : FAILURE in 54m 32s","accounts_in_message":[],"_revision_number":3},{"id":"ea42858182bef6ad03975b2bc6d584d3cfda30fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"date":"2026-01-28 10:35:09.000000000","message":"Uploaded patch set 4: New patch set was added with same tree, parent tree, and commit message as Patch Set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"7e8d72a2294238370ce779984b42118b6fc6e19a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 11:45:05.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/867ffc04393744eca250615f37f3daa2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4332c5ef0b8447039b8efae1e0d112b4 : SUCCESS in 6m 37s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/b0c4625fb06a48bbb8dd66a51eb1e1f3 : SUCCESS in 5m 04s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/2c91a8e442c54b5c99f7411d1d2e8335 : SUCCESS in 3m 10s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/020f7addfe3f4417b6746366a8bba9a6 : SUCCESS in 7m 26s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/06a7f0fada88469a8220834a59eb7dde : SUCCESS in 1h 08m 18s","accounts_in_message":[],"_revision_number":4},{"id":"fe7eb31515509a125cbda56c38b74fc564c19558","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-01-29 02:00:53.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"418833b126cc628170864bae0163864ee6ba0739","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 02:02:53.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"93c4284cea99a5641de3f402932ed8f4252c5d77","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 04:10:18.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06efd8ceddb543f2b288ccdf86fd20b0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b67c2ef15d444afa64a9a3e79c0aae4 : SUCCESS in 6m 17s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/09303bb320794e9186e32a608f9cd0ca : SUCCESS in 13m 04s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/5d8b136b022f45b38c7e6c5c363b53e2 : SUCCESS in 7m 41s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/797d94b0e629448ba3b33a6e86f6f229 : SUCCESS in 7m 56s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/2939aba3fc7f4567b1d265aa12407900 : SUCCESS in 1h 27m 50s","accounts_in_message":[],"_revision_number":4},{"id":"fbf76e64b2871cf96bd6583c3e8553111fb3ff7d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 04:10:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"bea9c0d752a0245bbb28ae86154c1dc93b08133c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 04:11:18.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26f68fba51cf4f648991780eefa92596\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc0e9ce0b086495ca32f976c0456dbed : SUCCESS in 47s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"982412ce64f7a537d78cdce2a081b04dc6e596b4","revisions":{"eda2a4efea85c1580d038ce6a971f48c73743280":{"kind":"REWORK","_number":1,"created":"2026-01-26 12:39:23.000000000","uploader":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"ref":"refs/changes/20/974720/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/20/974720/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/974720/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/974720/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/974720/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/20/974720/1"}}},"commit":{"parents":[{"commit":"e3d4e427eecddc57bbb5b7c0a740cc83475e0a2e","subject":"Merge \"docs: add missing pymysql dependency for database connectivity\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e3d4e427eecddc57bbb5b7c0a740cc83475e0a2e"}]}],"author":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2026-01-26 12:37:16.000000000","tz":330},"committer":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2026-01-26 12:37:16.000000000","tz":330},"subject":"feat: Add instance reset state functionality","message":"feat: Add instance reset state functionality\n\nChange-Id: Ic09a19c0a57c04bb4034b8d27b5ac0967410a062\nSigned-off-by: Sowmya Nethi \u003csowmya.kamavaram@rackspace.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/eda2a4efea85c1580d038ce6a971f48c73743280"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/eda2a4efea85c1580d038ce6a971f48c73743280"}]},"branch":"refs/heads/master"},"f1b777ff0f7becc7912cbf5b961e795932d00b3d":{"kind":"REWORK","_number":2,"created":"2026-01-27 05:25:16.000000000","uploader":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"ref":"refs/changes/20/974720/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/20/974720/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/974720/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/974720/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/974720/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/20/974720/2"}}},"commit":{"parents":[{"commit":"e3d4e427eecddc57bbb5b7c0a740cc83475e0a2e","subject":"Merge \"docs: add missing pymysql dependency for database connectivity\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e3d4e427eecddc57bbb5b7c0a740cc83475e0a2e"}]}],"author":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2026-01-26 12:37:16.000000000","tz":330},"committer":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2026-01-27 05:24:13.000000000","tz":330},"subject":"feat: Add instance reset state functionality","message":"feat: Add instance reset state functionality\n\nChange-Id: Ic09a19c0a57c04bb4034b8d27b5ac0967410a062\nSigned-off-by: Sowmya Nethi \u003csowmya.kamavaram@rackspace.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f1b777ff0f7becc7912cbf5b961e795932d00b3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f1b777ff0f7becc7912cbf5b961e795932d00b3d"}]},"branch":"refs/heads/master"},"78c7ed0f322f87e6ffc0f9c7f8fa843abe5ab27d":{"kind":"REWORK","_number":3,"created":"2026-01-28 09:05:18.000000000","uploader":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"ref":"refs/changes/20/974720/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/20/974720/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/974720/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/974720/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/974720/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/20/974720/3"}}},"commit":{"parents":[{"commit":"e3d4e427eecddc57bbb5b7c0a740cc83475e0a2e","subject":"Merge \"docs: add missing pymysql dependency for database connectivity\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e3d4e427eecddc57bbb5b7c0a740cc83475e0a2e"}]}],"author":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2026-01-26 12:37:16.000000000","tz":330},"committer":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2026-01-28 09:04:59.000000000","tz":330},"subject":"feat: Add instance reset state functionality","message":"feat: Add instance reset state functionality\n\nChange-Id: Ic09a19c0a57c04bb4034b8d27b5ac0967410a062\nSigned-off-by: Sowmya Nethi \u003csowmya.kamavaram@rackspace.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/78c7ed0f322f87e6ffc0f9c7f8fa843abe5ab27d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/78c7ed0f322f87e6ffc0f9c7f8fa843abe5ab27d"}]},"branch":"refs/heads/master"},"982412ce64f7a537d78cdce2a081b04dc6e596b4":{"kind":"NO_CHANGE","_number":4,"created":"2026-01-28 10:35:09.000000000","uploader":{"_account_id":34517,"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","username":"sowmya.kamavaram"},"ref":"refs/changes/20/974720/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/20/974720/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/974720/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/974720/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/20/974720/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/20/974720/4"}}},"commit":{"parents":[{"commit":"e3d4e427eecddc57bbb5b7c0a740cc83475e0a2e","subject":"Merge \"docs: add missing pymysql dependency for database connectivity\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e3d4e427eecddc57bbb5b7c0a740cc83475e0a2e"}]}],"author":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2026-01-26 12:37:16.000000000","tz":330},"committer":{"name":"Sowmya Nethi","email":"sowmya.kamavaram@rackspace.com","date":"2026-01-28 10:29:09.000000000","tz":330},"subject":"feat: Add instance reset state functionality","message":"feat: Add instance reset state functionality\n\nChange-Id: Ic09a19c0a57c04bb4034b8d27b5ac0967410a062\nSigned-off-by: Sowmya Nethi \u003csowmya.kamavaram@rackspace.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/982412ce64f7a537d78cdce2a081b04dc6e596b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/982412ce64f7a537d78cdce2a081b04dc6e596b4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
