)]}'
{"id":"openstack%2Fskyline-console~974724","triplet_id":"openstack%2Fskyline-console~stable%2F2025.1~Idb3c3c1220b1fea3e032f1c6094f9a3700a97b66","project":"openstack/skyline-console","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-01-27 02:41:58.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"36709":{"account":{"_account_id":36709,"name":"Uwe Jäger","display_name":"Uwe","email":"uwe.jaeger@valiton.com","username":"uwej711"},"last_update":"2026-01-27 05:25:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idb3c3c1220b1fea3e032f1c6094f9a3700a97b66","subject":"fix: Delete Loadbalancer not possible, fixes 2102154","status":"MERGED","created":"2026-01-26 13:22:07.000000000","updated":"2026-01-27 05:26:38.000000000","submitted":"2026-01-27 05:25:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"974724","cherry_pick_of_change":974407,"cherry_pick_of_patch_set":2,"meta_rev_id":"d40c7472373386c75fd8a6694a0ca4004b5e2fc7","_number":974724,"virtual_id_number":974724,"owner":{"_account_id":36709,"name":"Uwe Jäger","display_name":"Uwe","email":"uwe.jaeger@valiton.com","username":"uwej711"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-27 05:25:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-01-27 02:41:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-01-26 13:41:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-01-27 02:41:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-26 13:41:43.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-01-26 13:41:43.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2026-01-26 14:27:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"255aa1d8c5724a43b201c15cd04c980df4a3ede9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36709,"name":"Uwe Jäger","display_name":"Uwe","email":"uwe.jaeger@valiton.com","username":"uwej711"},"date":"2026-01-26 13:22:07.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"531d363e7a9a1f8a7efad25a4a314da87231d657","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-26 13:41:43.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"2ebc525f84a4665c3cdc24978e303e0b73c24967","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-26 14:27:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d58e33eba034e12b928c28c50b81318\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8f0529f1413423bae624fbaa94618f6 : SUCCESS in 4m 28s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/304ded72b5294831938578da4e00b538 : SUCCESS in 8m 41s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/03aec196b2cf4880a1e1087d3a47fd81 : SUCCESS in 3m 33s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/c6bad15d597b4979bf25a67a80ff7b88 : SUCCESS in 4m 44s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/bcd866464ad540b5a6e7732f7ab95fd2 : SUCCESS in 1h 02m 35s","accounts_in_message":[],"_revision_number":1},{"id":"3ce102f5afa9b933723760f6605bb54806eac0ec","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-01-27 02:41:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9cb41f9456d583205e215069895fb22eba32742e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 02:42:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1882c7f8774367b4456ff27420a6f99f178716f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 05:25:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f084c73f13f4b9ab408fd1d91ac0197\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ffc1e94987c43dc8b9960f4b1aec6f4 : SUCCESS in 6m 29s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/675ab88156694e148f1ba214571e2c75 : SUCCESS in 10m 41s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/6124eb33681e465792fa100c69e1c901 : SUCCESS in 7m 29s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/f9f186bd35924b8e93f3bcf01505b5ac : SUCCESS in 4m 39s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/0b787d0a37f04acbb94ed1b4f3717ad5 : SUCCESS in 1h 02m 30s","accounts_in_message":[],"_revision_number":1},{"id":"86fc16ca88be56d2837f30897d1470c285677c82","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 05:25:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d40c7472373386c75fd8a6694a0ca4004b5e2fc7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 05:26:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/073c6bd1909c4fdbb6c583a254c437ee\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49111704b12b44ac98579a3d4b900ba2 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"409b1bb06ac4d126b6133cd41b5f5adb5dfd6c9c","revisions":{"409b1bb06ac4d126b6133cd41b5f5adb5dfd6c9c":{"kind":"REWORK","_number":1,"created":"2026-01-26 13:22:07.000000000","uploader":{"_account_id":36709,"name":"Uwe Jäger","display_name":"Uwe","email":"uwe.jaeger@valiton.com","username":"uwej711"},"ref":"refs/changes/24/974724/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/24/974724/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/24/974724/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/24/974724/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/24/974724/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/24/974724/1"}}},"commit":{"parents":[{"commit":"1dd9079c6ef8a088111b948ca4a329e865467c39","subject":"Monkey patch createHash function as workaround for hardcoded md4 hash","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/1dd9079c6ef8a088111b948ca4a329e865467c39"}]}],"author":{"name":"Uwe Jäger","email":"uwe.jaeger@valiton.com","date":"2026-01-23 10:11:27.000000000","tz":60},"committer":{"name":"Uwe Jäger","email":"uwe.jaeger@valiton.com","date":"2026-01-26 13:22:07.000000000","tz":0},"subject":"fix: Delete Loadbalancer not possible, fixes 2102154","message":"fix: Delete Loadbalancer not possible, fixes 2102154\n\nThis fixes https://bugs.launchpad.net/skyline-console/+bug/2102154 by\nmoving the cascade: true from the request body to url parameters.\n\nOtherwise nginx complains about the body being too large.\n\nCloses-Bug: #2102154\n\nChange-Id: Idb3c3c1220b1fea3e032f1c6094f9a3700a97b66\nSigned-off-by: Uwe Jäger \u003cuwe.jaeger@valiton.com\u003e\n(cherry picked from commit 8f8bc87fd33fc31d6a11cc3b26eb6898bcea003d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/409b1bb06ac4d126b6133cd41b5f5adb5dfd6c9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/409b1bb06ac4d126b6133cd41b5f5adb5dfd6c9c"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
