)]}'
{"id":"openstack%2Fskyline-console~974790","triplet_id":"openstack%2Fskyline-console~master~I70ded0eca6b1fd4e2506a0a66ba49d38a171aa42","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-01-27 06:54:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I70ded0eca6b1fd4e2506a0a66ba49d38a171aa42","subject":"Hide Disk Format column on Instance Snapshots page","status":"MERGED","created":"2026-01-26 22:28:33.000000000","updated":"2026-01-27 06:55:57.000000000","submitted":"2026-01-27 06:54:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"974790","meta_rev_id":"237b7337d0c653a582db4435c4c02acb716378c4","_number":974790,"virtual_id_number":974790,"owner":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-27 06:54:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-01-27 03:00:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-01-27 03:00:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-27 00:09:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-27 03:00:58.000000000","updated_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"}],"messages":[{"id":"b525d9e45502d2c1d99f3d60bdf653135a595567","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-26 22:28:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7db515aeca466af07f1419dd84a064745fbe65ff","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-26 22:29:28.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"d828562c2e2cf4f88da8b4c2ca573b8532d8dda9","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-26 22:29:32.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"e099bc1b9d945e9d4cbcef05e9d3698f5ad4c37d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 00:09:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2cd396df1c3443ae9aae38c75bd0368e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08bb95aa5aa44603883e951f34cf1300 : SUCCESS in 4m 06s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/ab48175e5e01400fa1efb0820d35658f : SUCCESS in 13m 18s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/f4c8c9112bc248ce9ae3486f5c76a125 : SUCCESS in 4m 06s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/2e996fcfedee4303b0b3f8a7603ba84f : SUCCESS in 4m 55s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/8e1a18c04e5241c085da5d0c766e845e : SUCCESS in 1h 23m 29s","accounts_in_message":[],"_revision_number":2},{"id":"b19c745eccbfd715b5108c912824a8c2224018f7","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-01-27 03:00:58.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c90a9b5b113b20acb527d376557b530ee3be7850","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 03:01:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f9d4a5a9e58bf37fbd7f68c0e905f1e376598cab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 06:54:26.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3bbdd1e33ae430fbc9f7936ec406065\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6df772609257423abbce356f8cc53c8d : SUCCESS in 6m 32s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/135a050477f14d24b3f7b2ccb55e22c4 : SUCCESS in 10m 54s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/bc7ed744ead34145a7f9683c080aa51b : SUCCESS in 6m 06s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/a68b6f49f119473091ccdf614af704c2 : SUCCESS in 3m 49s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/d7515ee790da4dbf975d3c4393afdea8 : SUCCESS in 1h 19m 25s","accounts_in_message":[],"_revision_number":2},{"id":"08b85b17c712e54cde462a2b10a735954ddf5b3b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 06:54:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"237b7337d0c653a582db4435c4c02acb716378c4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 06:55:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1461d907b34e4b639abc62e4545ee15c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b19e47c9eb74acc94b1d228e72e6684 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7eba44e4c3df105db8a4d40c8b44f780cc3a9700","revisions":{"449e344ac2504e9da22f65768664c87d55580a75":{"kind":"REWORK","_number":1,"created":"2026-01-26 22:28:33.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/90/974790/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/90/974790/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/90/974790/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/90/974790/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/90/974790/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/90/974790/1"}}},"commit":{"parents":[{"commit":"e3d4e427eecddc57bbb5b7c0a740cc83475e0a2e","subject":"Merge \"docs: add missing pymysql dependency for database connectivity\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e3d4e427eecddc57bbb5b7c0a740cc83475e0a2e"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-26 22:28:33.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-26 22:28:33.000000000","tz":0},"subject":"Hide Disk Format column on Instance Snapshots page","message":"Hide Disk Format column on Instance Snapshots page\n\nHide the Disk Format column as it does not provide much value there\nand to improve readability on Instance Snapshots page.\n\nChange-Id: I70ded0eca6b1fd4e2506a0a66ba49d38a171aa42\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/449e344ac2504e9da22f65768664c87d55580a75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/449e344ac2504e9da22f65768664c87d55580a75"}]},"branch":"refs/heads/master"},"7eba44e4c3df105db8a4d40c8b44f780cc3a9700":{"kind":"REWORK","_number":2,"created":"2026-01-26 22:29:28.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/90/974790/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/90/974790/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/90/974790/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/90/974790/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/90/974790/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/90/974790/2"}}},"commit":{"parents":[{"commit":"e3d4e427eecddc57bbb5b7c0a740cc83475e0a2e","subject":"Merge \"docs: add missing pymysql dependency for database connectivity\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e3d4e427eecddc57bbb5b7c0a740cc83475e0a2e"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-26 22:28:33.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-26 22:29:28.000000000","tz":0},"subject":"Hide Disk Format column on Instance Snapshots page","message":"Hide Disk Format column on Instance Snapshots page\n\nHide the Disk Format column as it does not provide much value there\nand to improve readability on Instance Snapshots page.\n\nChange-Id: I70ded0eca6b1fd4e2506a0a66ba49d38a171aa42\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7eba44e4c3df105db8a4d40c8b44f780cc3a9700"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7eba44e4c3df105db8a4d40c8b44f780cc3a9700"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
