)]}'
{"id":"openstack%2Fskyline-console~974955","triplet_id":"openstack%2Fskyline-console~master~I17dec2bab0e2ef99b82fa34589918279c5512084","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-01-29 10:00:31.000000000","reason":"\u003cGERRIT_ACCOUNT_37632\u003e replied on the change","reason_account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-01-29 01:47:52.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-01-29 14:35:56.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"I17dec2bab0e2ef99b82fa34589918279c5512084","subject":"Confirm form, \"confirm\" button update to OK | Part 3","status":"MERGED","created":"2026-01-27 21:31:58.000000000","updated":"2026-01-29 17:33:39.000000000","submitted":"2026-01-29 17:32:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"974955","meta_rev_id":"e5e8c364fce06bc84aaefade7583752628e8a6d8","_number":974955,"virtual_id_number":974955,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-29 17:32:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-01-29 01:47:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-01-29 01:47:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-27 22:29:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-28 12:59:33.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"},{"updated":"2026-01-28 12:59:33.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"}],"messages":[{"id":"6087580e9b9576b27d30eaa7ac1ce949956401ef","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-27 21:31:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6676a9d2af43a7e6a30df74a16ae4802199824a2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-27 21:41:35.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"829d8969e65909528f076d4b820b9f70c82d280e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 22:29:50.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/12d01365be9741b4ae751d147f64c5ba\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e457466c73fb46aaa41eff707d29c751 : SUCCESS in 3m 26s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/16c989b04c9e4df2b5d3811743952d50 : FAILURE in 4m 16s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/dd659200fa3d44049e69ae536ce2dc48 : SUCCESS in 7m 33s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/a728a38ce36a4605acaf88cf25ad24f4 : SUCCESS in 4m 45s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/4342f485d64342968e26628d28df75d3 : FAILURE in 38m 47s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/4c9a8ab58f3c44808b19f2851001524f : FAILURE in 26m 55s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/c6d87d8ae14243af8ec6c034025aaa15 : FAILURE in 10m 14s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/97a3d424a2ef45a680d9b66c5209ad4f : FAILURE in 21m 46s","accounts_in_message":[],"_revision_number":2},{"id":"9fba8a732ffe4b9c142334b650242eb1726399f2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-27 22:44:24.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"c1ee00e8b555232916e37c4136007c57a103c13b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 00:07:31.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/178353a213c348419815f202d08dfaaa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/493038c1f8834a14919b7a0f42e44d3e : SUCCESS in 6m 12s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/92af23ecb1d4446195f18404e441a041 : FAILURE in 4m 15s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/7967e3b61e224790addf4c83f2c5cc4a : SUCCESS in 3m 16s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/8b427a6100244b0c849a86bd35ba3872 : SUCCESS in 3m 22s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/4d293e213e2a49a9af95127aa7e77a8c : SUCCESS in 1h 21m 27s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/b82f7ac90822480e93e9a76829b48122 : SUCCESS in 1h 06m 59s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/e1343524f66f4064b3623c945b13e174 : SUCCESS in 32m 30s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/a2535c2034ff4fd68607259125a2c711 : SUCCESS in 30m 38s","accounts_in_message":[],"_revision_number":3},{"id":"5b3dab378c05218a0bddfd434b21a1affedca198","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-28 08:08:45.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"3e754736a3191ed098c694204036e159ac72830c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 09:33:46.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ac037d6d530549d2b1ad58c06233287e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62f7878b3e1549ee8370b52ba0b54a24 : SUCCESS in 6m 02s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/a9ab3ecb438c42bc8ad8db0a9bd70f24 : SUCCESS in 10m 17s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/70831d33ebb041a286fb2ab3298811a4 : SUCCESS in 3m 58s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/d001470dbef04ee585354e90eef14572 : SUCCESS in 3m 22s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/6e06798e71e147a5b024bc8bc736d8c3 : FAILURE in 58m 18s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/190c94da6b7642da81ad2783373a5b5c : SUCCESS in 1h 19m 45s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/7daedf9ef9df4b7b99006d2c7299bcd6 : SUCCESS in 53m 57s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/67afd8f1b3a54f9794aad926c43c7c03 : SUCCESS in 56m 32s","accounts_in_message":[],"_revision_number":4},{"id":"5066f7a06995a3fb0f51a8833e8b3cbbc5ddd9c0","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-28 09:39:58.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"82acd19ecf25da87983149f35aa554bd638cfd79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 11:20:57.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0d55829c55c34801955f583cbbf56d52\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9133ef1fd34f4d099149325f315d0553 : SUCCESS in 3m 24s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/3a5307680fce4f9580065fe2a964f31f : SUCCESS in 6m 52s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/a17b91db81484a159af0aae520134007 : SUCCESS in 4m 52s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/b2c94e98467246d8b9ae7d0364642fdf : SUCCESS in 7m 37s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/bd9691e1c0be4873a544092e197cf780 : SUCCESS in 1h 24m 58s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/08a2fd8b36f3476fab8854a4171ffe2a : SUCCESS in 52m 17s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/13ccdd798f6a46baba29843805abebff : FAILURE in 1h 13m 18s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/35834985702f4e739bb23197bd35f5a9 : SUCCESS in 49m 09s","accounts_in_message":[],"_revision_number":4},{"id":"5efa32450efbdab94573348ce13514e97706002f","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-28 11:26:55.000000000","message":"Patch Set 4:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":4},{"id":"342c4cb2cc2b805404088324f2fc23365229befd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-28 11:26:59.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"2c20e26f9b106e31021b7adc82c03be64e382a5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 12:58:45.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f2a9a86b9944220ad9d15c05055549e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5fb01d3974d415b90c5e22ff6f3daff : SUCCESS in 4m 01s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/32cbd1a8b34148058fdad61323edbb6a : SUCCESS in 4m 16s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/9928419ea232402fb843feb02604b9c4 : SUCCESS in 7m 25s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/0130a1e3f12c4a5485f2a78fb917c2bd : SUCCESS in 5m 05s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/ebea4565db1040eaac2d5e8454c00ebd : SUCCESS in 1h 25m 41s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/42cb0030b1484025b5ae22b9be9ad15b : SUCCESS in 1h 18m 24s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/a4679b625569415aa5f0d7214e459464 : SUCCESS in 30m 31s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/d95db835b856483b8220a875feebd19f : SUCCESS in 50m 57s","accounts_in_message":[],"_revision_number":5},{"id":"446391b07e179133beb4568b0d278739c08c2b84","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-01-29 01:47:52.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"16207e8f7aa65fefabc0bfe45c272c6f9b6c4259","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 01:48:22.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"0f6fa3e941fbc007ae742a84a546ffb0422b6949","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 02:52:48.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3abe6482fe144600bd907c1adac4afe3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa0c01fd01944eecaab4d2ba8cdb7aaa : SUCCESS in 4m 10s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/d76560e5710c451886e9f64888beec73 : SUCCESS in 10m 50s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/30223c7e81ad45e7b8b45186d3ffdbef : SUCCESS in 8m 50s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/facdc3e17224461ead0f845dedb87c37 : SUCCESS in 6m 51s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/389b7467c5e64d82821e11deef4da91d : FAILURE in 53m 57s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/e4aff77f32af4736a072d07d7e2adc83 : SUCCESS in 1h 02m 15s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/fa08461b1e6e41ceb25f3b93b8a96d00 : FAILURE in 55m 29s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/687f110421284b66ad28bf7b348fee24 : SUCCESS in 52m 47s","accounts_in_message":[],"_revision_number":5},{"id":"2058b9c9e21ff977f48abda1705bdcc7b2437f0f","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-29 10:00:31.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"15175d401f4582c11cf10411db211d6c325a4457","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 11:36:06.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8b85552953734ec1b1682b8f9598ca4a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a6ad5daa1e7470fb183936cd01aaff6 : SUCCESS in 6m 26s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/c9cc4b7d63f64e85a2a9652b71893d09 : SUCCESS in 12m 07s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/a6b88361c5424e92b8a068d65b756a27 : SUCCESS in 9m 18s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/4a2d979814264e04b1b7139f5954da88 : SUCCESS in 7m 51s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/0457983cb9ba482aad39aa16c762f78a : SUCCESS in 1h 30m 00s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/9f6e476d545c4acb98b1005970c83a82 : SUCCESS in 1h 13m 08s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/8c124561082a42099912b8ca95544571 : FAILURE in 45m 40s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/aefbf60933a147da8bfadbeef0b116ba : SUCCESS in 41m 16s","accounts_in_message":[],"_revision_number":5},{"id":"37fed23586c3c009d0ced285f7b28ef901f8b94d","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-29 14:35:56.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"2f2921109c930c343d4cdaceaf2b425e77f339d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 15:48:05.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15755749861a4c6e80dea7a453ee6618\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a9fedfff83a4260ac7b00dfb5db7b6a : SUCCESS in 4m 19s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/06143a1738ff4437bed734cb2d631241 : SUCCESS in 10m 31s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/1754a202fc9c4ec996dbec452852b533 : SUCCESS in 7m 23s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/f61faa9cda8542c2915f16f49647e6ac : SUCCESS in 7m 28s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/d1f77b164fc747b1ab5b44990c65c8c2 : SUCCESS in 51m 13s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/b302a553b27b40f7b24a87a3bce3ed69 : SUCCESS in 1h 06m 37s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/18c46886760f4577996dd6ca99b7c6d8 : SUCCESS in 30m 56s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/5602aab4ba9843709ba64927be380312 : SUCCESS in 48m 16s","accounts_in_message":[],"_revision_number":5},{"id":"0f5e894935143018ca94f8a6be96788000c1e22f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 15:48:36.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"a13c169a58c80ca5d046569d05c822646adcf187","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 17:32:48.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7079695597e1419884b08315f6b920db\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/926d1e28848a447cad105108605484fd : SUCCESS in 6m 32s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/4799fd0c35724b2d95674efbf0792ba8 : SUCCESS in 8m 16s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/7ae358aa4609411ba6545e8594ff1217 : SUCCESS in 4m 50s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/ea69035e87ed41fdaee3d1ca543859ab : SUCCESS in 6m 40s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/31e7fcc2193f44debf5641cec3532f02 : SUCCESS in 1h 25m 24s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/dbf64e00b8654761a422d10ef903cb5f : SUCCESS in 1h 08m 32s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/13900ca43b344be0ab7e99d864acf09b : SUCCESS in 52m 04s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/7290e7d489fc491eaa3870fd17724e2d : SUCCESS in 29m 57s","accounts_in_message":[],"_revision_number":5},{"id":"3ffd92f2fe57190950936afab17d1d768204ddb9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 17:32:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"e5e8c364fce06bc84aaefade7583752628e8a6d8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 17:33:39.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac824647d4444d1c837719ff00410e3d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc78ae612f724aa6bad50ece41a8942d : SUCCESS in 40s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"55452991b5deb5e533cc71eeed349fe6431e2f68","revisions":{"fb5426f2ec5b41550da31d1babc1bed2e88f2185":{"kind":"REWORK","_number":1,"created":"2026-01-27 21:31:58.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/55/974955/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/55/974955/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/974955/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/974955/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/974955/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/55/974955/1"}}},"commit":{"parents":[{"commit":"7217e3af676d1cd1586701410fc4b5f63bdb2bb3","subject":"Merge \"Remove no need spaces\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7217e3af676d1cd1586701410fc4b5f63bdb2bb3"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-27 21:31:58.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-27 21:31:58.000000000","tz":0},"subject":"Confirm form, \"confirm\" button update to OK | Part 3","message":"Confirm form, \"confirm\" button update to OK | Part 3\n\nPart 1 in I0c5b68000cd75d57d6e3dee8ab33a9da0be90f22\nPart 2 in Ib95b9de87e6a065ab03f98d69c26eede3d3a87c3\n\nChange-Id: I17dec2bab0e2ef99b82fa34589918279c5512084\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/fb5426f2ec5b41550da31d1babc1bed2e88f2185"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/fb5426f2ec5b41550da31d1babc1bed2e88f2185"}]},"branch":"refs/heads/master"},"91336d5df6028e10ea3111ca76367f4f21685181":{"kind":"REWORK","_number":2,"created":"2026-01-27 21:41:35.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/55/974955/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/55/974955/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/974955/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/974955/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/974955/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/55/974955/2"}}},"commit":{"parents":[{"commit":"7217e3af676d1cd1586701410fc4b5f63bdb2bb3","subject":"Merge \"Remove no need spaces\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7217e3af676d1cd1586701410fc4b5f63bdb2bb3"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-27 21:31:58.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-27 21:41:35.000000000","tz":0},"subject":"Confirm form, \"confirm\" button update to OK | Part 3","message":"Confirm form, \"confirm\" button update to OK | Part 3\n\nPart 1 in I0c5b68000cd75d57d6e3dee8ab33a9da0be90f22\nPart 2 in Ib95b9de87e6a065ab03f98d69c26eede3d3a87c3\n\nChange-Id: I17dec2bab0e2ef99b82fa34589918279c5512084\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/91336d5df6028e10ea3111ca76367f4f21685181"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/91336d5df6028e10ea3111ca76367f4f21685181"}]},"branch":"refs/heads/master"},"63d8562b987b4b83bbf3fbf007bb0b9740753d07":{"kind":"REWORK","_number":3,"created":"2026-01-27 22:44:24.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/55/974955/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/55/974955/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/974955/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/974955/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/974955/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/55/974955/3"}}},"commit":{"parents":[{"commit":"7217e3af676d1cd1586701410fc4b5f63bdb2bb3","subject":"Merge \"Remove no need spaces\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7217e3af676d1cd1586701410fc4b5f63bdb2bb3"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-27 21:31:58.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-27 22:44:24.000000000","tz":0},"subject":"Confirm form, \"confirm\" button update to OK | Part 3","message":"Confirm form, \"confirm\" button update to OK | Part 3\n\nPart 1 in I0c5b68000cd75d57d6e3dee8ab33a9da0be90f22\nPart 2 in Ib95b9de87e6a065ab03f98d69c26eede3d3a87c3\n\nChange-Id: I17dec2bab0e2ef99b82fa34589918279c5512084\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/63d8562b987b4b83bbf3fbf007bb0b9740753d07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/63d8562b987b4b83bbf3fbf007bb0b9740753d07"}]},"branch":"refs/heads/master"},"7b6b84bcc38b0c002584208533663350941435d8":{"kind":"REWORK","_number":4,"created":"2026-01-28 08:08:45.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/55/974955/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/55/974955/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/974955/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/974955/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/974955/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/55/974955/4"}}},"commit":{"parents":[{"commit":"7217e3af676d1cd1586701410fc4b5f63bdb2bb3","subject":"Merge \"Remove no need spaces\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7217e3af676d1cd1586701410fc4b5f63bdb2bb3"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-27 21:31:58.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-28 08:08:45.000000000","tz":0},"subject":"Confirm form, \"confirm\" button update to OK | Part 3","message":"Confirm form, \"confirm\" button update to OK | Part 3\n\nPart 1 in I0c5b68000cd75d57d6e3dee8ab33a9da0be90f22\nPart 2 in Ib95b9de87e6a065ab03f98d69c26eede3d3a87c3\n\nChange-Id: I17dec2bab0e2ef99b82fa34589918279c5512084\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7b6b84bcc38b0c002584208533663350941435d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7b6b84bcc38b0c002584208533663350941435d8"}]},"branch":"refs/heads/master"},"55452991b5deb5e533cc71eeed349fe6431e2f68":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-01-28 11:26:59.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/55/974955/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/55/974955/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/974955/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/974955/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/55/974955/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/55/974955/5"}}},"commit":{"parents":[{"commit":"a1bbbfcfe11d36113e274a0ecc1aa67d3328d488","subject":"Merge \"Confirm step display cleanup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a1bbbfcfe11d36113e274a0ecc1aa67d3328d488"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-27 21:31:58.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-28 11:26:59.000000000","tz":0},"subject":"Confirm form, \"confirm\" button update to OK | Part 3","message":"Confirm form, \"confirm\" button update to OK | Part 3\n\nPart 1 in I0c5b68000cd75d57d6e3dee8ab33a9da0be90f22\nPart 2 in Ib95b9de87e6a065ab03f98d69c26eede3d3a87c3\n\nChange-Id: I17dec2bab0e2ef99b82fa34589918279c5512084\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/55452991b5deb5e533cc71eeed349fe6431e2f68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/55452991b5deb5e533cc71eeed349fe6431e2f68"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
