)]}'
{"id":"openstack%2Fskyline-console~974958","triplet_id":"openstack%2Fskyline-console~master~I49dc86c106679bf4aff92bd6c71fad93447719e0","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-01-29 04:10:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I49dc86c106679bf4aff92bd6c71fad93447719e0","subject":"Fix issue for User ID","status":"MERGED","created":"2026-01-27 21:40:26.000000000","updated":"2026-01-29 04:12:11.000000000","submitted":"2026-01-29 04:10:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"974958","meta_rev_id":"51b51200182afe6d0ac81c97d861391ec8b234e7","_number":974958,"virtual_id_number":974958,"owner":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-29 04:10:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-01-29 02:01:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-01-29 02:01:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-27 23:12:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-29 02:01:55.000000000","updated_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"}],"messages":[{"id":"97c4b9b8e4631760243cffdc9fac8b8432a9aae7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-27 21:40:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aa4d056f7ea09b61141e8fe22109bcc13e81d450","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-27 21:45:28.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"44ae8292efcf3b04c208967eed341ef160ccdef5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-27 23:12:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d9f23a839bf4d65a5955c47ecf2af19\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15110dd77ec44bf3bd7d477eb3c672a9 : SUCCESS in 5m 12s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/ffe51cfb6511466a8e90ce89d1749379 : SUCCESS in 10m 54s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/86d51103ba6f48039047bbe2faa26482 : SUCCESS in 7m 40s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/aa088181f4ed4d428c4470a1bad8a982 : SUCCESS in 8m 01s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/7e16d54a8fb74c099745a7d41a8e1e33 : SUCCESS in 1h 22m 26s","accounts_in_message":[],"_revision_number":2},{"id":"aadd5b3a3687281679e1b7980fe4e8d9a6ad52e2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-28 08:32:18.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"73a1bf2e18695202fc440db34afda645b4272259","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 10:05:35.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c752f33b4ff4c38bed5bb28f015ca75\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ced3a035df3c4135bc6129cbe40a0bd5 : SUCCESS in 7m 10s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/480790ea6f71488897b74fd69ccca06d : SUCCESS in 5m 17s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/fabe37e55c37499b81e3a0e162034bec : SUCCESS in 7m 48s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/af2e290d98d74e618b1e3cdbc84fe5b0 : SUCCESS in 3m 16s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/398ae7a5ae8a4638aec14f77ff0d2c09 : SUCCESS in 1h 26m 55s","accounts_in_message":[],"_revision_number":3},{"id":"1d2e552d662db2b03a34403c82104660cd75f78b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-29 00:13:17.000000000","message":"Patch Set 4: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"3a92e7914978abf912035ad5736d1675ff3c2fa7","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-29 00:13:20.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":4},{"id":"d3d160b0ac9e32ed1e9f3804da3b048e3d52496e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-29 00:14:09.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"2697369dec2a464ab8d58b8e4e446350a2209fde","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 01:42:26.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a09b47f41b8f49fabf57180f3f268c94\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d27ba8c703b4d33ae401d143bd682c0 : SUCCESS in 6m 58s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/ce3c35fe405a4328bfea685145f37d9d : SUCCESS in 8m 14s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/d92be7a2b3ff405f834b2f5ea7085070 : SUCCESS in 6m 20s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/c43747bfe8b64a9c8638df5bd0e71214 : SUCCESS in 3m 15s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/1e04397d48b54456bd8e6fe574819e84 : SUCCESS in 1h 21m 42s","accounts_in_message":[],"_revision_number":5},{"id":"43e91097e8736fa1d8b9ecdd61a2f6b73ad89bed","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-01-29 02:01:55.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"edf34534c045b07a62adc5554c4605310fcb1d98","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 02:09:47.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"c4c338d98e555bb0e05b7990518392a6b298c68b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 04:10:19.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61e636793ef2455fb4ce877c73b40021\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70baf6e0965046bf9f8c565d1c7aceac : SUCCESS in 7m 16s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/c6b49179f3a94d9099f99071e82b4119 : SUCCESS in 10m 41s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/8866d573188a4299aa27502ad3305ddd : SUCCESS in 8m 08s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/bf9b810ee06e40419ecba3a6a5629da6 : SUCCESS in 7m 52s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/930b607fd3c84585872ae54b48476971 : SUCCESS in 1h 19m 36s","accounts_in_message":[],"_revision_number":5},{"id":"eb1a51718d2d3debfe930a2083f8ba9918bc4923","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 04:10:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"51b51200182afe6d0ac81c97d861391ec8b234e7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 04:12:11.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/614604872b4845e596b6d8f6a649de3c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17b95f1093a545d8849f480189a8e1b8 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"a0bd667a996f866484e7a68c594ed943cb8428dc","revisions":{"350a7840f41bcdf3703b6770a6bf0fa4e3afd50c":{"kind":"REWORK","_number":1,"created":"2026-01-27 21:40:26.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/58/974958/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/58/974958/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/58/974958/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/58/974958/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/58/974958/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/58/974958/1"}}},"commit":{"parents":[{"commit":"7217e3af676d1cd1586701410fc4b5f63bdb2bb3","subject":"Merge \"Remove no need spaces\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7217e3af676d1cd1586701410fc4b5f63bdb2bb3"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-27 21:40:26.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-27 21:40:26.000000000","tz":0},"subject":"WIP","message":"WIP\n\nChange-Id: I49dc86c106679bf4aff92bd6c71fad93447719e0\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/350a7840f41bcdf3703b6770a6bf0fa4e3afd50c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/350a7840f41bcdf3703b6770a6bf0fa4e3afd50c"}]},"branch":"refs/heads/master"},"786565b254e8d1d95b9d29e5813431238deec469":{"kind":"REWORK","_number":2,"created":"2026-01-27 21:45:28.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/58/974958/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/58/974958/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/58/974958/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/58/974958/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/58/974958/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/58/974958/2"}}},"commit":{"parents":[{"commit":"7217e3af676d1cd1586701410fc4b5f63bdb2bb3","subject":"Merge \"Remove no need spaces\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7217e3af676d1cd1586701410fc4b5f63bdb2bb3"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-27 21:40:26.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-27 21:44:26.000000000","tz":0},"subject":"WIP","message":"WIP\n\nChange-Id: I49dc86c106679bf4aff92bd6c71fad93447719e0\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/786565b254e8d1d95b9d29e5813431238deec469"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/786565b254e8d1d95b9d29e5813431238deec469"}]},"branch":"refs/heads/master"},"8104effe6fa807f308689fe68bbf68cb031317cd":{"kind":"REWORK","_number":3,"created":"2026-01-28 08:32:18.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/58/974958/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/58/974958/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/58/974958/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/58/974958/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/58/974958/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/58/974958/3"}}},"commit":{"parents":[{"commit":"7217e3af676d1cd1586701410fc4b5f63bdb2bb3","subject":"Merge \"Remove no need spaces\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7217e3af676d1cd1586701410fc4b5f63bdb2bb3"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-27 21:40:26.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-28 08:32:11.000000000","tz":0},"subject":"WIP","message":"WIP\n\nChange-Id: I49dc86c106679bf4aff92bd6c71fad93447719e0\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/8104effe6fa807f308689fe68bbf68cb031317cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/8104effe6fa807f308689fe68bbf68cb031317cd"}]},"branch":"refs/heads/master"},"04bf125fce886c4246ade4c26a926e98aeea7d32":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2026-01-29 00:13:17.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/58/974958/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/58/974958/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/58/974958/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/58/974958/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/58/974958/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/58/974958/4"}}},"commit":{"parents":[{"commit":"7217e3af676d1cd1586701410fc4b5f63bdb2bb3","subject":"Merge \"Remove no need spaces\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7217e3af676d1cd1586701410fc4b5f63bdb2bb3"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-27 21:40:26.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-29 00:13:17.000000000","tz":0},"subject":"Fix issue for User ID","message":"Fix issue for User ID\n\nFix an issue where the User ID column displayed incorrect\ndata ([object Object]) instead of the actual user identifier.\n\nChange-Id: I49dc86c106679bf4aff92bd6c71fad93447719e0\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/04bf125fce886c4246ade4c26a926e98aeea7d32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/04bf125fce886c4246ade4c26a926e98aeea7d32"}]},"branch":"refs/heads/master","description":"Edit commit message"},"a0bd667a996f866484e7a68c594ed943cb8428dc":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2026-01-29 00:14:09.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/58/974958/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/58/974958/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/58/974958/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/58/974958/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/58/974958/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/58/974958/5"}}},"commit":{"parents":[{"commit":"7217e3af676d1cd1586701410fc4b5f63bdb2bb3","subject":"Merge \"Remove no need spaces\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7217e3af676d1cd1586701410fc4b5f63bdb2bb3"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-27 21:40:26.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-29 00:14:09.000000000","tz":0},"subject":"Fix issue for User ID","message":"Fix issue for User ID\n\nFix an issue where the User ID in Action logs column displayed\nincorrect data ([object Object]) instead of the actual\nuser identifier and add to translation.\n\nChange-Id: I49dc86c106679bf4aff92bd6c71fad93447719e0\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a0bd667a996f866484e7a68c594ed943cb8428dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a0bd667a996f866484e7a68c594ed943cb8428dc"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
