)]}'
{"id":"openstack%2Fskyline-console~974970","triplet_id":"openstack%2Fskyline-console~master~I3f5df0f588a6df13689c4f33aa24d14a114e4d21","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-01-28 05:14:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3f5df0f588a6df13689c4f33aa24d14a114e4d21","subject":"Correct Floating IP release confirmation text typo","status":"MERGED","created":"2026-01-27 23:13:42.000000000","updated":"2026-01-28 05:17:21.000000000","submitted":"2026-01-28 05:14:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"974970","meta_rev_id":"149a6552d5915bdeb399437397161b607cdc0444","_number":974970,"virtual_id_number":974970,"owner":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-28 05:14:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-01-28 03:36:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-01-28 03:36:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-28 00:20:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-28 03:36:25.000000000","updated_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"}],"messages":[{"id":"a618ae0d19283a958362e93699cda85c55db58ed","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-27 23:13:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3070e07580613b8a7c650e487d80e020ff797933","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-27 23:14:54.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"b0380451fdc9bc7efdaaf659b9f13f805f8111b8","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-27 23:14:56.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"6d4708406c373ad7b58926f9b330094a7a9fad7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 00:20:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7eb4d7e4c0b46318458034fd2c95175\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa54cbb1e4334cb2ba04a5e526651206 : SUCCESS in 5m 13s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/57ada53decc347cf8a3795d81a47c47e : SUCCESS in 4m 09s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/68915d2961db41b39cdada06d68bfa63 : SUCCESS in 7m 28s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/1433b0ab08df446589c109e45ceb16aa : SUCCESS in 7m 32s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/bbe22d7416874b929a67e0ee9ae4b4f2 : SUCCESS in 1h 01m 04s","accounts_in_message":[],"_revision_number":2},{"id":"8338df9ae91ae98aa67b9902d941edb432e8d106","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-01-28 03:36:25.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7dd1f0c7fd3c9c421ae6595a5a33497358a0d4ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 03:37:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a860fe910f54bb322d97897a5ca90c89f1de5fe3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 05:14:41.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e4dcd42035a4e57bc93d3a1e54614b4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/372060ee1d0546e5a9b327676d9c5235 : SUCCESS in 3m 32s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/dba6326c3191460a9546b31cf0e2fae8 : SUCCESS in 4m 23s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/f55987341a194d97ade9d823861ff7ac : SUCCESS in 4m 07s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/91f482cf83c748529fb5d4132db7beca : SUCCESS in 3m 35s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/003cf98ebdfd49468ab363ccef9ed3d2 : SUCCESS in 50m 50s","accounts_in_message":[],"_revision_number":2},{"id":"10d58cc1dd384cfae2afd719455e8ec9f0cb9251","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 05:14:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"149a6552d5915bdeb399437397161b607cdc0444","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 05:17:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60c0b051708c4658a2c9ba577f5ce7dc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc18cb66d51949c1985fd085b26d37f4 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"82c45d83258110ec7bff5992205c0233bebe3973","revisions":{"c187888baacf81ab99dac74b0c0da485c3ce245f":{"kind":"REWORK","_number":1,"created":"2026-01-27 23:13:42.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/70/974970/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/70/974970/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/70/974970/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/70/974970/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/70/974970/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/70/974970/1"}}},"commit":{"parents":[{"commit":"7217e3af676d1cd1586701410fc4b5f63bdb2bb3","subject":"Merge \"Remove no need spaces\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7217e3af676d1cd1586701410fc4b5f63bdb2bb3"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-27 23:13:42.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-27 23:13:42.000000000","tz":0},"subject":"Correct Floating IP release confirmation text typo","message":"Correct Floating IP release confirmation text typo\n\nRemove extra brace from the IP release\nconfirmation text\n\nChange-Id: I3f5df0f588a6df13689c4f33aa24d14a114e4d21\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/c187888baacf81ab99dac74b0c0da485c3ce245f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/c187888baacf81ab99dac74b0c0da485c3ce245f"}]},"branch":"refs/heads/master"},"82c45d83258110ec7bff5992205c0233bebe3973":{"kind":"REWORK","_number":2,"created":"2026-01-27 23:14:54.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/70/974970/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/70/974970/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/70/974970/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/70/974970/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/70/974970/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/70/974970/2"}}},"commit":{"parents":[{"commit":"7217e3af676d1cd1586701410fc4b5f63bdb2bb3","subject":"Merge \"Remove no need spaces\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/7217e3af676d1cd1586701410fc4b5f63bdb2bb3"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-27 23:13:42.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-27 23:14:53.000000000","tz":0},"subject":"Correct Floating IP release confirmation text typo","message":"Correct Floating IP release confirmation text typo\n\nRemove extra brace from the IP release\nconfirmation text\n\nChange-Id: I3f5df0f588a6df13689c4f33aa24d14a114e4d21\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/82c45d83258110ec7bff5992205c0233bebe3973"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/82c45d83258110ec7bff5992205c0233bebe3973"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
