)]}'
{"id":"openstack%2Fskyline-console~975004","triplet_id":"openstack%2Fskyline-console~master~Icbe486e2ad3c0057c990db903e854a4d76c34322","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-01-29 10:00:18.000000000","reason":"\u003cGERRIT_ACCOUNT_37632\u003e replied on the change","reason_account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-01-29 01:32:28.000000000","reason":"removed on reply"},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-01-29 11:54:55.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icbe486e2ad3c0057c990db903e854a4d76c34322","subject":"fix for Endpoint writing","status":"MERGED","created":"2026-01-28 13:13:22.000000000","updated":"2026-01-29 11:56:03.000000000","submitted":"2026-01-29 11:54:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"975004","meta_rev_id":"53e18ffd90c909909d9b83ed6747ac8832820252","_number":975004,"virtual_id_number":975004,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-29 11:54:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-01-29 01:32:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-01-29 01:32:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-28 13:16:25.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"},{"updated":"2026-01-28 13:16:25.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2026-01-28 14:19:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4b0777264b1080cb97a9ab26c652959c32bee5ac","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-28 13:13:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4d5cb047c70f7a781906a24fc1a9c6741be19d24","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-28 13:16:08.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"193f911f112386395abad8ea733474d52d2a3201","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-28 13:16:18.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"f2bb92101e1b7ef7d25abc992aca9849cdf4d437","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-28 13:16:25.000000000","message":"Patch Set 3:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"1c6032a01558604318797a174b588096dd1eee7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 14:19:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ef0b82507be4b519dfd877f8c8ddcba\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2bf0f1e6510b48b2acd38cc249774cea : SUCCESS in 4m 11s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/b779d1d86b2e464c811ba77dcb7b3068 : SUCCESS in 4m 55s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/d45e1e69fe6c493c959eb327abc81ddc : SUCCESS in 7m 56s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/540de0a6b91f475abf8a5f58859da9f8 : SUCCESS in 3m 16s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/374013ea941141bfa8419695b647ec86 : SUCCESS in 1h 00m 35s","accounts_in_message":[],"_revision_number":3},{"id":"e8358e09e6ee82a21b3e8bcfc8fb40de151443c8","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-01-29 01:32:28.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"9cc6c8745c6b3a3fd385dffff55e5ee6854c8199","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-01-29 01:32:35.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ea20287be78bf93e370245ffbb74971a8d2b5fcf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 01:32:58.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5096db8e47680b6176cad287e1564326d7668d56","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 02:04:45.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/26d746e948064f438d2b0d246ea872f6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97897da4ec4a4a10a1c911ac236a62bc : SUCCESS in 5m 21s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/2c1ddbc8fa6e45b982a54972a0752375 : FAILURE in 3m 23s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/477fb46120a64efa9f5125799d8d28d5 : SUCCESS in 5m 03s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/49171119eabe4065947916cb770847a4 : SUCCESS in 7m 37s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/f566838b65364dd4b091dd040d07f84b : SUCCESS in 30m 27s","accounts_in_message":[],"_revision_number":3},{"id":"7d4f96b1561aff61eee403323f8b9edcab826753","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-29 10:00:18.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"da5131794178ab64e0808f72ef0a56a5a67d0ea4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 10:33:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53f48f83a5b94f16b70a3ec3bc76a2ab\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/303c3ad48f274cad9d1df32c761666f7 : SUCCESS in 6m 18s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/fedd2fd5f5674a1fa7451a735f80ad4e : SUCCESS in 6m 50s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/e694545641a540b3bb978ed86051234f : SUCCESS in 7m 21s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/dd35cea63e644320b07c4496b9764d85 : SUCCESS in 8m 54s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/a2b8222b95af4b2c9a77ab4f751640ea : SUCCESS in 30m 31s","accounts_in_message":[],"_revision_number":3},{"id":"39c8e9637701ab537415ac34c856399830c5a3f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 10:33:46.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"43321e25e867eda004d893b7349527c841620c3e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 11:54:55.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c6d7f4d40a864726976c5824f6b0cb7a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bebd57d5af1643d6a8c03ea7cefd2075 : SUCCESS in 6m 47s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/e8d50154dd6c4f67a92fc64890b09f85 : SUCCESS in 4m 35s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/3630ce7210044f9cbf039713d359cdd5 : SUCCESS in 6m 43s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/e897b89b16cb4339be6e2158404e7e09 : SUCCESS in 7m 21s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/a3223430b4114de8a9d13adf90eca24d : SUCCESS in 1h 00m 50s","accounts_in_message":[],"_revision_number":3},{"id":"81d8c9e4eb9fcb697abdb48a0b70370d72f6bc87","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 11:54:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"53e18ffd90c909909d9b83ed6747ac8832820252","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 11:56:03.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56d90c16738a4ee2b1e18e7d9bcffbb8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26ef17bf29e64abc87b8bd83154da69c : SUCCESS in 47s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d0f6049e40da18ae78466c4252f3ee3ba69b2ac5","revisions":{"0b49f857eed6d38a9d8b67e58be3d396a0ede436":{"kind":"REWORK","_number":1,"created":"2026-01-28 13:13:22.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/04/975004/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/04/975004/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/04/975004/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/04/975004/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/04/975004/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/04/975004/1"}}},"commit":{"parents":[{"commit":"2bb504d704ff0cdc8e850dbc83080bca0fccda1c","subject":"Merge \"fix for Message Reason\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/2bb504d704ff0cdc8e850dbc83080bca0fccda1c"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-28 13:13:22.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-28 13:13:22.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: Icbe486e2ad3c0057c990db903e854a4d76c34322\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/0b49f857eed6d38a9d8b67e58be3d396a0ede436"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/0b49f857eed6d38a9d8b67e58be3d396a0ede436"}]},"branch":"refs/heads/master"},"6cf5e45ca56330b25601ca48b549e6666c0fef96":{"kind":"REWORK","_number":2,"created":"2026-01-28 13:16:08.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/04/975004/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/04/975004/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/04/975004/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/04/975004/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/04/975004/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/04/975004/2"}}},"commit":{"parents":[{"commit":"2bb504d704ff0cdc8e850dbc83080bca0fccda1c","subject":"Merge \"fix for Message Reason\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/2bb504d704ff0cdc8e850dbc83080bca0fccda1c"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-28 13:13:22.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-28 13:16:08.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: Icbe486e2ad3c0057c990db903e854a4d76c34322\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6cf5e45ca56330b25601ca48b549e6666c0fef96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6cf5e45ca56330b25601ca48b549e6666c0fef96"}]},"branch":"refs/heads/master"},"d0f6049e40da18ae78466c4252f3ee3ba69b2ac5":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-01-28 13:16:18.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/04/975004/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/04/975004/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/04/975004/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/04/975004/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/04/975004/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/04/975004/3"}}},"commit":{"parents":[{"commit":"2bb504d704ff0cdc8e850dbc83080bca0fccda1c","subject":"Merge \"fix for Message Reason\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/2bb504d704ff0cdc8e850dbc83080bca0fccda1c"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-28 13:13:22.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-01-28 13:16:18.000000000","tz":0},"subject":"fix for Endpoint writing","message":"fix for Endpoint writing\n\nChange-Id: Icbe486e2ad3c0057c990db903e854a4d76c34322\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d0f6049e40da18ae78466c4252f3ee3ba69b2ac5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d0f6049e40da18ae78466c4252f3ee3ba69b2ac5"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
