)]}'
{"id":"openstack%2Fskyline-console~975099","triplet_id":"openstack%2Fskyline-console~master~I475b340d8c5951a9c8852f40de7f0e46096d42b9","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-01-29 04:20:45.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I475b340d8c5951a9c8852f40de7f0e46096d42b9","subject":"Improve UX for Manual IP Assignment Field","status":"MERGED","created":"2026-01-28 21:34:43.000000000","updated":"2026-01-29 04:21:37.000000000","submitted":"2026-01-29 04:20:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"975099","meta_rev_id":"57437dc44c70110df60d460b9776047f9a4d10b2","_number":975099,"virtual_id_number":975099,"owner":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-29 04:20:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-01-29 02:03:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-01-29 02:03:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-28 23:04:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-29 02:03:23.000000000","updated_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"}],"messages":[{"id":"08146ff3485158d588175a540c4f9c96f94964e6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-28 21:34:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"665f06317eba79f48650916526212fb4d3f2f61e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-28 21:39:31.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"1b81001ec524f65fe8b577e7d299799cb09297a0","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-28 21:39:40.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"fd769448625ad27e7eb616a977a4ec851652ead5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-28 23:04:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2521fce9b88440049742c2973d17fb12\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd78140745b34783b3f27dc0881ed785 : SUCCESS in 5m 26s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/b720cc8f75084b56a8cbea5073230df6 : SUCCESS in 11m 04s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/e290cac0f69e4f91a2a25dab92e1de2b : SUCCESS in 7m 43s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/9a51825c756040ad9ea1bf00cdeeae3e : SUCCESS in 7m 37s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/ceb61456c90a4072b9f52dd2262d6082 : SUCCESS in 1h 23m 58s","accounts_in_message":[],"_revision_number":2},{"id":"52f548d8d3d5b76f81c7122e9162b0dd55e05d62","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-01-29 02:03:23.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0b527c76753a0bc8d2b14a5cdf3c2b60132eca06","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 02:15:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d4193bc8817e4302a8d0c9afa039218be9c5fb79","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 04:20:45.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c30680a3931f4c528324c1927590144a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5efadc13c36443978bc09c4a42200280 : SUCCESS in 8m 10s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/e66796fe807b4f43bf8fd80cb15f8836 : SUCCESS in 12m 30s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/32744d5e14064a6bb76b8488f442d9ce : SUCCESS in 7m 12s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/ead9e62f3dd0433cb8275b5af64a7356 : SUCCESS in 4m 30s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/53a68837df114225b48c8b7fb90eefb3 : SUCCESS in 1h 22m 54s","accounts_in_message":[],"_revision_number":2},{"id":"cbef77fadc1d410a7d5982d20b3bded2a53821d3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 04:20:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"57437dc44c70110df60d460b9776047f9a4d10b2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 04:21:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9be22b5499f74763af93f9a1387a0d21\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba2c4de2aaef4add9364e66afca81822 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ea8a8e1a953968667e539189359eb7c3b6447456","revisions":{"05e9c9471e1b5fc747e316034d65f4c42a4b718f":{"kind":"REWORK","_number":1,"created":"2026-01-28 21:34:43.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/99/975099/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/99/975099/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/99/975099/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/99/975099/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/99/975099/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/99/975099/1"}}},"commit":{"parents":[{"commit":"a88f8b6f908917ad9188c377f12beecab57da46d","subject":"Merge \"feat: enhance Object Storage UX with public URL access, clearer actions, and improved error messages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a88f8b6f908917ad9188c377f12beecab57da46d"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-28 21:34:43.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-28 21:34:43.000000000","tz":0},"subject":"Improve UX for Manual IP Assignment Field","message":"Improve UX for Manual IP Assignment Field\n\nSimplifythe interface attachment form by\nrenaming the field from “Given IP” to “IP Address”\n\nChange-Id: I475b340d8c5951a9c8852f40de7f0e46096d42b9\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/05e9c9471e1b5fc747e316034d65f4c42a4b718f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/05e9c9471e1b5fc747e316034d65f4c42a4b718f"}]},"branch":"refs/heads/master"},"ea8a8e1a953968667e539189359eb7c3b6447456":{"kind":"REWORK","_number":2,"created":"2026-01-28 21:39:31.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/99/975099/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/99/975099/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/99/975099/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/99/975099/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/99/975099/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/99/975099/2"}}},"commit":{"parents":[{"commit":"a88f8b6f908917ad9188c377f12beecab57da46d","subject":"Merge \"feat: enhance Object Storage UX with public URL access, clearer actions, and improved error messages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a88f8b6f908917ad9188c377f12beecab57da46d"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-28 21:34:43.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-28 21:39:28.000000000","tz":0},"subject":"Improve UX for Manual IP Assignment Field","message":"Improve UX for Manual IP Assignment Field\n\nSimplifythe interface attachment form by\nrenaming the field from “Given IP” to “IP Address”\n\nChange-Id: I475b340d8c5951a9c8852f40de7f0e46096d42b9\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ea8a8e1a953968667e539189359eb7c3b6447456"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/ea8a8e1a953968667e539189359eb7c3b6447456"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
