)]}'
{"id":"openstack%2Fskyline-console~975131","triplet_id":"openstack%2Fskyline-console~master~Icdd7871c433f762d681ffae4f1889a8ca0e8e2bd","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-01-30 14:41:58.000000000","reason":"\u003cGERRIT_ACCOUNT_37632\u003e replied on the change","reason_account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-01-30 03:08:58.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}},"hashtags":[],"change_id":"Icdd7871c433f762d681ffae4f1889a8ca0e8e2bd","subject":"Rename QEMU Guest Agent Label","status":"MERGED","created":"2026-01-29 10:32:35.000000000","updated":"2026-01-30 17:01:57.000000000","submitted":"2026-01-30 17:01:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"975131","meta_rev_id":"d608d5851a98e17ea0f87cab711f956a4edb7b58","_number":975131,"virtual_id_number":975131,"owner":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-30 17:01:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-01-30 03:08:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-01-29 14:36:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-01-30 03:08:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-29 12:15:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-29 14:36:43.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-01-29 14:36:43.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"}],"messages":[{"id":"f69de1040000c0693fce5f31579e11d923c82356","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-29 10:32:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9b673bd61a60b4aa26d2a161e20870941409112d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-29 10:34:28.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"f255891dc6d095135bd2e92447e96100bdfb4029","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-29 10:34:32.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"0fa7cca98de381c23942d90155c307e5cff4c216","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 12:15:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6ca030cf25f47bbabd73b46cd127fb7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d1bd1701dc54ef3a4ea4f8c4b1cb073 : SUCCESS in 4m 12s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/b4d2103dd3b545d49a4d7934117a3f3c : SUCCESS in 4m 20s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/8edb72aec8ab446795d161cc2f707ac1 : SUCCESS in 10m 27s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/4f806028ee8c4fb3ac1cbea510cf04ee : SUCCESS in 7m 23s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/7e7779cd83be4672be3f5434cd89e996 : SUCCESS in 1h 39m 28s","accounts_in_message":[],"_revision_number":2},{"id":"cd8624431e94c56976077ca5f3b8c8e48cd0f440","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-29 14:36:43.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"e1869494a594d0ab7d3b55ff3fd187be7f9482b4","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-01-30 03:08:58.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2ffee580d5f0e75f2f8075d110c1d9c890687efd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-30 03:09:21.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"07c2594994aa8f77df838a3d6f577002ec2b97c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-30 04:20:06.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e8070f4ae8cf48d7bdef40d990be8817\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98938521171847e8b5a1d4a861329746 : SUCCESS in 5m 55s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/74a6a562cb1341828f259e09a5eb5068 : SUCCESS in 7m 23s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/dc2ffbc04b264ead8595808ad3dada86 : SUCCESS in 7m 15s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/8f32d1c410874254bca01394e713c691 : SUCCESS in 7m 04s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/6a377ccd4a1e41e08f62198b18eafccf : FAILURE in 1h 08m 33s","accounts_in_message":[],"_revision_number":2},{"id":"72f690fa1397b292d5d9e9d26936d033b4fad9b5","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-30 07:47:03.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0a8e7028519ea5d7f91b1ddb197cd284e23da63e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-30 09:23:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48f677396d2e495986da19b3af82cab6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5f7c01061164f53b24d59c14c2be187 : SUCCESS in 6m 05s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/1ad8948036d24f039af4987a442b8b8f : SUCCESS in 11m 14s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/bd59a04508744696aa8322bb10ef1ea0 : SUCCESS in 3m 38s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/43fc59be508446f1aceff0d4cf85e3fc : SUCCESS in 9m 54s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/17f7e05f56a24e378324dfb74bd80830 : SUCCESS in 1h 31m 00s","accounts_in_message":[],"_revision_number":2},{"id":"38609f8698609a8a1304250ce596274164f56ae3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-30 09:23:37.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3e75119456dd14222aee0090af97a6b810e49e13","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-30 10:51:43.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4638cc0ff88c4df4855ece8c9087377a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1487469bf0b241b099ecd433bcbd69a2 : SUCCESS in 4m 37s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/080f3879355343bda725b0e75eda550a : SUCCESS in 9m 00s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/be5cf432c6c24abb81417a60e062a17f : SUCCESS in 4m 17s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/bbbc97041b7c4d1ebe76da23b3fc7c9f : SUCCESS in 4m 22s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/bfabf23ff4cf4a98a823f91982539064 : FAILURE in 1h 26m 44s","accounts_in_message":[],"_revision_number":2},{"id":"5c575641b78995350e850b6958c7a1de02142ac9","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-30 11:27:55.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d7898143dd4e5a9d6f32d5fd660289f7256a28d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-30 13:00:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ad608a63c5b4d3a9c1a54d0338c3668\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b2c2654d6c747a4bf61db0c4e241b89 : SUCCESS in 6m 35s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/79f0dff84ca2432dacece70108c615fc : SUCCESS in 11m 34s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/571361861ad944febef2355b557afbcd : SUCCESS in 9m 33s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/44f41339b06d4d83a7ebdc661cc6d0f1 : SUCCESS in 8m 11s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/bcee3d872c724859ad02989fac26c8ec : SUCCESS in 1h 26m 42s","accounts_in_message":[],"_revision_number":2},{"id":"1791be0b4b17253eb6005703564d36e3fc07928c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-30 13:00:22.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7c38cc8ae45fe6bfe870b415caea98b2bc4e34c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-30 13:52:09.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fa75b2f211a04ef9b7467d0b6d7592f9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d90cd44e9e9f493c87405552529b87fe : SUCCESS in 6m 21s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/7d3b85497e85499596c95648aa654ffc : SUCCESS in 4m 47s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/130531f52a2b4a8baff9b7fd9d571a77 : FAILURE in 1m 48s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/82593d2c100545168e59cc8537d35c32 : SUCCESS in 3m 58s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/0134aeebd5d0418098ea1e58815b0aa2 : SUCCESS in 51m 21s","accounts_in_message":[],"_revision_number":2},{"id":"c869d6532d943cfd5ec357b48d904ed39bea652f","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-30 14:41:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"aef72e325f16bf5a4de55f6ccb8cfad62ecb4748","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-30 15:56:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17ad85a796c646c9a05504f5e688774e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/677e215f333943998cbc3b2f01e2e54f : SUCCESS in 6m 37s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/39458a63f5d1497c8857b4d9dafc2cf3 : SUCCESS in 7m 36s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/ebaf3e882f7d4c279b0df1c95466f6b2 : SUCCESS in 3m 56s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/91610123d7b043dc835422cd207bb888 : SUCCESS in 4m 45s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/c2c044d8c6c34293a9afdfe4815e012c : SUCCESS in 1h 11m 42s","accounts_in_message":[],"_revision_number":2},{"id":"f9554000778fbc4c7fdded648f434e0069c9b4a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-30 15:57:16.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ad2c53177f69ffcaf83e820939a76cf916e140fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-30 17:01:05.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e11cf98f7d474dc6b3944ad5ce131116\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a67b63772849487e968d8131bed88a59 : SUCCESS in 6m 36s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/29f698ccb2764480a6c90c04d92f4700 : SUCCESS in 8m 14s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/670c218003e04aa0882beec0f9483d9a : SUCCESS in 3m 20s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/231edb37f31e4abfb186d46b08156c58 : SUCCESS in 10m 11s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/8799d809cb414bf8bee93bd1fc28da89 : SUCCESS in 1h 03m 28s","accounts_in_message":[],"_revision_number":2},{"id":"eb8044b615fe67e7f132da2aa0c1a33efe125f91","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-30 17:01:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"d608d5851a98e17ea0f87cab711f956a4edb7b58","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-30 17:01:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd6725feb8a64287a8cdd3419482ef87\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8332052aeceb4dfc8a26a3553681f072 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"539a823ec609fa1c8a75f208e7717870f91bcc20","revisions":{"a77911752682160528c134b208fd7c52e36370b8":{"kind":"REWORK","_number":1,"created":"2026-01-29 10:32:35.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/31/975131/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/31/975131/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/31/975131/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/31/975131/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/31/975131/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/31/975131/1"}}},"commit":{"parents":[{"commit":"3e850f5d724decf053cc15f05f7c76043600a8ea","subject":"Merge \"Improve UX for Manual IP Assignment Field\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/3e850f5d724decf053cc15f05f7c76043600a8ea"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-29 10:32:35.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-29 10:32:35.000000000","tz":0},"subject":"Rename QEMU Guest Agent Label","message":"Rename QEMU Guest Agent Label\n\nReplace the technical field with a human-readable\n label to improve UI clarity\n\nChange-Id: Icdd7871c433f762d681ffae4f1889a8ca0e8e2bd\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a77911752682160528c134b208fd7c52e36370b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a77911752682160528c134b208fd7c52e36370b8"}]},"branch":"refs/heads/master"},"539a823ec609fa1c8a75f208e7717870f91bcc20":{"kind":"REWORK","_number":2,"created":"2026-01-29 10:34:28.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/31/975131/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/31/975131/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/31/975131/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/31/975131/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/31/975131/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/31/975131/2"}}},"commit":{"parents":[{"commit":"3e850f5d724decf053cc15f05f7c76043600a8ea","subject":"Merge \"Improve UX for Manual IP Assignment Field\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/3e850f5d724decf053cc15f05f7c76043600a8ea"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-29 10:32:35.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-29 10:34:25.000000000","tz":0},"subject":"Rename QEMU Guest Agent Label","message":"Rename QEMU Guest Agent Label\n\nReplace the technical field with a human-readable\n label to improve UI clarity\n\nChange-Id: Icdd7871c433f762d681ffae4f1889a8ca0e8e2bd\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/539a823ec609fa1c8a75f208e7717870f91bcc20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/539a823ec609fa1c8a75f208e7717870f91bcc20"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
