)]}'
{"id":"openstack%2Fskyline-console~975156","triplet_id":"openstack%2Fskyline-console~master~Iad8a01baf19e6be75cd3826a89f276905a7d30e5","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-01-30 05:45:00.000000000","reason":"Change was submitted"},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-01-30 03:33:28.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}},"hashtags":[],"change_id":"Iad8a01baf19e6be75cd3826a89f276905a7d30e5","subject":"Fix duplicated affinity policies","status":"MERGED","created":"2026-01-29 13:51:15.000000000","updated":"2026-01-30 05:45:49.000000000","submitted":"2026-01-30 05:45:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"975156","meta_rev_id":"5a1a0ac7e3559b2fa2c688c01980a3ceaae32ff2","_number":975156,"virtual_id_number":975156,"owner":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-30 05:45:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-01-30 03:33:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-01-29 14:37:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-01-30 03:33:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-29 14:37:01.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-01-29 14:37:01.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2026-01-29 14:45:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9867fda1c94755fb75a77160322e64f589c916ad","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-29 13:51:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a220ce92d35135754290e50afc3c889d6c6cf589","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-29 13:52:13.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"2c6b9b22522d6d293a562429bbefa339b80af563","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-29 13:52:15.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"bebcd10874c90e4ed6ce22185ffcef087976d369","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-29 14:37:01.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"07ab7e006d5f60241baac519b9244db7934c087a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 14:45:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3d9373c48354c839c094843064c560b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bdfd34b0c2a42758595c5605426f3bb : SUCCESS in 4m 02s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/037eacb6f32a45e2ae3ca02691b4ae1d : SUCCESS in 12m 17s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/2b644e90dc814c87856b018f261d8e4a : SUCCESS in 3m 18s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/fa963297e5cb49b5b3a98777cff682b1 : SUCCESS in 3m 20s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/9fbec0f293c1494490f0bdd381cc8123 : SUCCESS in 52m 57s","accounts_in_message":[],"_revision_number":2},{"id":"2d06b0c9f51c9f8b3f5bb9217ab038a7ad7b7007","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-01-30 03:33:28.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ad395256faf2db7f5c7a4249635215f5b6de7286","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-30 03:33:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1f4fb9de13b305895cb2ea57d0812984d6cc9189","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-30 05:45:00.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58b20177180b43c19acf019838f8a942\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3b983d8285642a181b933cf9149aafa : SUCCESS in 6m 04s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/5c6c3b429e394efd8a0555c134a552f0 : SUCCESS in 8m 48s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/0e5031a515dd424eb278a966a6317a90 : SUCCESS in 7m 40s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/3ea52b011b394edd982e19beafc500db : SUCCESS in 3m 20s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/ad7a7194abc2466da6b30f15063ad457 : SUCCESS in 1h 22m 31s","accounts_in_message":[],"_revision_number":2},{"id":"b58e704df041d724d349a7e951160173008292e0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-30 05:45:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"5a1a0ac7e3559b2fa2c688c01980a3ceaae32ff2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-30 05:45:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5766e53a2a148bebc640f5a2aa8da2e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9f07e3303134bbb9a73e61d4866f3df : SUCCESS in 41s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1902dd1fd336afff2548c0f373ff47fde8ba4fed","revisions":{"18f07dc6cb3dbc0178f02d42d49028a09ef8d43d":{"kind":"REWORK","_number":1,"created":"2026-01-29 13:51:15.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/56/975156/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/56/975156/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/56/975156/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/56/975156/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/56/975156/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/56/975156/1"}}},"commit":{"parents":[{"commit":"0d43333f3f87fd4daf666b57eac64fea48606fcf","subject":"Merge \"fix for Endpoint writing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/0d43333f3f87fd4daf666b57eac64fea48606fcf"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-29 13:51:15.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-29 13:51:15.000000000","tz":0},"subject":"Fix duplicated affinity policies","message":"Fix duplicated affinity policies\n\nCorrect policy labels and descriptions by renaming duplicated\n“Affinity” and “Anti-affinity” options to Soft Affinity and Soft Anti-affinity\n\nChange-Id: Iad8a01baf19e6be75cd3826a89f276905a7d30e5\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/18f07dc6cb3dbc0178f02d42d49028a09ef8d43d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/18f07dc6cb3dbc0178f02d42d49028a09ef8d43d"}]},"branch":"refs/heads/master"},"1902dd1fd336afff2548c0f373ff47fde8ba4fed":{"kind":"REWORK","_number":2,"created":"2026-01-29 13:52:13.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/56/975156/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/56/975156/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/56/975156/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/56/975156/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/56/975156/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/56/975156/2"}}},"commit":{"parents":[{"commit":"0d43333f3f87fd4daf666b57eac64fea48606fcf","subject":"Merge \"fix for Endpoint writing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/0d43333f3f87fd4daf666b57eac64fea48606fcf"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-29 13:51:15.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-29 13:52:04.000000000","tz":0},"subject":"Fix duplicated affinity policies","message":"Fix duplicated affinity policies\n\nCorrect policy labels and descriptions by renaming duplicated\n“Affinity” and “Anti-affinity” options to Soft Affinity and Soft Anti-affinity\n\nChange-Id: Iad8a01baf19e6be75cd3826a89f276905a7d30e5\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/1902dd1fd336afff2548c0f373ff47fde8ba4fed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/1902dd1fd336afff2548c0f373ff47fde8ba4fed"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
