)]}'
{"id":"openstack%2Fskyline-console~975180","triplet_id":"openstack%2Fskyline-console~master~If7c0c07e893375a99d7504fa3c3c57e00cc88d42","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-01-30 06:07:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If7c0c07e893375a99d7504fa3c3c57e00cc88d42","subject":"Rename Virtual Adapter to Ports","status":"MERGED","created":"2026-01-29 16:21:58.000000000","updated":"2026-01-30 06:08:39.000000000","submitted":"2026-01-30 06:07:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"975180","meta_rev_id":"676c4ca54aeed4535480798535f5aa6ab53695f6","_number":975180,"virtual_id_number":975180,"owner":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-30 06:07:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-01-30 03:34:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-01-30 03:34:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-29 18:02:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-01-30 03:34:47.000000000","updated_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"}],"messages":[{"id":"80f4fe4a32e009b16bad055e59b45cf9c2db3a42","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-29 16:21:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cdd1ab7bb0fe7319311ddd68200f74b158ad565b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-29 16:36:58.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"3afd5a4b29e6a06249d88c9b66ae3d500c99aab6","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-29 16:37:01.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"4008fd2fe982a305f850aca39278c4d3ded4ef89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-29 18:02:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c3b44e4607a46f1a9df38d1138057b6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4e5825a79eb4d7191ab2947383bf019 : SUCCESS in 3m 29s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/1de1a6635ea54c02812c5492346296a7 : SUCCESS in 6m 40s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/f084d3f71243497f98b249d743719b03 : SUCCESS in 3m 13s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/c66898b6811f4cb08ba8d0656de9ad74 : SUCCESS in 6m 43s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/d862b286166b43d58904f6855cf4669d : SUCCESS in 1h 23m 20s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/14917582248e4445992473332220415d : SUCCESS in 43m 14s","accounts_in_message":[],"_revision_number":2},{"id":"126a112ad7943b0aa7517d358e6c1e9e491e3914","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-01-30 03:34:47.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e62e3b16cf3a04b4a44f151e17df2ff0f9edb9b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-30 03:35:22.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"28cd77765bbacbcdaffdd1958c66de6f6e5cd62e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-30 06:07:48.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b92c4e886bac4f5a8db29096c4249187\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db9eb88387f043fbb016f5f040e1525d : SUCCESS in 4m 06s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/2c372d4083fc4423a91a11fe73a25b40 : SUCCESS in 9m 20s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/6978d0b6b8a74abaad51873940a76ff4 : SUCCESS in 3m 16s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/5b49c34b296244cbac80ccf2fe3af246 : SUCCESS in 3m 15s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/182acc934a31400790cd41303e0177a9 : SUCCESS in 1h 35m 27s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/97ce2ea005ff411f84d2fc9322e6da0c : SUCCESS in 1h 06m 11s","accounts_in_message":[],"_revision_number":2},{"id":"bc799eede6c8d529fbdbaf6c3e0e2bdba5f0265b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-30 06:07:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"676c4ca54aeed4535480798535f5aa6ab53695f6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-30 06:08:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e18618f4dd664b248e058ac5adc19660\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6c8ccc32f1c4139b4cdf8f0a175d840 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2300242a25523674d3ec3160f99980587a58acb9","revisions":{"08a9a7c960d9123f6e64e3efb12707ca00d5ea48":{"kind":"REWORK","_number":1,"created":"2026-01-29 16:21:58.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/80/975180/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/80/975180/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/80/975180/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/80/975180/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/80/975180/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/80/975180/1"}}},"commit":{"parents":[{"commit":"0d43333f3f87fd4daf666b57eac64fea48606fcf","subject":"Merge \"fix for Endpoint writing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/0d43333f3f87fd4daf666b57eac64fea48606fcf"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-29 16:21:58.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-29 16:21:58.000000000","tz":0},"subject":"Rename Virtual Adapter to Ports","message":"Rename Virtual Adapter to Ports\n\nStandardize network object naming by using “Ports” \nin accordance with OpenStack definitions.\n\nChange-Id: If7c0c07e893375a99d7504fa3c3c57e00cc88d42\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/08a9a7c960d9123f6e64e3efb12707ca00d5ea48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/08a9a7c960d9123f6e64e3efb12707ca00d5ea48"}]},"branch":"refs/heads/master"},"2300242a25523674d3ec3160f99980587a58acb9":{"kind":"REWORK","_number":2,"created":"2026-01-29 16:36:58.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/80/975180/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/80/975180/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/80/975180/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/80/975180/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/80/975180/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/80/975180/2"}}},"commit":{"parents":[{"commit":"0d43333f3f87fd4daf666b57eac64fea48606fcf","subject":"Merge \"fix for Endpoint writing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/0d43333f3f87fd4daf666b57eac64fea48606fcf"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-29 16:21:58.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-29 16:36:42.000000000","tz":0},"subject":"Rename Virtual Adapter to Ports","message":"Rename Virtual Adapter to Ports\n\nStandardize network object naming by using “Ports” \nin accordance with OpenStack definitions.\n\nChange-Id: If7c0c07e893375a99d7504fa3c3c57e00cc88d42\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/2300242a25523674d3ec3160f99980587a58acb9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/2300242a25523674d3ec3160f99980587a58acb9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
