)]}'
{"id":"openstack%2Fskyline-console~975272","triplet_id":"openstack%2Fskyline-console~master~I9d57df2436019ede207b3e65b57977c661a10a9e","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-02-07 05:47:02.000000000","reason":"Change was submitted"},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-02-04 05:51:57.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}},"hashtags":[],"change_id":"I9d57df2436019ede207b3e65b57977c661a10a9e","subject":"Hide Monitor Center menu by default","status":"MERGED","created":"2026-01-30 12:52:42.000000000","updated":"2026-02-07 05:47:51.000000000","submitted":"2026-02-07 05:47:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":18,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"975272","meta_rev_id":"ad7d9d73e3dae09d7b808f1723d0b987265d7975","_number":975272,"virtual_id_number":975272,"owner":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-07 05:47:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-02-04 05:51:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-02-04 05:51:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-30 14:29:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-01 23:15:32.000000000","updated_by":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2026-02-02 13:19:16.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2026-02-07 03:26:00.000000000","updated_by":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"reviewer":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"state":"CC"}],"messages":[{"id":"87b6b2a1dc978d145d43b57ff60a63ea0ad423f8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-30 12:52:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"96547d6aa0e602c4e70bee818fe5df968afb68d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-30 14:29:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/08fcccd5d4b84e1aba6795a0cad3ac86\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3967e34171dc481fbc5df0ac0acbbf8f : SUCCESS in 3m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/58ae37e7fa8648799cac71f21ece5e21 : SUCCESS in 2m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ec0bbb5c2354592a0d26fa48e81ee13 : SUCCESS in 6m 52s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/999792e13bf547f2a4c688df095bebbf : SUCCESS in 9m 47s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/8c4ebbe1d02a4df8aae6a7c80a945d47 : SUCCESS in 7m 02s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/c203308077d04a6dadb1ce9a6b286513 : SUCCESS in 3m 44s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/731ee3cb06e141888ae05dd56cf1cf93 : SUCCESS in 1h 25m 19s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/13ad569adf0f4631a9b343f0b4fcda6e : SUCCESS in 58m 24s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/caf092e495d2432a9df3cb64b8665a7b : FAILURE in 45m 02s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/c8e71a0a9115435aa48ea493df08029e : SUCCESS in 55m 08s","accounts_in_message":[],"_revision_number":1},{"id":"4b2c8a1caf19afa91aef1993404238a1d9845272","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-30 14:36:24.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"4eb94e9713aab9b288f2a387b361b9697ae77d95","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-30 14:36:48.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"385c34d879b9a653f171d7b119ab1b288542617c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-30 15:07:32.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/84c44e1bf24a4643a3cba56a3d4dfe3f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9e6d03c29654dfe94f93fd8892ab3e4 : SUCCESS in 6m 31s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/44d5c44cc38c481c9a4240a433fbc8d0 : FAILURE in 5m 34s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/1e0336ea66e84d68b7e1fefe4640e7cb : SUCCESS in 9m 16s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/3dbe9441399344ca8a2722137f87d7da : SUCCESS in 6m 09s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/8153052ac3804292a1c7e9d0391e7bcd : FAILURE in 27m 30s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/ab89b241896d40e78ff9503ab7fe0468 : FAILURE in 14m 56s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/5aaafbc7856a4cdea2cc688df4025ca0 : FAILURE in 15m 32s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/af6944e2392748d19f17db4d0a628e5a : FAILURE in 25m 28s","accounts_in_message":[],"_revision_number":3},{"id":"7215f98e9bcd865f17323f8459b57e9b3a57a950","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-31 10:15:46.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"8f73d265a52195d2bb0f041c9186e08f13c6327e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-31 11:44:53.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0e3578464b5f463cad29b2310849d770\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f61666d7115141de9413aee6d2bf55df : SUCCESS in 3m 16s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/a172b67d581f427bbce3c006e7e3ac14 : FAILURE in 8m 42s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/a28f1c06832f49f7adf6198e05932ab7 : SUCCESS in 3m 07s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/725db905c0a54c32b9e331e14aefd4e5 : SUCCESS in 3m 10s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/7021fdbcabef45ef967e82652d209fc1 : SUCCESS in 1h 19m 09s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/2a556a91781a405e9ce16449ab0b5442 : SUCCESS in 41m 42s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/e73f83251ce04bd8886c942810dd81a8 : FAILURE in 51m 17s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/e02e33a7ffd248b3aba741ce009ce2fd : SUCCESS in 51m 04s","accounts_in_message":[],"_revision_number":4},{"id":"c7dfe9138d09c84a420e176c73da2c4e321fb9b0","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-31 16:09:00.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"997e15f7618eabd68aaed5d89a213c0bfd1ee31b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-31 17:38:06.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/058e683fd8a84cd082624dfe362527ac\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd8fcb54fa1942ec833a84c49f4e0e30 : SUCCESS in 5m 50s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/39eb7dad02b84ee78be80a61f131c992 : FAILURE in 11m 09s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/5a6a6f0bab9f40a5a4f4c814c552db64 : SUCCESS in 4m 16s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/7fda7e3b5d5e492cad512b5fca4845b3 : SUCCESS in 7m 24s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/eaeb01eba3f74ebfab9fe7ba0162f526 : SUCCESS in 1h 23m 53s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/f2a9a9d3ed6848a6bbebc61b298129f2 : SUCCESS in 1h 01m 04s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/91eecf1650eb4e268f0ec4c5a165e124 : SUCCESS in 52m 15s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/4bb43951eaee4606902d8c4c11acfd38 : SUCCESS in 32m 53s","accounts_in_message":[],"_revision_number":4},{"id":"bf3fcf2889145f9684f5bb1e3201331c8b77b9e0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-01-31 21:26:02.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"65cf4b80ea3dba623d67c29660ffd68cd129eab0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-31 22:53:55.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c14fd51e75ad4fea9bb92c9ab68a57ce\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01b16efc36e54c96a13bffab207575e8 : SUCCESS in 3m 41s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/120dcdb8fa504c53b81240219d6ee330 : SUCCESS in 10m 57s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/e7531218210d418ebb84f467de57fc8f : SUCCESS in 3m 09s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/9615797cd499460ab905f79f434fa414 : SUCCESS in 4m 44s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/e5cd063cba074ee698c58edaa60e7322 : SUCCESS in 1h 21m 58s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/4a45333399e549f4b3e3a0e9d4ebcf53 : SUCCESS in 44m 11s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/83b08deeb8514fed9cbce9ba8502b359 : SUCCESS in 54m 31s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/16fa2cce1985448ba079d82620c4f89d : SUCCESS in 37m 53s","accounts_in_message":[],"_revision_number":5},{"id":"eacd2659ce34b2b557755d49d0851cd1232333aa","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-02-01 21:08:15.000000000","message":"Patch Set 6: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"561450616bd9a89393ea095b68390afe4359ba29","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-02-01 21:08:19.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":6},{"id":"54b3354b142ba688d706fe7de28b78d493382418","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-02-01 21:12:40.000000000","message":"Patch Set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"df28afad07adb2a68445fea067b0b999d9f42ec9","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-02-01 23:15:32.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"0fb75089b9329c5f1e5bbbdde8daeb8bca3eb8ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-02-02 09:28:32.000000000","message":"Patch Set 8: Commit message was updated.","accounts_in_message":[],"_revision_number":8},{"id":"a23b42619e36c5e8a848b414d8a8f2128a2f9a5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-02 10:46:15.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f34a4a8d33024d228da66a7177fc324d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03dc357a87d44959966d2b822540e54d : SUCCESS in 6m 14s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/0782131d8a1e4244afea13f4f58da302 : SUCCESS in 4m 11s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/fb4e621a55d847619e25c962516e7a2c : SUCCESS in 3m 12s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/1db8fc8465b34ca2a80de5e457b1609c : SUCCESS in 7m 49s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/cf05b9ad83634bb1933835db16797c6b : SUCCESS in 1h 00m 48s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/0c028df6a74c4644bf838ee7bf10b79a : SUCCESS in 1h 12m 15s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/cb1b254749904d61a95f965d12a0e7a4 : SUCCESS in 53m 46s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/f61d9e5cda7444b69f91997d76dc3a2d : FAILURE in 8m 15s","accounts_in_message":[],"_revision_number":8},{"id":"bbfa888a323c5bdc6f5451f8d340b383330b1f7f","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-02-02 11:45:04.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"031c2129179809ab496ce622581509a310bbf99f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-02 13:18:40.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/431bc9b2c57148e988e14794f6f1bf81\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7507aacc717f4744abd41dcc4474370f : SUCCESS in 5m 07s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/aaa909d077054f5798e414c454e55e81 : SUCCESS in 7m 06s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/1c6a4e251bd0451e9f63df7c606f477a : SUCCESS in 4m 42s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/b35caa066d7743cd82a0f8f89a948d66 : SUCCESS in 6m 23s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/bb7f1bdf9f1d41b687d2c046b7f0c2d6 : SUCCESS in 1h 21m 22s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/f3e8bdd682d14c14a3bb80bc05c4f693 : SUCCESS in 59m 50s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/a91cb038b4e943248322c217323f1953 : FAILURE in 40m 04s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/4fa2011a1c1e45bfaba780ebe7b16891 : SUCCESS in 51m 13s","accounts_in_message":[],"_revision_number":8},{"id":"701b58c1806cbc3ea18c3339051b5e5ebddf456b","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-02 13:19:16.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"0cc2cab2582dde431e391f47d6bbca86a06c6a89","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"real_author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-02 13:20:36.000000000","message":"Patch Set 9: Patch Set 8 was rebased on behalf of \u003cGERRIT_ACCOUNT_37632\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}],"_revision_number":9},{"id":"bfc170b269f731fca586c76ff60c7c6a6910e8e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-02 14:24:37.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e574454175594150ab0e6fc7f6160fd3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5949e9c8a88c44be95af2b6e177da171 : SUCCESS in 6m 16s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/2b70ff8903e74e989619c74d1049be9d : SUCCESS in 11m 37s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/fc52cc68e1df4198ab527f5958c04e07 : SUCCESS in 4m 12s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/0d05b553d4f444a6a96072784c2099d8 : SUCCESS in 9m 00s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/851d70e8c1204d7bb474c84575faa941 : FAILURE in 53m 21s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/58fcf278be1f4606b94c3edeb1393fac : SUCCESS in 1h 02m 48s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/38dc24892e0345fe9f340481337b67b1 : SUCCESS in 54m 16s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/f457c2cf8a394e4bbf13ba3ac382fb42 : SUCCESS in 34m 33s","accounts_in_message":[],"_revision_number":9},{"id":"1a47aa81d2876f217d95038c525e35b6b6e7cabd","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-02 14:30:49.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"9a52122de2384e00ef373f82e15ebf2bc64a6031","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-02 16:23:35.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ee1d26fd86e84fd1be21bcc7ba8d4063\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f511769095a9488293d308555c5cc753 : SUCCESS in 6m 44s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/da663597dcb34ce787a580437df65f0b : SUCCESS in 10m 50s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/d58d686b840544eba9d997523323c6c7 : SUCCESS in 7m 14s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/f0da847e46064e8c898b09c3b618d735 : SUCCESS in 7m 30s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/81de74cb4d384822b6d265b48488ea80 : SUCCESS in 1h 06m 17s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/801feddfc4044869be361a67c56c7187 : SUCCESS in 1h 42m 39s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/bdcee4840d124864b5b239b059fef76e : FAILURE in 1h 01m 38s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/2a29da19ba074ce0b505f9a34e152fc7 : SUCCESS in 54m 52s","accounts_in_message":[],"_revision_number":9},{"id":"1aa3685fb595bee2b8833df2f3f33d4e91ffde61","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-02 16:50:35.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"fb94c5cec6729d00cba77b234f62bf2a5feab918","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-02 17:54:28.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94ae7aaae6ed445abac6b1052b74aab2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f59622dfc2b4473853e0355bd6e7362 : SUCCESS in 6m 28s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/4b0160334a614b63b61ff350a299cb47 : SUCCESS in 12m 44s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/d1fcced5974c431abfebaa7fd337aa5d : SUCCESS in 4m 38s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/1115e8b1c26f45a2b309c3b52f024b0f : SUCCESS in 4m 53s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/8e73839ab9a64432a5f500af892ea41d : SUCCESS in 51m 43s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/3ffd7eae4f16438a8b3889e4dd118e7f : SUCCESS in 31m 26s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/7f8e1a5e9a9d4eb2994e3efa0399a0f5 : SUCCESS in 52m 25s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/28c13f0c23cd48b0be91d32b9fda201b : SUCCESS in 57m 45s","accounts_in_message":[],"_revision_number":9},{"id":"ed89f160f997977a6434a3f75f8b7d5cfca995a1","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-02-04 05:51:57.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"0bd4404638a9f01e2d4832d228a7e1ba4dc86fa8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-04 05:52:28.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"b3da5fefe4a89ccf6536e0b4189ca198e416479b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-04 06:43:53.000000000","message":"Patch Set 9: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d44493b98338494faacb0e1e71d15d9e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0ef62a72a9c472fa5bcbe25eb9805c6 : SUCCESS in 3m 46s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/93ef49ace11a4733b51ffb6e764483f7 : SUCCESS in 9m 15s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/5de5781742c44520a1f5b23c4ef61c6b : SUCCESS in 3m 41s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/58ac1fb8395a45fa88afe0eb9ca0f010 : SUCCESS in 3m 13s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/f091c714e78348fb9a6e5cff76713925 : SUCCESS in 48m 09s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/f1b2941b71864305a49cb9608fb7b444 : SUCCESS in 30m 41s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/3a241fbeaa00458b82fae12e2b0887b9 : FAILURE in 40m 00s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/21175ceeefce47a2ac9f0084c5664711 : SUCCESS in 30m 24s","accounts_in_message":[],"_revision_number":9},{"id":"a05a8d5eabfe9b09fc547ece23e4954c6c9712ef","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-02-04 07:19:58.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"c639b38e9ef7b777a98737560a57b99a712144b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-04 08:53:54.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fbc989a21aef4e1fba0c4ecefeb204da\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73af0acbb3dd480bb0aa4d1d9476d3af : SUCCESS in 5m 45s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/a532c3e541e0493db0ca20124a2858f2 : SUCCESS in 6m 17s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/73157454e5444598b1bd3450a28e2896 : SUCCESS in 4m 29s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/bddcfe3b39f345a7b09d5e66d79e8f85 : SUCCESS in 8m 20s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/35f1e704524043d2bc7ab5ed924f4f22 : SUCCESS in 1h 21m 53s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/9caf9eba07f648e3936b32c85195ac1c : SUCCESS in 29m 47s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/61eaf25d6d12494796f4c89bb777985d : FAILURE in 50m 04s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/aaadc238928e46fcab617db12c00e0c6 : SUCCESS in 53m 12s","accounts_in_message":[],"_revision_number":9},{"id":"aa47b75b40f7e5f47d77a1bcd4c91371f4b3cea4","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-02-04 09:58:58.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"ac93d975935f92b7423c1145e0f019df78257bf0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-04 11:41:13.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18fe723d595f41dd89a48bf3251b3cfa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54c7fb4dac2247ea9bb5731d2f7a055f : SUCCESS in 5m 13s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/6031dfef56e44a7788b5a6aa938a9693 : SUCCESS in 11m 07s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/f858473b04a34467ad4fbabdd04c26e4 : SUCCESS in 7m 16s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/f0e62a39315f4d0c9a9454d8ee319858 : SUCCESS in 4m 48s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/f317701b7c5848f5be3eeecae575da2f : SUCCESS in 1h 31m 53s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/d7b9978ac19148bca874cabb6049e0a9 : SUCCESS in 41m 52s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/92f54c4046fe4394b8873558dc40af76 : SUCCESS in 53m 31s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/a8e60b542c024baa85c1160012a78328 : SUCCESS in 30m 05s","accounts_in_message":[],"_revision_number":9},{"id":"57ccc16fa3833d0c4c762f9a9f0e96b35a64ffae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-04 11:41:41.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"af064a6ad0bf2d5c7e8a552f99fb39836a9ae987","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-04 12:50:57.000000000","message":"Patch Set 9: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/19a5283f3df84e7292f141737b1862d1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb5471f5ba6d492a992a132dd2e57e10 : SUCCESS in 6m 53s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/da5dcd0135914d40bc588194dc214c21 : SUCCESS in 10m 37s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/33d928cbf9bd4461ab5f14e4dded279b : SUCCESS in 4m 28s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/dc8b1c7ad0054798a5a214dfe93407ef : SUCCESS in 4m 56s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/9da202b4c5994a038de6bb8634f6001d : FAILURE in 27m 18s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/f4b1f60f57b74df3ac152c035f8cf946 : SUCCESS in 1h 00m 23s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/6bcf844e04da466283532097825c99e6 : SUCCESS in 53m 37s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/526afac5f46d4a479d4cd312b4c9b2e2 : SUCCESS in 59m 55s","accounts_in_message":[],"_revision_number":9},{"id":"1f78aac6c3996e5102ba58387fab17e45b480778","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-02-04 13:04:08.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"ae00d9f097fabd4cf13434cfbcab5ed6875bb897","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-04 14:32:43.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2c8abfc0f3594292aa365fcd5699193d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3eede6397c39445a862d57051ca9e515 : SUCCESS in 5m 35s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/cb8d351206ec436fb90a1a7b798c8426 : SUCCESS in 11m 40s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/172cd6ad215e4bd8846bf92a7f87f1d0 : SUCCESS in 3m 16s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/4006586400534a5a827339fde098c284 : SUCCESS in 5m 08s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/83074e98acdb4b9085a652e737e2c961 : SUCCESS in 1h 23m 07s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/fc11f286466b4d44a3886e47717d55a0 : SUCCESS in 1h 08m 17s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/aa6c53d4d4a04d76875e35bec0f0232f : FAILURE in 43m 08s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/dd2fbb4517b44c779c279270e0828a83 : SUCCESS in 50m 09s","accounts_in_message":[],"_revision_number":9},{"id":"b05f2f2097ba362ae8872f474f417406ec321d33","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-02-04 15:01:02.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"2e6ebcb26a1eb1ef2fa9d6e41094a568a39227f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-04 16:31:35.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65b1437141c349bb80d5b71793971cea\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f06aa9f9ef1742128e5e253384c65e61 : SUCCESS in 7m 03s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/17228fb25f4e45ed89ec1c611d538dd8 : SUCCESS in 7m 32s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/8d175df23c854318b59a2843a8bc4b54 : SUCCESS in 7m 52s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/b953966cfaff4e78b58283230949035a : SUCCESS in 7m 40s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/7b3214aa82934246bcab3d226e2d2ad7 : SUCCESS in 1h 24m 52s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/a97e9524f72d46839398a3b99e67c1c3 : SUCCESS in 1h 17m 52s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/c1f7527962c947d69efbfd9f2b7b7641 : SUCCESS in 53m 54s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/5480be1952ee4e9c816cf6f5dde650ea : SUCCESS in 32m 07s","accounts_in_message":[],"_revision_number":9},{"id":"1b7ce15366ff3de950ae013e48a3e32a2d406df1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-04 16:32:04.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"55fa576964b5f60caa2c95655cf15bd0e9977c11","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-04 17:32:08.000000000","message":"Patch Set 9: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/eb9a741a56f14cad87c098d28ecf9d0f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbf2ac8fdd9b498dbd1cfc755d0f63f8 : SUCCESS in 6m 29s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/fef9c26f73f94d2ba0b8702d57234bee : SUCCESS in 8m 14s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/1a3339e78a4e4614928ab08ca4c7a585 : SUCCESS in 4m 31s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/315b5debf7c84a96b7ec557e7f7a19d9 : SUCCESS in 7m 42s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/b4557582982444fb9c5c89171f23a005 : FAILURE in 54m 18s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/5522bbb6151f484084c826bfccac1cf0 : SUCCESS in 41m 19s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/e93a4a4f1d634c89923fae9b7f68458f : SUCCESS in 53m 15s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/80b21e1e61544c0d9fc8fa67ad39ad70 : SUCCESS in 41m 47s","accounts_in_message":[],"_revision_number":9},{"id":"d6385d78c7c503258dca767faf8dac2dd8471c55","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-02-04 19:35:05.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"91dc07e16d35cdcd25cfadf26ced156345740f7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-04 20:46:45.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/382c9d3012ce4000acb46d3351fc03a0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e92242fabc474423a3191dba23558c2a : SUCCESS in 6m 14s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/5fb9d9c30f0740b1a0dc091e9a92daa6 : SUCCESS in 11m 24s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/6bdda0b0772547a98c57d6775d6d4d35 : SUCCESS in 3m 20s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/1ccb4fc946f04e0c9d65231dd9b42390 : SUCCESS in 3m 25s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/86f892c2b8d843559c9a6fbce5a7ed27 : SUCCESS in 1h 09m 49s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/a0a2110f93ef4a91ba30ff5de5333ef0 : FAILURE in 42m 49s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/de83f9e0144d46ab9e0c8688a578b675 : FAILURE in 40m 25s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/1f6115d12548403b9eac7091c7426db8 : SUCCESS in 33m 10s","accounts_in_message":[],"_revision_number":9},{"id":"a44b2b4a711e3698a7ead87eb4c765c1de437475","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-02-04 21:13:53.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"4a9045b277bce614223fc9aa85aebb913b3dadeb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-04 22:43:48.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f3f3aa60dbca499eb00a264b40009b5c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76bee74cef204cedb015d9cc46009215 : SUCCESS in 6m 27s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/ba9a511e21a04585a29546e179270245 : SUCCESS in 11m 18s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/a907f86b2318424f8a9e166aa5b17f32 : SUCCESS in 8m 29s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/45d562cfb58240deac523d95cbed5643 : SUCCESS in 3m 52s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/07f22fdeb1f44938bed73a945b7d2d94 : SUCCESS in 1h 23m 31s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/a056d971835748258a2158b673ef8397 : FAILURE in 36m 17s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/5d11de2fe94b4060addcbfaf0bf22e8d : SUCCESS in 56m 43s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/d5ccc31815844044b9bbbb2823ccf712 : SUCCESS in 59m 29s","accounts_in_message":[],"_revision_number":9},{"id":"cc12dffd8dad440da75c983347e888851db8cfed","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-02-04 22:44:51.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"36fb0ed504063722c1068a8a0ed72fd3076b05a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-05 00:12:59.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/29390ef6d8704fc2b4172e740d46cbf5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c03ea916795f4c7b97bc71dd4e3ff00c : SUCCESS in 3m 45s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/b865503b29534073958f8f0abcca0892 : SUCCESS in 9m 21s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/d4809d467e9c4bb0af0d4e1ea0f00720 : SUCCESS in 3m 39s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/4d60b5c7281f496b983eac446abffeea : SUCCESS in 7m 17s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/6d8100c6a7794ee7a748cf2303fbaff0 : SUCCESS in 1h 22m 39s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/af7a6eb39f824a0881be14731d0c1eeb : FAILURE in 35m 03s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/32bd8c6883c4492cbaa920e13773e612 : SUCCESS in 51m 02s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/971e097a5f834ed6838562df4797bc0a : SUCCESS in 41m 57s","accounts_in_message":[],"_revision_number":9},{"id":"7d97aabded333daae2013b9a8be48ca72e395414","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-02-05 09:11:05.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"c86d8f04e7664537512764f5b3b331acc0a83959","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-05 10:42:23.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5748ec4309c54806b8e693595b164703\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84c35cf53b8445908231cd93e3852188 : SUCCESS in 4m 15s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/f49d00e058be47e4b3bbb5bf833e2406 : SUCCESS in 8m 29s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/db5261ddb2614b28a1bb44d9e0edd1af : SUCCESS in 8m 24s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/6bd4cc61e71a4ed28180474842ce9440 : SUCCESS in 6m 32s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/af3e461272334cee8f6e64bd8cb91910 : SUCCESS in 1h 26m 07s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/c2f0e93e6aab40a7906380f7300bda1f : FAILURE in 39m 03s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/fdf8aa381e1147bf85dc34c01567ebf4 : SUCCESS in 56m 51s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/28c325b9bd2449749e1d58b71ae9e733 : SUCCESS in 47m 56s","accounts_in_message":[],"_revision_number":9},{"id":"d831dbc252ef84adf5c2863e3f5d183692d00c26","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-06 12:44:59.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"3a3d90bfb5175cd8dbf2de8def12fc57bdde8941","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-06 14:16:31.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55c8af8bdde947bab0d4dcab19224d38\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eae7defc96a74e8db08273d140838837 : SUCCESS in 6m 48s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/1cefee568e2f4fc69e1fac21083d55b6 : SUCCESS in 11m 28s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/7150160647c240e59550bde3687a8339 : SUCCESS in 7m 26s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/277f9c85195144ada9e64722f2da176f : SUCCESS in 7m 10s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/8e434bea93824c2391f4fbb098fed16b : SUCCESS in 1h 26m 08s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/e1b333715bb0427983e61e2f08a23fd8 : SUCCESS in 1h 17m 38s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/fb41b3ca1991442897ef489f28f5714b : SUCCESS in 30m 49s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/6cf248a312964f859231be15a32b4ec4 : SUCCESS in 51m 39s","accounts_in_message":[],"_revision_number":9},{"id":"4e8d8bd776510d8143a7a27c9cc5d7e9e48b6e1c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-06 14:17:00.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"5008ede608e3bb6c15e1c18a97c3276d73f33e86","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-06 15:05:03.000000000","message":"Patch Set 9: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1d564b92c73748738250875fd1fe8a6d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8d33caaad924668a03665d9b80b0a4f : SUCCESS in 5m 50s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/4d212b73a9b446c9b1685b366dcaa2bf : SUCCESS in 4m 23s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/956f5cf0c5d6490f86e59c09e1471b0d : SUCCESS in 3m 16s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/16aec68d4e8e4c13b29efd95b3314839 : SUCCESS in 3m 16s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/85647066479344869782cb4b99489d3a : SUCCESS in 47m 09s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/e785c58c3ace47e98fa70072bfa6eb3e : SUCCESS in 30m 05s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/1cd90d91f7454be28913f5ef7b3897e6 : FAILURE in 40m 42s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/5244499fb6bd49168385397365242517 : SUCCESS in 30m 23s","accounts_in_message":[],"_revision_number":9},{"id":"f1cbc699c75180702bb2c26d09055654b20caeec","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-06 18:28:14.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"c85367613bcfaba1c76be4f36606c275369778b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-06 19:57:32.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c572f2131e94d29896d12c2897939a2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70415e2e809747f4b332dc2594dfc8fa : SUCCESS in 6m 45s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/95cc5fa0e4d6470f9a22271f7cb112a3 : SUCCESS in 11m 38s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/04d6f72d5ad9449ca7d27008fadda15a : SUCCESS in 3m 16s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/1f19f25c723a4a5a8842487303bdbafa : SUCCESS in 4m 48s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/ae91079d56944972a4935fa94b5b6870 : SUCCESS in 1h 21m 52s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/7c7e9a12738c4cf7bb498a93674ece17 : SUCCESS in 1h 00m 37s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/4070e849f5714c70965784dee4a2cbdd : SUCCESS in 37m 39s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/d3347cf9bf97445697eedd6a2e69febb : SUCCESS in 38m 06s","accounts_in_message":[],"_revision_number":9},{"id":"a6165430289e51486dfd95bd930d2500e7b42003","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-06 19:58:03.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"e71edd9c1ba476c39bebda3ab60a55d2580a0604","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-06 21:27:18.000000000","message":"Patch Set 9: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6cc124b0d8e2446e8f939564d5c1080e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fff892c9236c4b8282d5b8ba5510bb1f : SUCCESS in 3m 55s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/217962bff7114a36b32daf273ef9ab82 : SUCCESS in 10m 59s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/6aec148f3dca41c8afc4c34b25489b71 : SUCCESS in 8m 17s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/6fb027645bbb4464bf53b9487008b861 : SUCCESS in 4m 40s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/a7eed5776c0d498b87d1c5bb281d8eba : SUCCESS in 1h 24m 22s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/f8c9671b314e420ea5a795c39e70e911 : SUCCESS in 57m 17s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/b1b8493aecd14a4bba5c71c4a7fac611 : FAILURE in 40m 33s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/4dbe57a0f85c46d1928c3a8f1c64a7d5 : SUCCESS in 50m 57s","accounts_in_message":[],"_revision_number":9},{"id":"418b13eef6047fe4460721adb937293086c664ff","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-06 21:40:37.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"ba71a5ec4c7af1fe5a3889a674eaf63bdda57e37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-06 23:06:39.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6efa84040844500b195366d0e12860f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/babf7e68e33943f3b0dc03417fa79f1f : SUCCESS in 6m 08s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/6fef67aabd244191b4aeac48650786e2 : SUCCESS in 11m 48s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/4d793e91d3ac43968590c02ca4bff4aa : SUCCESS in 4m 04s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/fc7ef6a2dcf34a448b4efe1dd6ab27e3 : SUCCESS in 6m 44s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/8a345e84c2c94c2fa761163a540a83e0 : SUCCESS in 1h 20m 45s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/562f20b36c0a4cb2bbd1c465071753df : SUCCESS in 1h 00m 18s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/f3c40000b47a4365a784c91f8f539d39 : SUCCESS in 35m 54s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/b02df3c94531491e9b460ea7d441b3d2 : SUCCESS in 43m 28s","accounts_in_message":[],"_revision_number":9},{"id":"009066c299cde56745e5c616412583bd0c9ae7ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-06 23:07:07.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"e3cab0a0aed9cfe97cd2f098ee04532d395f5a62","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-07 00:06:01.000000000","message":"Patch Set 9: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a75475d997ca4cae862647a3ac89aa09\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f14dfd36acf4d67935bef196fb0335a : SUCCESS in 3m 22s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/ee08d7d3804d4e0a808fb8180c6dd2db : SUCCESS in 4m 49s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/75fb14c051ca446cbb4bb158ba75adba : SUCCESS in 3m 07s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/5e163c28ef7c44b486b3dc123f6eada5 : SUCCESS in 3m 17s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/61b7d6cc00be4ea0bb84518964f3fd43 : FAILURE in 53m 37s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/39f578d17b88415191edde02032520c8 : SUCCESS in 30m 06s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/c9853c514c9e475d96a87b20ac7c0eec : SUCCESS in 54m 08s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/91d79dcbe9404c37bd586a788e5c6450 : SUCCESS in 50m 52s","accounts_in_message":[],"_revision_number":9},{"id":"1c2ea194503f6ed1cf3c0b2f2b8f3fd974de257d","author":{"_account_id":28706,"name":"Boxiang Zhu","email":"bxzhu_5355@163.com","username":"ZhuBoxiang"},"date":"2026-02-07 03:26:00.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"798262ee8294a1e4c5c77c863799b7f5a9e7fcf0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-07 04:22:03.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f198b7bc6885463fb5f999ed61204af5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb96fd64ca7e4cb980329960f2dac20a : SUCCESS in 5m 36s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/e92069ecf77649149489051a4075c087 : SUCCESS in 11m 37s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/b107bacb54604f6c8dbfdcd1d97fc01a : SUCCESS in 4m 39s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/9e1da5f9f88e4830bf21a92ad53cc9d1 : SUCCESS in 4m 47s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/67f25d9e6da04d92b4e06ee7e5f7a5ad : SUCCESS in 49m 28s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/cacce2befdd54133b34ba672b5e72a0e : SUCCESS in 30m 32s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/40aaa337c33040c0adf372f928a100e4 : SUCCESS in 43m 09s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/9b3adfacfd9742d78a6c92302d7d5040 : SUCCESS in 50m 52s","accounts_in_message":[],"_revision_number":9},{"id":"dd17409c71742d732ca89ea46ef1381bfdab4826","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-07 04:22:35.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"5b71b909df9479501f4b86830b4357cec769f488","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-07 05:47:02.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af649768a58845f19d97ce8b88cb227d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4545dc182b2543a792b731507821d35f : SUCCESS in 6m 21s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/d2665e45e59b4f70a8a53aabd68acd1f : SUCCESS in 9m 01s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/759aab05869a48e1a6bc89642f087fde : SUCCESS in 5m 00s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/b80b2e42eb0e401c8ee8465cf539080a : SUCCESS in 4m 44s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/b0a2421886a541739f2d57b70a6dcb3f : SUCCESS in 1h 19m 45s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/513b367f2edb47868a95ae4dfaa2e090 : SUCCESS in 1h 00m 48s\n- skyline-console-devstack-e2etests-other https://zuul.opendev.org/t/openstack/build/919a08cb681c450b8265c35ab2ebcc79 : SUCCESS in 38m 10s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/e19ffcf4c3f04b8890a6f991ea01dbc4 : SUCCESS in 29m 55s","accounts_in_message":[],"_revision_number":9},{"id":"d2f2ef6b7e5395c2d33b1945751f69feb8411860","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-07 05:47:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9},{"id":"ad7d9d73e3dae09d7b808f1723d0b987265d7975","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-07 05:47:51.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bde0f1413f164ecca7b0afb4e27d837a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b00cc644a85c48aa9ba10af884c0b232 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"083bfa19cdc2c3323e88dd548e8a87c1c464bc56","revisions":{"360519432c12e4cf8f78e3ebcc03448811fc3025":{"kind":"REWORK","_number":1,"created":"2026-01-30 12:52:42.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/72/975272/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/72/975272/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/72/975272/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/72/975272/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/72/975272/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/72/975272/1"}}},"commit":{"parents":[{"commit":"154c2609a9c6fd5e27009807e808cdef879533bc","subject":"Merge \"Fix formatting in Member Detail Page\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/154c2609a9c6fd5e27009807e808cdef879533bc"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-30 12:52:42.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-30 12:52:42.000000000","tz":0},"subject":"WIP","message":"WIP\n\nChange-Id: I9d57df2436019ede207b3e65b57977c661a10a9e\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/360519432c12e4cf8f78e3ebcc03448811fc3025"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/360519432c12e4cf8f78e3ebcc03448811fc3025"}]},"branch":"refs/heads/master"},"bf9b1c9782b3fd25bb5dfaba9c9b25e6011f8cf5":{"kind":"REWORK","_number":2,"created":"2026-01-30 14:36:24.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/72/975272/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/72/975272/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/72/975272/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/72/975272/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/72/975272/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/72/975272/2"}}},"commit":{"parents":[{"commit":"154c2609a9c6fd5e27009807e808cdef879533bc","subject":"Merge \"Fix formatting in Member Detail Page\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/154c2609a9c6fd5e27009807e808cdef879533bc"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-30 12:52:42.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-30 14:36:02.000000000","tz":0},"subject":"WIP","message":"WIP\n\nChange-Id: I9d57df2436019ede207b3e65b57977c661a10a9e\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/bf9b1c9782b3fd25bb5dfaba9c9b25e6011f8cf5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/bf9b1c9782b3fd25bb5dfaba9c9b25e6011f8cf5"}]},"branch":"refs/heads/master"},"a5a7d7adea480fab7e2eb50a91ea2b98537fb657":{"kind":"REWORK","_number":3,"created":"2026-01-30 14:36:48.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/72/975272/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/72/975272/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/72/975272/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/72/975272/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/72/975272/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/72/975272/3"}}},"commit":{"parents":[{"commit":"154c2609a9c6fd5e27009807e808cdef879533bc","subject":"Merge \"Fix formatting in Member Detail Page\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/154c2609a9c6fd5e27009807e808cdef879533bc"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-30 12:52:42.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-30 14:36:46.000000000","tz":0},"subject":"WIP","message":"WIP\n\nChange-Id: I9d57df2436019ede207b3e65b57977c661a10a9e\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a5a7d7adea480fab7e2eb50a91ea2b98537fb657"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/a5a7d7adea480fab7e2eb50a91ea2b98537fb657"}]},"branch":"refs/heads/master"},"9dfb3d19777a8f88adc8cefd9603e793dd6f28d3":{"kind":"REWORK","_number":4,"created":"2026-01-31 10:15:46.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/72/975272/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/72/975272/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/72/975272/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/72/975272/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/72/975272/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/72/975272/4"}}},"commit":{"parents":[{"commit":"154c2609a9c6fd5e27009807e808cdef879533bc","subject":"Merge \"Fix formatting in Member Detail Page\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/154c2609a9c6fd5e27009807e808cdef879533bc"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-30 12:52:42.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-31 10:15:45.000000000","tz":0},"subject":"WIP","message":"WIP\n\nChange-Id: I9d57df2436019ede207b3e65b57977c661a10a9e\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/9dfb3d19777a8f88adc8cefd9603e793dd6f28d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/9dfb3d19777a8f88adc8cefd9603e793dd6f28d3"}]},"branch":"refs/heads/master"},"b87f9411fe2bbe39d3d4bc8a10452140b1c39935":{"kind":"REWORK","_number":5,"created":"2026-01-31 21:26:02.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/72/975272/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/72/975272/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/72/975272/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/72/975272/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/72/975272/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/72/975272/5"}}},"commit":{"parents":[{"commit":"154c2609a9c6fd5e27009807e808cdef879533bc","subject":"Merge \"Fix formatting in Member Detail Page\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/154c2609a9c6fd5e27009807e808cdef879533bc"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-30 12:52:42.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-31 21:26:02.000000000","tz":0},"subject":"WIP","message":"WIP\n\nChange-Id: I9d57df2436019ede207b3e65b57977c661a10a9e\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b87f9411fe2bbe39d3d4bc8a10452140b1c39935"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b87f9411fe2bbe39d3d4bc8a10452140b1c39935"}]},"branch":"refs/heads/master"},"f229b01083c88b333c0e68a912ca3fde9cd4848a":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2026-02-01 21:08:15.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/72/975272/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/72/975272/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/72/975272/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/72/975272/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/72/975272/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/72/975272/6"}}},"commit":{"parents":[{"commit":"154c2609a9c6fd5e27009807e808cdef879533bc","subject":"Merge \"Fix formatting in Member Detail Page\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/154c2609a9c6fd5e27009807e808cdef879533bc"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-30 12:52:42.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-02-01 21:08:15.000000000","tz":0},"subject":"Hide Monitor Center menu by default","message":"Hide Monitor Center menu by default\n\nMonitor Center requires Prometheus for monitoring data, which is\nnot included in the default deployment.\n\nChange-Id: I9d57df2436019ede207b3e65b57977c661a10a9e\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f229b01083c88b333c0e68a912ca3fde9cd4848a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/f229b01083c88b333c0e68a912ca3fde9cd4848a"}]},"branch":"refs/heads/master","description":"Edit commit message"},"8d109d07a3d12043260dacf16b9218a3b79e3464":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2026-02-01 21:12:40.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/72/975272/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/72/975272/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/72/975272/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/72/975272/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/72/975272/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/72/975272/7"}}},"commit":{"parents":[{"commit":"154c2609a9c6fd5e27009807e808cdef879533bc","subject":"Merge \"Fix formatting in Member Detail Page\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/154c2609a9c6fd5e27009807e808cdef879533bc"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-30 12:52:42.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-02-01 21:12:40.000000000","tz":0},"subject":"Hide Monitor Center menu by default","message":"Hide Monitor Center menu by default\n\nMonitor Center requires Prometheus for monitoring data, which is\nnot included in the default deployment.\n\n\nCloses-Bug: #2138157\nChange-Id: I9d57df2436019ede207b3e65b57977c661a10a9e\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/8d109d07a3d12043260dacf16b9218a3b79e3464"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/8d109d07a3d12043260dacf16b9218a3b79e3464"}]},"branch":"refs/heads/master","description":"Edit commit message"},"b61f380fb13c97054ca730067fd4c7e209cb97cf":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2026-02-02 09:28:32.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/72/975272/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/72/975272/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/72/975272/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/72/975272/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/72/975272/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/72/975272/8"}}},"commit":{"parents":[{"commit":"154c2609a9c6fd5e27009807e808cdef879533bc","subject":"Merge \"Fix formatting in Member Detail Page\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/154c2609a9c6fd5e27009807e808cdef879533bc"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-30 12:52:42.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-02-02 09:28:32.000000000","tz":0},"subject":"Hide Monitor Center menu by default","message":"Hide Monitor Center menu by default\n\nMonitor Center requires Prometheus for monitoring data, which is\nnot included in the default deployment.\n\nCloses-Bug: #2138157\nChange-Id: I9d57df2436019ede207b3e65b57977c661a10a9e\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b61f380fb13c97054ca730067fd4c7e209cb97cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b61f380fb13c97054ca730067fd4c7e209cb97cf"}]},"branch":"refs/heads/master","description":"Edit commit message"},"083bfa19cdc2c3323e88dd548e8a87c1c464bc56":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2026-02-02 13:20:36.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"real_uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/72/975272/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/72/975272/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/72/975272/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/72/975272/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/72/975272/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/72/975272/9"}}},"commit":{"parents":[{"commit":"cd3f344b6947b209925914390c9fc30d98f9603e","subject":"Merge \"Fix IPv6 subnet fields visibility in network form\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/cd3f344b6947b209925914390c9fc30d98f9603e"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-01-30 12:52:42.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-02-02 13:20:36.000000000","tz":0},"subject":"Hide Monitor Center menu by default","message":"Hide Monitor Center menu by default\n\nMonitor Center requires Prometheus for monitoring data, which is\nnot included in the default deployment.\n\nCloses-Bug: #2138157\nChange-Id: I9d57df2436019ede207b3e65b57977c661a10a9e\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/083bfa19cdc2c3323e88dd548e8a87c1c464bc56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/083bfa19cdc2c3323e88dd548e8a87c1c464bc56"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
