)]}'
{"id":"openstack%2Fskyline-console~975374","triplet_id":"openstack%2Fskyline-console~master~I049a8dff49989c693518bff93683a8c778bb0002","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-02-02 13:06:11.000000000","reason":"Change was submitted"},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-02-02 12:04:01.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}},"hashtags":[],"change_id":"I049a8dff49989c693518bff93683a8c778bb0002","subject":"Fix IPv6 subnet fields visibility in network form","status":"MERGED","created":"2026-02-02 09:41:57.000000000","updated":"2026-02-02 13:07:04.000000000","submitted":"2026-02-02 13:06:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"975374","meta_rev_id":"3fff4c46bfae2ebec16c416a9c3e1503c166c094","_number":975374,"virtual_id_number":975374,"owner":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-02 13:06:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-02-02 12:04:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-02-02 11:47:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-02-02 12:04:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-02 10:34:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-02 11:47:38.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-02-02 11:47:38.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"}],"messages":[{"id":"2e3f9b4b81cc6f83f97646876be6cb72c141d2fe","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-02-02 09:41:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ac5a80cbd7418a4bf35c42a3a8c764a7581cfe67","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-02-02 09:48:13.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"fb34372c4095ed23d4f691aa35f40e0d20f028a0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-02-02 09:48:20.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"6d0a33032304b50ebce0fcd5265e749db5694de2","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-02-02 10:22:17.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":3},{"id":"b304a77132760b579b38fd841c2af553ef615b51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-02 10:34:12.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0eab43fcb3a448ea2b6717992ee90a3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cccce98b29a540bca83ca50d13799436 : SUCCESS in 6m 13s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/d4e1b461213249c495bca7be0d9ee753 : SUCCESS in 10m 57s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/26599b33d56a441fa6f66d6d63c83ad9 : SUCCESS in 3m 36s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/3fc79be7627b40469165631039f39646 : SUCCESS in 6m 51s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/77c9a0181e6a45b598c56662bf3d2f4a : SUCCESS in 44m 31s","accounts_in_message":[],"_revision_number":3},{"id":"17a64bba7622146ace45641087fbad0555251526","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-02 11:47:38.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"26b4305ed08819752449b918c5d4286278416d43","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-02-02 12:04:01.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"9d4f650defc09cb366019c358fbb060b7bd94ef4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-02 12:04:26.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a662fb4247dca6a8c883026bd30b98d15807b71c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-02 13:06:11.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e26e370ca414c3290fc97f9654ebed9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13269312f9be48978bf1ab158b84e05a : SUCCESS in 6m 23s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/f0457dbb10c447239dbf9fddc6ccc7bd : SUCCESS in 11m 02s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/bb4a5ba8263a4c768ac5299d791d67bf : SUCCESS in 4m 42s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/22509a8aeb08432884326b97f264a8b9 : SUCCESS in 7m 35s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/66a7e13325de43ea983b24125e1d6e16 : SUCCESS in 1h 01m 41s","accounts_in_message":[],"_revision_number":3},{"id":"ae467b2ccd51dbcd050c964a6c625e15f29951b2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-02 13:06:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"3fff4c46bfae2ebec16c416a9c3e1503c166c094","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-02 13:07:04.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/11aa9c187da8431eb91ebcce46210dcd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d0e2fbbc8a04195ad51c7f498a43c59 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6b12a08e6c32319c492ad5445b576c81126ac3ea","revisions":{"2d42cbfec347d5dcb1fc1471e1d979bcc6f903d0":{"kind":"REWORK","_number":1,"created":"2026-02-02 09:41:57.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/74/975374/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/74/975374/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/74/975374/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/74/975374/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/74/975374/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/74/975374/1"}}},"commit":{"parents":[{"commit":"57adeccbe2af451f1f54ddd39018d0a45b8b72c8","subject":"Merge \"Rename \"Weights\" to \"Weight\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/57adeccbe2af451f1f54ddd39018d0a45b8b72c8"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-02-02 09:41:57.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-02-02 09:41:57.000000000","tz":0},"subject":"Fix IPv6 subnet fields visibility in network form","message":"Fix IPv6 subnet fields visibility in network form\n\nHide IPv6 subnet fields when subnet creation is\n disabled\n\nChange-Id: I049a8dff49989c693518bff93683a8c778bb0002\nCloses-Bug: #2125188\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/2d42cbfec347d5dcb1fc1471e1d979bcc6f903d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/2d42cbfec347d5dcb1fc1471e1d979bcc6f903d0"}]},"branch":"refs/heads/master"},"b13472a88d1fbf62fe94b88d29067f063549dc8f":{"kind":"REWORK","_number":2,"created":"2026-02-02 09:48:13.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/74/975374/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/74/975374/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/74/975374/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/74/975374/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/74/975374/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/74/975374/2"}}},"commit":{"parents":[{"commit":"57adeccbe2af451f1f54ddd39018d0a45b8b72c8","subject":"Merge \"Rename \"Weights\" to \"Weight\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/57adeccbe2af451f1f54ddd39018d0a45b8b72c8"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-02-02 09:41:57.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-02-02 09:47:53.000000000","tz":0},"subject":"Fix IPv6 subnet fields visibility in network form","message":"Fix IPv6 subnet fields visibility in network form\n\nHide IPv6 subnet fields when subnet creation is\n disabled\n\nChange-Id: I049a8dff49989c693518bff93683a8c778bb0002\nCloses-Bug: #2125188\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b13472a88d1fbf62fe94b88d29067f063549dc8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b13472a88d1fbf62fe94b88d29067f063549dc8f"}]},"branch":"refs/heads/master"},"6b12a08e6c32319c492ad5445b576c81126ac3ea":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-02-02 09:48:20.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/74/975374/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/74/975374/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/74/975374/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/74/975374/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/74/975374/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/74/975374/3"}}},"commit":{"parents":[{"commit":"57adeccbe2af451f1f54ddd39018d0a45b8b72c8","subject":"Merge \"Rename \"Weights\" to \"Weight\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/57adeccbe2af451f1f54ddd39018d0a45b8b72c8"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-02-02 09:41:57.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-02-02 09:48:20.000000000","tz":0},"subject":"Fix IPv6 subnet fields visibility in network form","message":"Fix IPv6 subnet fields visibility in network form\n\nHide IPv6 subnet fields when subnet creation is\ndisabled\n\nChange-Id: I049a8dff49989c693518bff93683a8c778bb0002\nCloses-Bug: #2125188\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6b12a08e6c32319c492ad5445b576c81126ac3ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/6b12a08e6c32319c492ad5445b576c81126ac3ea"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
