)]}'
{"id":"openstack%2Fskyline-console~975385","triplet_id":"openstack%2Fskyline-console~master~I0b24bc84815e4e7c025dd0cdccbc634033a22c2b","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-02-04 04:22:24.000000000","reason":"Change was submitted"},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-02-04 03:08:02.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-02-04 04:22:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0b24bc84815e4e7c025dd0cdccbc634033a22c2b","subject":"Source_IP replace to source IP","status":"MERGED","created":"2026-02-02 11:57:19.000000000","updated":"2026-02-04 04:23:19.000000000","submitted":"2026-02-04 04:22:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"975385","meta_rev_id":"b67f51b07ebe9242e12dc202680bed9048af8436","_number":975385,"virtual_id_number":975385,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-04 04:22:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-02-04 03:08:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-02-04 03:08:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-02 12:00:15.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"},{"updated":"2026-02-02 12:00:15.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2026-02-02 12:55:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c77101780919ba7752168dfe83a9a8b1544d66ea","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-02 11:57:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"adce2c148d91490de7b19205fee97317845b4d17","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-02 11:59:52.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"0dc2ebf7197cecd4cbda5f0818de7e1500d9c9a4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-02 12:00:04.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"d3979e0b75067148d137eb6020855b74eed69013","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-02 12:00:15.000000000","message":"Patch Set 3:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"15551f31ccbca3990c02843d943a8ab09db90743","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-02 12:02:55.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"935a02171847044de96285d4f90a6e553a9d5434","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-02 12:03:10.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"b3dfc5171bf92831d8afe3ecd1e319d02153e120","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-02 12:55:20.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75bc4a49599d4f709bbda69115ca8156\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bea1a0810d6f47d1b953246ff4c02a08 : SUCCESS in 6m 25s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/6e03131f95854b53bda2538615d491ac : SUCCESS in 11m 18s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/88e2992014454a2c9c9e0779b0122593 : SUCCESS in 7m 45s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/4d6d1a19a0a244d49ff2bb93ccb9f846 : SUCCESS in 6m 19s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/562b88257a3947f1af6c8f00950b89d7 : SUCCESS in 48m 34s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/3514ba932cab4c0aa832d5377537cdd3 : SUCCESS in 51m 41s","accounts_in_message":[],"_revision_number":5},{"id":"64df4cf236242db58b6d2bebb8945b3c1341cda5","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-02-04 03:08:02.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"6213662b8763206b20c99ce8448600b920eb969e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-04 03:08:28.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"4c81122149b46c31decad392e26d1a43bbf66709","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-04 04:22:24.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00b5e7e1b5634e559daa6efac3016a93\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efff753537354bf19e8fe7815e8def8f : SUCCESS in 4m 15s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/ab8ff6ec71b0477883fa19aba0a67ab2 : SUCCESS in 11m 33s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/01066ae8dded4b4e9c2007762bee8c2f : SUCCESS in 7m 42s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/e878cb485910441dacd494ed607f5a01 : SUCCESS in 8m 28s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/a83463c758ca4f239c0688b848be6f3d : SUCCESS in 1h 05m 13s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/b5d27002b89e465098a1dc7e0b85672a : SUCCESS in 50m 28s","accounts_in_message":[],"_revision_number":5},{"id":"47ddd123fd4596103b5ad714a05d5ae75b95b6a1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-04 04:22:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"b67f51b07ebe9242e12dc202680bed9048af8436","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-04 04:23:19.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/086b195727be4df1bffdcf5f2ba6b9f6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa4dfc8055544929b6bef60d865e2bb9 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"30eeed148d0ec599a7b002293a6aebfaef237f77","revisions":{"e702328a1b047236ba5c6c0dc603e108ab8b1ace":{"kind":"REWORK","_number":1,"created":"2026-02-02 11:57:19.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/85/975385/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/85/975385/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/85/975385/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/85/975385/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/85/975385/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/85/975385/1"}}},"commit":{"parents":[{"commit":"57adeccbe2af451f1f54ddd39018d0a45b8b72c8","subject":"Merge \"Rename \"Weights\" to \"Weight\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/57adeccbe2af451f1f54ddd39018d0a45b8b72c8"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-02 11:57:19.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-02 11:57:19.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: I0b24bc84815e4e7c025dd0cdccbc634033a22c2b\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e702328a1b047236ba5c6c0dc603e108ab8b1ace"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e702328a1b047236ba5c6c0dc603e108ab8b1ace"}]},"branch":"refs/heads/master"},"647dc722e5f33155ebe66a300f7b731e3022f341":{"kind":"REWORK","_number":2,"created":"2026-02-02 11:59:52.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/85/975385/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/85/975385/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/85/975385/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/85/975385/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/85/975385/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/85/975385/2"}}},"commit":{"parents":[{"commit":"57adeccbe2af451f1f54ddd39018d0a45b8b72c8","subject":"Merge \"Rename \"Weights\" to \"Weight\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/57adeccbe2af451f1f54ddd39018d0a45b8b72c8"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-02 11:57:19.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-02 11:59:52.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: I0b24bc84815e4e7c025dd0cdccbc634033a22c2b\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/647dc722e5f33155ebe66a300f7b731e3022f341"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/647dc722e5f33155ebe66a300f7b731e3022f341"}]},"branch":"refs/heads/master"},"62f7ca4222df46daa9a8b44f323e158af884c6e6":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-02-02 12:00:04.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/85/975385/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/85/975385/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/85/975385/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/85/975385/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/85/975385/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/85/975385/3"}}},"commit":{"parents":[{"commit":"57adeccbe2af451f1f54ddd39018d0a45b8b72c8","subject":"Merge \"Rename \"Weights\" to \"Weight\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/57adeccbe2af451f1f54ddd39018d0a45b8b72c8"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-02 11:57:19.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-02 12:00:04.000000000","tz":0},"subject":"Source_IP replace to source IP","message":"Source_IP replace to source IP\n\nChange-Id: I0b24bc84815e4e7c025dd0cdccbc634033a22c2b\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/62f7ca4222df46daa9a8b44f323e158af884c6e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/62f7ca4222df46daa9a8b44f323e158af884c6e6"}]},"branch":"refs/heads/master","description":"Edit commit message"},"8e2a34688dd8c68f6182c790ce6bbd9c42cc2422":{"kind":"REWORK","_number":4,"created":"2026-02-02 12:02:55.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/85/975385/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/85/975385/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/85/975385/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/85/975385/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/85/975385/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/85/975385/4"}}},"commit":{"parents":[{"commit":"57adeccbe2af451f1f54ddd39018d0a45b8b72c8","subject":"Merge \"Rename \"Weights\" to \"Weight\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/57adeccbe2af451f1f54ddd39018d0a45b8b72c8"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-02 11:57:19.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-02 12:02:55.000000000","tz":0},"subject":"Source_IP replace to source IP","message":"Source_IP replace to source IP\n\nChange-Id: I0b24bc84815e4e7c025dd0cdccbc634033a22c2b\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/8e2a34688dd8c68f6182c790ce6bbd9c42cc2422"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/8e2a34688dd8c68f6182c790ce6bbd9c42cc2422"}]},"branch":"refs/heads/master"},"30eeed148d0ec599a7b002293a6aebfaef237f77":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2026-02-02 12:03:10.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/85/975385/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/85/975385/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/85/975385/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/85/975385/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/85/975385/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/85/975385/5"}}},"commit":{"parents":[{"commit":"57adeccbe2af451f1f54ddd39018d0a45b8b72c8","subject":"Merge \"Rename \"Weights\" to \"Weight\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/57adeccbe2af451f1f54ddd39018d0a45b8b72c8"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-02 11:57:19.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-02 12:03:10.000000000","tz":0},"subject":"Source_IP replace to source IP","message":"Source_IP replace to source IP\n\nand small fix for QoS writing\n\nChange-Id: I0b24bc84815e4e7c025dd0cdccbc634033a22c2b\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/30eeed148d0ec599a7b002293a6aebfaef237f77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/30eeed148d0ec599a7b002293a6aebfaef237f77"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
