)]}'
{"id":"openstack%2Fskyline-console~975417","triplet_id":"openstack%2Fskyline-console~stable%2F2025.1~I049a8dff49989c693518bff93683a8c778bb0002","project":"openstack/skyline-console","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-02-04 13:42:01.000000000","reason":"Change was submitted"},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-02-04 03:20:00.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}},"hashtags":[],"change_id":"I049a8dff49989c693518bff93683a8c778bb0002","subject":"Fix IPv6 subnet fields visibility in network form","status":"MERGED","created":"2026-02-02 16:16:56.000000000","updated":"2026-02-04 13:42:52.000000000","submitted":"2026-02-04 13:42:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"975417","cherry_pick_of_change":975374,"cherry_pick_of_patch_set":3,"meta_rev_id":"4769366ec57dd444cf38d07c67adee366faf46c1","_number":975417,"virtual_id_number":975417,"owner":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-04 13:42:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-02-04 03:20:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-02-02 16:50:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-02-04 03:20:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-02 16:50:54.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-02-02 16:50:54.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2026-02-02 17:35:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0c96876e35acc62c89b388de7de295f5c1c3432b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-02-02 16:16:56.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"5da5777974d187e10909ea39edfe8f6eeeabed82","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-02 16:50:54.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"40f75092a2e750e5051b955a68daa54a7316df8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-02 17:35:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a7c4e41589c4cf9b39a0b3d42092f1d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89e782dafc594730a3af4573beb9e7d1 : SUCCESS in 5m 48s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/312184ae71304317bbbbdeeecbbd6fe4 : SUCCESS in 12m 26s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/bf7226b8eaa04404bbf3834c3318bd5f : SUCCESS in 9m 06s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/919a0162a3214112b708cf141ebc7172 : SUCCESS in 8m 02s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/02e0ddd7459f47d692677da1122aa93b : SUCCESS in 1h 03m 40s","accounts_in_message":[],"_revision_number":1},{"id":"e13463fc91b07b8d530a6e7f1d105013bb279f71","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-02-04 03:20:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8df3124a40ecd4162a158a63a535c3bad007521e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-04 03:20:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8e415ccc397a3106aaf3ff101240bad6c3134732","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-04 04:35:41.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3a61271eead1424eb36d2740bfffd9cb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a3022dfef3c4624869c8c27813d2093 : SUCCESS in 7m 12s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/c13105c1310d4d98a614825c094f1128 : SUCCESS in 10m 47s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/192a51b79b6d4522ba7f801323258f7a : SUCCESS in 5m 44s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/992013a8132746a7aa99e57381900847 : FAILURE in 2m 28s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/bfd77ca2bedc47049768b103aae62896 : SUCCESS in 1h 08m 06s","accounts_in_message":[],"_revision_number":1},{"id":"9dab7e882bd9327332feba970a1451f0fe3cd1ba","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-02-04 05:52:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"07f2b7bc3158ab86b51f0e17fa10c1581db88379","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-04 06:42:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff96a7c0b89d4f77937e1a5e9028c741\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8791769688ea4b52a2223a07571b85c5 : SUCCESS in 6m 43s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/72af1a8077764a57ab716f1e5ce27aaf : SUCCESS in 11m 28s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/b7072355b842424d892a16f56fbf05ff : SUCCESS in 7m 44s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/01ff74bf02384814add20c0768d3cdd4 : SUCCESS in 5m 23s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/e3502590cbfa4be9804b14c78207993b : SUCCESS in 46m 08s","accounts_in_message":[],"_revision_number":1},{"id":"0227b78aae32515d9ec9a6fda6e0f6708e38302c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-04 06:42:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8b3fe2c1119d8cb8a713368c2625b4b1b9270a70","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-04 07:07:08.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b64f8811da374b3d84ce5d31aa9e8e25\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9765bcbdf054c7ebb64a3b7aa5b6048 : SUCCESS in 6m 50s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/e91d70c06b794082ac23747d711a6b41 : SUCCESS in 7m 20s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/b4021e5fed764e649f3f0a2c89732bc4 : SUCCESS in 4m 39s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/56fcaf9a92e6423dbc2401b6c6abf971 : SUCCESS in 6m 10s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/8c8d5fd3a67a42cf96115fe20a2eadfb : FAILURE in 17m 10s","accounts_in_message":[],"_revision_number":1},{"id":"4c364992ac4415ce9f764330a27d2b10ed9d7716","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2026-02-04 07:15:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1af2317c0773bbb1e36e48283462d7e39865e322","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-04 07:48:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d24a2078bbf348068db380d7c5448c4f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2da51413cb984910926fc61366bdb192 : SUCCESS in 5m 57s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/cfec72669f1c49ffb39299d7843612d7 : SUCCESS in 6m 47s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/0b79146bdba34634b43884df66903a9c : SUCCESS in 4m 47s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/af488c113fc846039e5326dbf04a5c70 : SUCCESS in 4m 52s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/592985523bde49b6a7fc3e99fec184d6 : FAILURE in 31m 54s","accounts_in_message":[],"_revision_number":1},{"id":"19e2a207c140b9d902350c7510a00296018d2ff0","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-04 10:06:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1ba00c163c2e566ffe40446530594eb50a7413d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-04 10:50:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c90ee0a6e2d045d4b5e902ad47157a30\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e7f6cde68fd48b8828d951ed3f3702f : SUCCESS in 6m 08s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/5cf0dfae74624a5e8d8219b3cb570237 : SUCCESS in 8m 22s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/71850ddaf8024fa18abfd32f723850e8 : SUCCESS in 7m 29s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/9a894ee4a6744cb1816aea9a74926813 : SUCCESS in 3m 52s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/f2952094d7954a3b9a71a40c17ea7c26 : SUCCESS in 42m 45s","accounts_in_message":[],"_revision_number":1},{"id":"995793684043e9d96d1432ad68367d542eda2d04","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-04 10:50:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8158215da592d5785053998a8cc87e76b09376ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-04 11:59:01.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/df755adb2b9b42f29f719be936dd91b2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1714f2b288ff480e84f2557604325217 : FAILURE in 5m 27s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/14e28186bbe245baafd17e021cd409d4 : SUCCESS in 10m 40s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/e3772bbb91fe4a1f98497b0159aa18f3 : SUCCESS in 4m 24s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/1fd6e30e3ed140f1934e82d3bf3f7697 : SUCCESS in 7m 49s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/6d2a9cda9f12406d870c1adc72669cfc : SUCCESS in 1h 03m 21s","accounts_in_message":[],"_revision_number":1},{"id":"68262261809bbf113df1e2dc2edd3a1dbec1312f","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-04 11:59:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5431d10d2b2149c3c6c77c207c44620a06841534","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-04 12:37:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/343e800d8389453c97fee000c2a6ab78\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a04e8c3296954ce798556cff04ac32fe : SUCCESS in 6m 07s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/a7df6601cb564fbfb16cfd3b3812e552 : SUCCESS in 10m 49s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/ce7f903577f6438aa35ca8a702f531c0 : SUCCESS in 7m 48s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/d83445daadb24a1fb0138fa80cd0d5de : SUCCESS in 3m 59s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/11ee5d03f00c43db8c8c9af9f3ad089c : SUCCESS in 37m 03s","accounts_in_message":[],"_revision_number":1},{"id":"647fa3e9d53e1a40e3e53169b6c5f89d440866cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-04 12:37:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"829177abad8923c087932e183835cd8d0ac90d73","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-04 13:42:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c97a17c77f9e49a9a62556f3108474f7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b00d39056f584d4db565481e20f6587f : SUCCESS in 6m 32s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/ae5a9b0e7b73496bb8f56dd88ae097db : SUCCESS in 7m 38s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/3e8c1f6553bb441180c27de3299067d5 : SUCCESS in 6m 57s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/375909e5f04f407c97009a3816af6acf : SUCCESS in 7m 32s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/40dd3eb3380c4968a3a844ddebe335f7 : SUCCESS in 1h 02m 53s","accounts_in_message":[],"_revision_number":1},{"id":"a51570a0440a6b5ce5f030cb05c947246815aa55","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-04 13:42:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4769366ec57dd444cf38d07c67adee366faf46c1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-04 13:42:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5acb0485d64d440ea9fce23e9e695391\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf5255bb2b8c4b4991302ba1e44b2410 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"46fc2803ed3d895421adf45c1b719bad0dd930c7","revisions":{"46fc2803ed3d895421adf45c1b719bad0dd930c7":{"kind":"REWORK","_number":1,"created":"2026-02-02 16:16:56.000000000","uploader":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"ref":"refs/changes/17/975417/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/17/975417/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/17/975417/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/17/975417/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/17/975417/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/17/975417/1"}}},"commit":{"parents":[{"commit":"409b1bb06ac4d126b6133cd41b5f5adb5dfd6c9c","subject":"fix: Delete Loadbalancer not possible, fixes 2102154","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/409b1bb06ac4d126b6133cd41b5f5adb5dfd6c9c"}]}],"author":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-02-02 09:41:57.000000000","tz":0},"committer":{"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","date":"2026-02-02 16:16:56.000000000","tz":0},"subject":"Fix IPv6 subnet fields visibility in network form","message":"Fix IPv6 subnet fields visibility in network form\n\nHide IPv6 subnet fields when subnet creation is\ndisabled\n\nChange-Id: I049a8dff49989c693518bff93683a8c778bb0002\nCloses-Bug: #2125188\nSigned-off-by: Dmitriy Chubinidze \u003cdcu995@gmail.com\u003e\n(cherry picked from commit 6b12a08e6c32319c492ad5445b576c81126ac3ea)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/46fc2803ed3d895421adf45c1b719bad0dd930c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/46fc2803ed3d895421adf45c1b719bad0dd930c7"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
