)]}'
{"id":"openstack%2Fskyline-console~975425","triplet_id":"openstack%2Fskyline-console~master~I0eec751d684212b99a6f6631aabcae8307e96d08","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-02-04 06:07:45.000000000","reason":"Change was submitted"},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-02-04 03:51:49.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-02-04 06:07:45.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0eec751d684212b99a6f6631aabcae8307e96d08","subject":"Correct various grammar mistakes","status":"MERGED","created":"2026-02-02 16:51:25.000000000","updated":"2026-02-04 06:08:35.000000000","submitted":"2026-02-04 06:07:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"975425","meta_rev_id":"cf9dc1eee8e32cbe59f704fd40442c0a5366e3e1","_number":975425,"virtual_id_number":975425,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-04 06:07:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-02-04 03:51:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-02-04 03:51:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-02 17:36:38.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2026-02-02 17:36:38.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"},{"updated":"2026-02-02 19:02:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"561100ee8c4a12d8f2d72c4236b3650530185161","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-02 16:51:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8c7dd34ac2d66973177998155eae7b3f8c2caed4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-02 17:22:10.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"9a959d17a8ae529086b80f057373625d64f43c18","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-02 17:28:01.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"2c0c955f28c96feb69d9a28dd156d2ddb1a44479","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-02 17:34:26.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"62e9799c964e8da0c18534c79c8dbca9184c0fe0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-02 17:34:59.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"e5190e0788d6d93834bec14371516ebca6a0c3f0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-02 17:36:23.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"9e26ef7c6741ad799c319457c78447aaf5f1659f","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-02 17:36:38.000000000","message":"Patch Set 6:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":6},{"id":"d34ced221c4548c5450ab8b4322c3915c514480c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-02 19:02:21.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5646d3638b9c4ee394f15d8c9c0c7768\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd25ec6f266e415f86134b444cd34b1a : SUCCESS in 6m 35s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/a024266ba3924e4da344466f0982c342 : SUCCESS in 4m 13s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/91885f3d77a6475080a67138db223c1f : SUCCESS in 8m 31s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/07c0c4ea1dce4035bebc791f2cc225c1 : SUCCESS in 7m 01s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/d3e64a35322c47f48ce581c0d367f64e : SUCCESS in 1h 25m 08s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/a843fbc2c5c4427eb7fb3d372c334374 : SUCCESS in 1h 05m 40s","accounts_in_message":[],"_revision_number":6},{"id":"258e23e4209f171a363d62d0f4c55d55bf7eab68","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-02-04 03:51:49.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"d5c7f85e9c7723757891ab44db7db367dc8071a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-04 03:52:27.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"1117151db0cc16e5d98a51a0605beeba0d61ecdf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-04 06:07:45.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f4dace83b9849e3ba24ead4662bf8ea\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ac8cc14058f4e538b5473e256be2761 : SUCCESS in 5m 01s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/74ae50bd4dc14050a8ca30e5891adbdd : SUCCESS in 7m 35s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/27093a955f4a4feebf01f5bb65da7f83 : SUCCESS in 7m 53s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/e0e59b22a8dd45508d2f58e552a64095 : SUCCESS in 7m 04s\n- skyline-console-devstack-e2etests-compute https://zuul.opendev.org/t/openstack/build/800cb105c9f04f7d8aad0b0d242198b5 : SUCCESS in 1h 24m 11s\n- skyline-console-devstack-e2etests-network https://zuul.opendev.org/t/openstack/build/eaaac7e742e44e158fe1580886aceed2 : SUCCESS in 46m 28s","accounts_in_message":[],"_revision_number":6},{"id":"b1ea869b9945f0907c5024fe535a2f29342985e8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-04 06:07:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"cf9dc1eee8e32cbe59f704fd40442c0a5366e3e1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-04 06:08:35.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6461d0ec034c4bde9d5fbd18082e8c8e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78814beaa7f84d9a921c2e3bf1a0bd1c : SUCCESS in 42s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"b7cf0356a9698d2e22ac32c506bb095581303cfd","revisions":{"641becc668910984823692703e2aa68ab93f116c":{"kind":"REWORK","_number":1,"created":"2026-02-02 16:51:25.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/25/975425/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/25/975425/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/25/975425/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/25/975425/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/25/975425/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/25/975425/1"}}},"commit":{"parents":[{"commit":"cd3f344b6947b209925914390c9fc30d98f9603e","subject":"Merge \"Fix IPv6 subnet fields visibility in network form\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/cd3f344b6947b209925914390c9fc30d98f9603e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-02 16:51:25.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-02 16:51:25.000000000","tz":0},"subject":"wip","message":"wip\n\n\nChange-Id: I0eec751d684212b99a6f6631aabcae8307e96d08\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/641becc668910984823692703e2aa68ab93f116c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/641becc668910984823692703e2aa68ab93f116c"}]},"branch":"refs/heads/master"},"81d4c1889508e9fd443f195d61f1790b1577f23f":{"kind":"REWORK","_number":2,"created":"2026-02-02 17:22:10.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/25/975425/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/25/975425/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/25/975425/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/25/975425/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/25/975425/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/25/975425/2"}}},"commit":{"parents":[{"commit":"cd3f344b6947b209925914390c9fc30d98f9603e","subject":"Merge \"Fix IPv6 subnet fields visibility in network form\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/cd3f344b6947b209925914390c9fc30d98f9603e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-02 16:51:25.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-02 17:22:06.000000000","tz":0},"subject":"wip","message":"wip\n\n\nChange-Id: I0eec751d684212b99a6f6631aabcae8307e96d08\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/81d4c1889508e9fd443f195d61f1790b1577f23f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/81d4c1889508e9fd443f195d61f1790b1577f23f"}]},"branch":"refs/heads/master"},"e6a3039a63ff61c1790c3d0eae97e5c13138c0e4":{"kind":"REWORK","_number":3,"created":"2026-02-02 17:28:01.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/25/975425/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/25/975425/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/25/975425/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/25/975425/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/25/975425/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/25/975425/3"}}},"commit":{"parents":[{"commit":"cd3f344b6947b209925914390c9fc30d98f9603e","subject":"Merge \"Fix IPv6 subnet fields visibility in network form\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/cd3f344b6947b209925914390c9fc30d98f9603e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-02 16:51:25.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-02 17:28:01.000000000","tz":0},"subject":"wip","message":"wip\n\n\nChange-Id: I0eec751d684212b99a6f6631aabcae8307e96d08\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e6a3039a63ff61c1790c3d0eae97e5c13138c0e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/e6a3039a63ff61c1790c3d0eae97e5c13138c0e4"}]},"branch":"refs/heads/master"},"512130948a6d452515dc8709ec7e6c002eb3fd28":{"kind":"REWORK","_number":4,"created":"2026-02-02 17:34:26.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/25/975425/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/25/975425/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/25/975425/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/25/975425/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/25/975425/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/25/975425/4"}}},"commit":{"parents":[{"commit":"cd3f344b6947b209925914390c9fc30d98f9603e","subject":"Merge \"Fix IPv6 subnet fields visibility in network form\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/cd3f344b6947b209925914390c9fc30d98f9603e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-02 16:51:25.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-02 17:30:44.000000000","tz":0},"subject":"wip","message":"wip\n\n\nChange-Id: I0eec751d684212b99a6f6631aabcae8307e96d08\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/512130948a6d452515dc8709ec7e6c002eb3fd28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/512130948a6d452515dc8709ec7e6c002eb3fd28"}]},"branch":"refs/heads/master"},"c002eeea4682e8d337f2248f22cbb450903f1baa":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2026-02-02 17:34:59.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/25/975425/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/25/975425/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/25/975425/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/25/975425/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/25/975425/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/25/975425/5"}}},"commit":{"parents":[{"commit":"cd3f344b6947b209925914390c9fc30d98f9603e","subject":"Merge \"Fix IPv6 subnet fields visibility in network form\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/cd3f344b6947b209925914390c9fc30d98f9603e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-02 16:51:25.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-02 17:34:59.000000000","tz":0},"subject":"Correct various grammar mistakes","message":"Correct various grammar mistakes\n\nChange-Id: I0eec751d684212b99a6f6631aabcae8307e96d08\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/c002eeea4682e8d337f2248f22cbb450903f1baa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/c002eeea4682e8d337f2248f22cbb450903f1baa"}]},"branch":"refs/heads/master","description":"Edit commit message"},"b7cf0356a9698d2e22ac32c506bb095581303cfd":{"kind":"REWORK","_number":6,"created":"2026-02-02 17:36:23.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/25/975425/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/25/975425/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/25/975425/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/25/975425/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/25/975425/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/25/975425/6"}}},"commit":{"parents":[{"commit":"cd3f344b6947b209925914390c9fc30d98f9603e","subject":"Merge \"Fix IPv6 subnet fields visibility in network form\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/cd3f344b6947b209925914390c9fc30d98f9603e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-02 16:51:25.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-02 17:36:23.000000000","tz":0},"subject":"Correct various grammar mistakes","message":"Correct various grammar mistakes\n\nChange-Id: I0eec751d684212b99a6f6631aabcae8307e96d08\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b7cf0356a9698d2e22ac32c506bb095581303cfd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b7cf0356a9698d2e22ac32c506bb095581303cfd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
