)]}'
{"id":"openstack%2Fskyline-console~975504","triplet_id":"openstack%2Fskyline-console~master~I5f907cf91a0dea11244ee9433acc108bbd975392","project":"openstack/skyline-console","branch":"master","attention_set":{},"removed_from_attention_set":{"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2026-02-04 06:07:46.000000000","reason":"Change was submitted"},"30434":{"account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"last_update":"2026-02-04 03:52:12.000000000","reason":"\u003cGERRIT_ACCOUNT_30434\u003e replied on the change","reason_account":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-02-04 06:07:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5f907cf91a0dea11244ee9433acc108bbd975392","subject":"Object Storage - rename form, fix for \"New Name\" field","status":"MERGED","created":"2026-02-03 13:25:54.000000000","updated":"2026-02-04 06:09:36.000000000","submitted":"2026-02-04 06:07:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"975504","meta_rev_id":"58346731c125435fa10c1f3dc5a9c3419ac7d314","_number":975504,"virtual_id_number":975504,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-04 06:07:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":2,"date":"2026-02-04 03:52:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"all":[{"value":1,"date":"2026-02-04 03:52:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-03 13:29:19.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"},{"updated":"2026-02-03 13:29:19.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"},{"updated":"2026-02-03 14:20:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"96f88520677168a39e1cb6b06bffd95e9e015618","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-03 13:25:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7846428c76d2f65e03db2f28e65d8b3dc5dc129a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-03 13:27:55.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"82a79feb2aabf42a9a16931bb47eecc547108a44","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-03 13:29:09.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"ae18847ef4507f55ac6617b828c4ac5ebf2e2621","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-03 13:29:19.000000000","message":"Patch Set 3:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"62af77ac9e417de2d52015e0b02a81aa23361672","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-03 13:29:34.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"84bd3fb54f4537cc0c9440c82f015dc957d7e2a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-03 14:20:47.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50609641c5cd4ac5a746bebaea02eb05\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf7ce2918ad94bbabd1bb6e56d7fe195 : SUCCESS in 7m 32s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/495ce7d014e04d23b2e8e95fef29e490 : SUCCESS in 12m 29s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/2de73894ac774485a0a82613d7805c16 : SUCCESS in 3m 14s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/caf5339312b24186898d92403a43f9de : SUCCESS in 9m 01s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/cf80da2dc5e145cdbba665d5fab6b158 : SUCCESS in 47m 56s","accounts_in_message":[],"_revision_number":4},{"id":"7f76f42947bb5e127bf9a1f3d11b02252dad0399","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2026-02-04 03:52:12.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"8e9c3adb99df71218967f88995d59c7f42a66675","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-04 03:52:44.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"965cb94ed3e046af515698e2082caed814e41b2a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-04 06:07:46.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1fe33f943974c3694bc9e51d05377cf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eead28efb6b64e5bb7e7df76b625499d : SUCCESS in 3m 15s\n- skyline-nodejs16-run-lint-src https://zuul.opendev.org/t/openstack/build/0f42bfe8c52e44f08f24516ce953e5cf : SUCCESS in 11m 11s\n- skyline-nodejs16-run-lint-test https://zuul.opendev.org/t/openstack/build/ce2c9021c8c7472e9d10f9b18b255723 : SUCCESS in 5m 00s\n- skyline-nodejs16-run-unit-test https://zuul.opendev.org/t/openstack/build/44bc0d008a9a496ab22dd7949c04a62d : SUCCESS in 4m 24s\n- skyline-console-devstack-e2etests-storage https://zuul.opendev.org/t/openstack/build/006c2a47348b49559ab30ed4bd0615eb : SUCCESS in 49m 50s","accounts_in_message":[],"_revision_number":4},{"id":"0ff7e8725216072050caa437c511c9132a765ffa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-04 06:07:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"58346731c125435fa10c1f3dc5a9c3419ac7d314","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-04 06:09:36.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c6a43ee590764eef9044db6861a23f8f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1db5b37b48594832bc5889524bdeab6b : SUCCESS in 44s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"2f1078b9d2005b3c9224b759f101fcabeac2b853","revisions":{"dceca8d408580b837168708d31ce7ebfadc84884":{"kind":"REWORK","_number":1,"created":"2026-02-03 13:25:54.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/04/975504/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/04/975504/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/04/975504/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/04/975504/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/04/975504/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/04/975504/1"}}},"commit":{"parents":[{"commit":"cd3f344b6947b209925914390c9fc30d98f9603e","subject":"Merge \"Fix IPv6 subnet fields visibility in network form\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/cd3f344b6947b209925914390c9fc30d98f9603e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-03 13:25:54.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-03 13:25:54.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: I5f907cf91a0dea11244ee9433acc108bbd975392\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/dceca8d408580b837168708d31ce7ebfadc84884"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/dceca8d408580b837168708d31ce7ebfadc84884"}]},"branch":"refs/heads/master"},"d6abba809e81fd4f2d2943815f031fa303fba7e6":{"kind":"REWORK","_number":2,"created":"2026-02-03 13:27:55.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/04/975504/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/04/975504/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/04/975504/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/04/975504/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/04/975504/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/04/975504/2"}}},"commit":{"parents":[{"commit":"cd3f344b6947b209925914390c9fc30d98f9603e","subject":"Merge \"Fix IPv6 subnet fields visibility in network form\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/cd3f344b6947b209925914390c9fc30d98f9603e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-03 13:25:54.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-03 13:27:55.000000000","tz":0},"subject":"wip","message":"wip\n\nChange-Id: I5f907cf91a0dea11244ee9433acc108bbd975392\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d6abba809e81fd4f2d2943815f031fa303fba7e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/d6abba809e81fd4f2d2943815f031fa303fba7e6"}]},"branch":"refs/heads/master"},"b4cc972371e629525a07e0c415d405dfe0a6181c":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-02-03 13:29:09.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/04/975504/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/04/975504/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/04/975504/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/04/975504/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/04/975504/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/04/975504/3"}}},"commit":{"parents":[{"commit":"cd3f344b6947b209925914390c9fc30d98f9603e","subject":"Merge \"Fix IPv6 subnet fields visibility in network form\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/cd3f344b6947b209925914390c9fc30d98f9603e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-03 13:25:54.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-03 13:29:09.000000000","tz":0},"subject":"Object Storage - rename form, fix for \"New Name\" field","message":"Object Storage - rename form, fix for \"New Name\" field\n\nFor prevent any confusion with understanding from users side.\n\nChange-Id: I5f907cf91a0dea11244ee9433acc108bbd975392\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b4cc972371e629525a07e0c415d405dfe0a6181c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/b4cc972371e629525a07e0c415d405dfe0a6181c"}]},"branch":"refs/heads/master","description":"Edit commit message"},"2f1078b9d2005b3c9224b759f101fcabeac2b853":{"kind":"REWORK","_number":4,"created":"2026-02-03 13:29:34.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/04/975504/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/skyline-console","ref":"refs/changes/04/975504/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/04/975504/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/04/975504/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/skyline-console refs/changes/04/975504/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/skyline-console refs/changes/04/975504/4"}}},"commit":{"parents":[{"commit":"cd3f344b6947b209925914390c9fc30d98f9603e","subject":"Merge \"Fix IPv6 subnet fields visibility in network form\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/cd3f344b6947b209925914390c9fc30d98f9603e"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-03 13:25:54.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-02-03 13:29:34.000000000","tz":0},"subject":"Object Storage - rename form, fix for \"New Name\" field","message":"Object Storage - rename form, fix for \"New Name\" field\n\nFor prevent any confusion with understanding from users side.\n\nChange-Id: I5f907cf91a0dea11244ee9433acc108bbd975392\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/2f1078b9d2005b3c9224b759f101fcabeac2b853"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/skyline-console/commit/2f1078b9d2005b3c9224b759f101fcabeac2b853"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
